)]}'
{"id":"openstack%2Fgovernance~341421","triplet_id":"openstack%2Fgovernance~master~Ia9007a8df86ce0a50f117cc20b4ef28cf7fc0fc0","project":"openstack/governance","branch":"master","topic":"add-charm-aodh","hashtags":[],"change_id":"Ia9007a8df86ce0a50f117cc20b4ef28cf7fc0fc0","subject":"Add charm-aodh to OpenStack Charms project","status":"MERGED","created":"2016-07-13 10:49:01.000000000","updated":"2016-07-22 10:21:08.000000000","submitted":"2016-07-22 10:21:08.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"301f1a982063dadebafe014a98fcf14ad89f2f07","_number":341421,"virtual_id_number":341421,"owner":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2016-07-22 10:21:08.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2016-07-13 10:51:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-07-22 10:16:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2016-07-13 11:09:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-07-22 10:16:45.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-07-22 10:16:45.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2016-07-13 11:09:30.000000000","_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-07-13 10:51:49.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2016-07-13 11:09:30.000000000","updated_by":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"reviewer":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"state":"REVIEWER"},{"updated":"2016-07-22 10:16:45.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2016-07-22 10:21:08.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"649e7921d340b517dd07b8f5e385e3bdcd6c1429","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2016-07-13 10:49:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94b59f607e28c2bd4ad67afe36df669b7db8c0aa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-13 10:50:18.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"b5b47e340757c91f1f4131154f12affa1d016af8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2016-07-13 10:51:49.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"784c921b4e5f7a32585994ee94e6133b49a73ed3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-13 11:02:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-governance-docs http://docs-draft.openstack.org/21/341421/1/check/gate-governance-docs/b49ec6e//doc/build/html/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1204b98ebef5f33ecde244af387598ff4a8a1ab5","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2016-07-13 11:09:30.000000000","message":"Patch Set 1: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"eae4d873553ff35c9a31977fa9cc70b4998cf3b2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2016-07-22 10:16:45.000000000","message":"Patch Set 1: Code-Review+1 Rollcall-Vote+1 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b2e6122845e7679ac9a73e25399140d677e0e7ef","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-22 10:16:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"6b45babb3e2e3674565eec0ae6dbc7b915ccfb39","date":"2016-07-22 10:21:08.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"68678d90efd428c89fe9862f1955b8444b2029b3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-22 10:21:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-governance-docs http://docs-draft.openstack.org/21/341421/1/gate/gate-governance-docs/81af924//doc/build/html/ : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b8a339c259120c8a6fafecd81261a24bdf236de","revisions":{"9b8a339c259120c8a6fafecd81261a24bdf236de":{"kind":"REWORK","_number":1,"created":"2016-07-13 10:49:01.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/21/341421/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/21/341421/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/21/341421/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/21/341421/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/21/341421/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/21/341421/1"}}},"commit":{"parents":[{"commit":"712e16eb312e3b317fd1d31b0b8230b6c185998d","subject":"Merge \"Exclude inactive core reviewers from core metrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/712e16eb312e3b317fd1d31b0b8230b6c185998d"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2016-07-13 10:47:28.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@ubuntu.com","date":"2016-07-13 10:47:28.000000000","tz":60},"subject":"Add charm-aodh to OpenStack Charms project","message":"Add charm-aodh to OpenStack Charms project\n\nAdd the new Aodh charm to the OpenStack Charms project.\n\nChange-Id: Ia9007a8df86ce0a50f117cc20b4ef28cf7fc0fc0\nDepends-On: I4dcb020366b2c94a4ed2fdda28669699da9694c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/9b8a339c259120c8a6fafecd81261a24bdf236de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/9b8a339c259120c8a6fafecd81261a24bdf236de"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"712e16eb312e3b317fd1d31b0b8230b6c185998d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
