)]}'
{"id":"openstack%2Fgovernance~451625","triplet_id":"openstack%2Fgovernance~master~Ie8c0805e14e6952cce2132a6c5c9a0844a2edbab","project":"openstack/governance","branch":"master","topic":"formal-vote","hashtags":[],"change_id":"Ie8c0805e14e6952cce2132a6c5c9a0844a2edbab","subject":"[I18n] Renew \u0026 add Zanata dev members as extra ATC","status":"MERGED","created":"2017-03-30 02:30:51.000000000","updated":"2017-04-05 09:20:43.000000000","submitted":"2017-04-05 09:20:43.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b202cea192e1235b9650976f35422374c42c98da","_number":451625,"virtual_id_number":451625,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2017-04-05 09:20:43.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2017-04-05 08:45:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-04-05 08:45:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":1,"date":"2017-04-05 08:45:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-04-05 09:17:04.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2017-04-05 08:45:59.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2017-04-05 08:45:59.000000000","_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-04-05 08:45:59.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2017-04-05 08:45:59.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":1,"date":"2017-04-05 08:45:59.000000000","_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":1,"date":"2017-04-05 08:45:59.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-30 03:52:31.000000000","updated_by":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"reviewer":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"state":"REVIEWER"},{"updated":"2017-03-31 12:25:25.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2017-04-03 15:59:57.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2017-04-03 18:09:17.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2017-04-04 16:48:26.000000000","updated_by":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"reviewer":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"state":"REVIEWER"},{"updated":"2017-04-04 20:03:58.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2017-04-05 09:17:04.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2017-04-05 09:20:43.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"5e5573d3219b65c9a94624f99384d7745f2c5b0e","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2017-03-30 02:30:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbfcfe4a10a4128671eb4ef538c7826cf6abe2ff","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2017-03-30 02:31:47.000000000","message":"Topic changed from bp/s to i18n-extra-atc","accounts_in_message":[],"_revision_number":1},{"id":"c3f2fcd91d8381b77c20487eab0eda24ef45be7f","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2017-03-30 02:36:56.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9bc9818d084673f19cab8f05e9e7b3cbf5b9cbb1","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2017-03-30 02:37:33.000000000","message":"Topic changed from bp/s to i18n-extra-atc","accounts_in_message":[],"_revision_number":2},{"id":"eeccaa5d5ba822edd2929c78a6a7896dee4a7c92","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-30 02:51:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-governance-linters-ubuntu-xenial http://logs.openstack.org/25/451625/2/check/gate-governance-linters-ubuntu-xenial/0772363/ : SUCCESS in 1m 43s\n- gate-governance-docs-ubuntu-xenial http://docs-draft.openstack.org/25/451625/2/check/gate-governance-docs-ubuntu-xenial/7137461//doc/build/html/ : SUCCESS in 2m 59s\n- gate-governance-tox-validate-ubuntu-xenial http://logs.openstack.org/25/451625/2/check/gate-governance-tox-validate-ubuntu-xenial/827804e/ : SUCCESS in 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1dd922f2255965148232cf02e726a274c91aaa9c","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2017-03-30 03:52:31.000000000","message":"Patch Set 2: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"5977f6dd08267471cc9bdffd1b39475a04d660ea","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2017-03-31 09:22:58.000000000","message":"Patch Set 2: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"27586726a7f6585e954a71bec8c6342bd29f84db","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2017-03-31 09:23:06.000000000","message":"Topic changed from i18n-extra-atc to formal-vote","accounts_in_message":[],"_revision_number":2},{"id":"4d5ded1890076514ae8b76c9f61ad86c51997c22","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2017-03-31 12:25:25.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"86d5c6c4bb98dfd22b930ec1c4f7d8f38a64f974","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2017-04-03 15:59:57.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"6b2ea544586a36d17f6b082c136fdc89fa6c2cbc","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-04-03 18:09:17.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"56a2c641e3d2a3810e6918130d1c0af77322219d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2017-04-04 09:13:43.000000000","message":"Patch Set 2: Code-Review-1 -Rollcall-Vote\n\n(3 comments)\n\nExtra-ATCs need to be active Foundation members, and I could not match some of those. Could you ask them to register or give us their Foundation membership link ?","accounts_in_message":[],"_revision_number":2},{"id":"f9a598190b2f85ee45fe1c6cf650db8f453e9f06","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2017-04-04 10:19:02.000000000","message":"Patch Set 2:\n\n\u003e Could you ask them to register or give us\n \u003e their Foundation membership link ?\n\nYep I will ask them and update here. Thanks!","accounts_in_message":[],"_revision_number":2},{"id":"d13117ce21335e8dae17c89abf3a5c42908c8c55","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2017-04-04 16:48:26.000000000","message":"Patch Set 2: Code-Review+1\n\nwill vote as soon as there\u0027s a matching membership for the folks that are currently missing one","accounts_in_message":[],"_revision_number":2},{"id":"1c97c76f2bf12ea4cec8247de8f72ce8224edeaa","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2017-04-04 20:03:58.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\nI am fine with this merging as needed, if everything is ship shape.","accounts_in_message":[],"_revision_number":2},{"id":"95eda32897007a8ee2cf346f3a49fbba199d379b","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2017-04-05 01:52:13.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nI replied all the link for their Foundation membership by asking them.\n\nAlso, I have updated https://wiki.openstack.org/wiki/I18nTeam#People not to forget the links :) Thanks!\n\n \u003e Could you ask them to register or give us\n \u003e their Foundation membership link ?","accounts_in_message":[],"_revision_number":2},{"id":"576826057f4281933074d8b538f3d74c3d0ee7d0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2017-04-05 08:14:56.000000000","message":"Patch Set 2: Code-Review+1 Rollcall-Vote+1 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a4dbed165ffee9896bcc160d4ff9a5660f33169d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-05 08:15:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"4ab74b86941dffd83761bfc30960cb66c6792a85","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-05 08:17:57.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-governance-linters-ubuntu-xenial http://logs.openstack.org/25/451625/2/gate/gate-governance-linters-ubuntu-xenial/a1bac88/ : SUCCESS in 2m 19s\n- gate-governance-docs-ubuntu-xenial http://logs.openstack.org/25/451625/2/gate/gate-governance-docs-ubuntu-xenial/36a4880/ : FAILURE in 1m 48s","accounts_in_message":[],"_revision_number":2},{"id":"5fd49e25fc0a676026b4be684a722a37508189e9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2017-04-05 08:45:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5d4311cb8861fcf09f45f8777bd1e54d1daccf79","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-05 09:05:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-governance-linters-ubuntu-xenial http://logs.openstack.org/25/451625/3/check/gate-governance-linters-ubuntu-xenial/9d25e29/ : SUCCESS in 2m 12s\n- gate-governance-docs-ubuntu-xenial http://docs-draft.openstack.org/25/451625/3/check/gate-governance-docs-ubuntu-xenial/cc4f35d//doc/build/html/ : SUCCESS in 2m 08s\n- gate-governance-tox-validate-ubuntu-xenial http://logs.openstack.org/25/451625/3/check/gate-governance-tox-validate-ubuntu-xenial/203cada/ : SUCCESS in 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8c413d8707c8d77b74ab7c527458fb3fe2788ded","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2017-04-05 09:17:04.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e0e183ef61dd43a1a2e7f8e7fb4b8b771cba5104","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-05 09:17:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"c1152a90eba0eb2b7f7fc17dc4174f9d9c5883c2","date":"2017-04-05 09:20:43.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"b202cea192e1235b9650976f35422374c42c98da","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-05 09:20:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-governance-linters-ubuntu-xenial http://logs.openstack.org/25/451625/3/gate/gate-governance-linters-ubuntu-xenial/2c10130/ : SUCCESS in 2m 57s\n- gate-governance-docs-ubuntu-xenial http://docs-draft.openstack.org/25/451625/3/gate/gate-governance-docs-ubuntu-xenial/4833344//doc/build/html/ : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dc21ca44f72e321c96653fd638cfbec613b6d9d1","revisions":{"8e905ef7607702774344e85c22dfdcc2acfb3044":{"kind":"REWORK","_number":1,"created":"2017-03-30 02:30:51.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/25/451625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/25/451625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/25/451625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/25/451625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/25/451625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/25/451625/1"}}},"commit":{"parents":[{"commit":"139ce763fc8268817e3ca3d724fadf0f20c7e46e","subject":"Merge \"Remove Gnocchi from Telemetry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/139ce763fc8268817e3ca3d724fadf0f20c7e46e"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2017-03-30 02:12:42.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2017-03-30 02:28:17.000000000","tz":540},"subject":"[I18n] Renew \u0026 add Zanata dev members as extra ATC","message":"[I18n] Renew \u0026 add Zanata dev members as extra ATC\n\nZanata dev team has continuously helped I18n team and Infra team\nto setup our translation system with open source software Zanata [1].\nAlso, they collect our requirements for Zanata and enhance Zanata.\nSince they are not translators, they do not satisfy our standard\nATC requirements [2], but they helped a lot with kind opinions and\ncomments on translation platform such as Zanata upgrade [3].\n\nMore details:\n - Carlos is the manager in Zanata development team.\n   He encouages Alex and Patrick to help I18n team with answering\n   to OpenStack I18n opinions like [4].\n - Alex has helped a lot to collect Zanata feedback and Zanata upgrade\n   with detail configuration suggestions [5].\n - Patrick also helped [5] and kindly answered Zanata upgrade details\n   with opinions in openstack-i18n and openstack-infra IRC channel [6].\n\n[1] https://translate.zanata.org/\n[2] https://docs.openstack.org/developer/i18n/official_translator.html#atc-status-in-i18n-project\n[2] https://blueprints.launchpad.net/openstack-i18n/+spec/ocata-zanata-upgrade\n[3] http://lists.openstack.org/pipermail/openstack-i18n/2016-July/002300.html\n[4] https://review.openstack.org/#/c/422373/\n[5] http://eavesdrop.openstack.org/irclogs/%23openstack-infra/%23openstack-infra.2017-01-19.log.html#t2017-01-19T03:01:09\n\nChange-Id: Ie8c0805e14e6952cce2132a6c5c9a0844a2edbab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/8e905ef7607702774344e85c22dfdcc2acfb3044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/8e905ef7607702774344e85c22dfdcc2acfb3044"}]},"branch":"refs/heads/master"},"59a73cef7e3bfc0830f407462eda7a355600630c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2017-03-30 02:36:56.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/25/451625/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/25/451625/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/25/451625/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/25/451625/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/25/451625/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/25/451625/2"}}},"commit":{"parents":[{"commit":"139ce763fc8268817e3ca3d724fadf0f20c7e46e","subject":"Merge \"Remove Gnocchi from Telemetry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/139ce763fc8268817e3ca3d724fadf0f20c7e46e"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2017-03-30 02:12:42.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2017-03-30 02:36:38.000000000","tz":540},"subject":"[I18n] Renew \u0026 add Zanata dev members as extra ATC","message":"[I18n] Renew \u0026 add Zanata dev members as extra ATC\n\nZanata dev team has continuously helped I18n team and Infra team\nto setup our translation system with open source software Zanata [1].\nAlso, they collect our requirements for Zanata and enhance Zanata.\nSince they are not translators, they do not satisfy our standard\nATC requirements [2], but they helped a lot with kind opinions and\ncomments on translation platform such as Zanata upgrade [3].\n\nMore details:\n - Carlos is the manager in Zanata development team.\n   He encouages Alex and Patrick to help I18n team with answering\n   to OpenStack I18n opinions like [4].\n - Alex has helped a lot to collect Zanata feedback and Zanata upgrade\n   with detail configuration suggestions [5].\n - Patrick also helped [5] and kindly answered Zanata upgrade details\n   with opinions in openstack-i18n and openstack-infra IRC channel [6].\n\n[1] https://translate.zanata.org/\n[2] https://docs.openstack.org/developer/i18n/official_translator.html#atc-status-in-i18n-project\n[3] https://blueprints.launchpad.net/openstack-i18n/+spec/ocata-zanata-upgrade\n[4] http://lists.openstack.org/pipermail/openstack-i18n/2016-July/002300.html\n[5] https://review.openstack.org/#/c/422373/\n[6] http://eavesdrop.openstack.org/irclogs/%23openstack-infra/%23openstack-infra.2017-01-19.log.html#t2017-01-19T03:01:09\n\nChange-Id: Ie8c0805e14e6952cce2132a6c5c9a0844a2edbab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/59a73cef7e3bfc0830f407462eda7a355600630c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/59a73cef7e3bfc0830f407462eda7a355600630c"}]},"branch":"refs/heads/master"},"dc21ca44f72e321c96653fd638cfbec613b6d9d1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2017-04-05 08:45:59.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/25/451625/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/25/451625/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/25/451625/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/25/451625/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/25/451625/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/25/451625/3"}}},"commit":{"parents":[{"commit":"5ddcdce53abd279fcc0ab3942248aef1018ab6e5","subject":"Blacklist Pillow 4.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/5ddcdce53abd279fcc0ab3942248aef1018ab6e5"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2017-03-30 02:12:42.000000000","tz":540},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2017-04-05 08:45:59.000000000","tz":0},"subject":"[I18n] Renew \u0026 add Zanata dev members as extra ATC","message":"[I18n] Renew \u0026 add Zanata dev members as extra ATC\n\nZanata dev team has continuously helped I18n team and Infra team\nto setup our translation system with open source software Zanata [1].\nAlso, they collect our requirements for Zanata and enhance Zanata.\nSince they are not translators, they do not satisfy our standard\nATC requirements [2], but they helped a lot with kind opinions and\ncomments on translation platform such as Zanata upgrade [3].\n\nMore details:\n - Carlos is the manager in Zanata development team.\n   He encouages Alex and Patrick to help I18n team with answering\n   to OpenStack I18n opinions like [4].\n - Alex has helped a lot to collect Zanata feedback and Zanata upgrade\n   with detail configuration suggestions [5].\n - Patrick also helped [5] and kindly answered Zanata upgrade details\n   with opinions in openstack-i18n and openstack-infra IRC channel [6].\n\n[1] https://translate.zanata.org/\n[2] https://docs.openstack.org/developer/i18n/official_translator.html#atc-status-in-i18n-project\n[3] https://blueprints.launchpad.net/openstack-i18n/+spec/ocata-zanata-upgrade\n[4] http://lists.openstack.org/pipermail/openstack-i18n/2016-July/002300.html\n[5] https://review.openstack.org/#/c/422373/\n[6] http://eavesdrop.openstack.org/irclogs/%23openstack-infra/%23openstack-infra.2017-01-19.log.html#t2017-01-19T03:01:09\n\nChange-Id: Ie8c0805e14e6952cce2132a6c5c9a0844a2edbab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/dc21ca44f72e321c96653fd638cfbec613b6d9d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/dc21ca44f72e321c96653fd638cfbec613b6d9d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
