)]}'
{"id":"openstack%2Fgovernance~655010","triplet_id":"openstack%2Fgovernance~master~I64ca9232d168f6b8203b7b19cbf7280635dded52","project":"openstack/governance","branch":"master","topic":"typo-fix","hashtags":[],"change_id":"I64ca9232d168f6b8203b7b19cbf7280635dded52","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 11:38:15.000000000","updated":"2019-06-27 11:49:50.000000000","submitted":"2019-06-27 11:49:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655010-1561636190830-be057232","meta_rev_id":"3355e476ae566d7aaf5c3208b3dc6864ab931ed3","_number":655010,"virtual_id_number":655010,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":2,"date":"2019-06-27 11:49:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-27 11:41:02.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2019-06-27 09:46:51.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-27 11:41:02.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 12:00:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-24 03:09:26.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-05-29 00:02:33.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-06-13 08:20:21.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-06-25 09:26:57.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-06-27 09:46:51.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-06-27 11:41:02.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-27 11:49:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6682da5621d7df89067d5712103ab4dc87fc58fb","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 11:38:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9392ec8c08a25e88e60b91d3fd219bea90ca43d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 11:55:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/10/655010/1/check/test-release-openstack/d02b16e/ : SUCCESS in 5m 39s\n- openstack-tox-docs http://logs.openstack.org/10/655010/1/check/openstack-tox-docs/df9efc9/html/ : SUCCESS in 7m 10s\n- openstack-tox-py36 http://logs.openstack.org/10/655010/1/check/openstack-tox-py36/4e7aa7f/ : SUCCESS in 4m 42s\n- openstack-tox-linters http://logs.openstack.org/10/655010/1/check/openstack-tox-linters/7135944/ : FAILURE in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"133972aaad9d5b09009d8d4d4863c77c9d6984b1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-23 12:00:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nLot\u0027s of incorrect links.","accounts_in_message":[],"_revision_number":1},{"id":"38a5a8f9970fa6b004c2612ddc801f8f326f7ede","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-24 03:09:26.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nmost of the links are 404.","accounts_in_message":[],"_revision_number":1},{"id":"eadd1a73c93c70f0b9913467c8f404e9a6badd76","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-25 20:26:43.000000000","message":"Patch Set 1:\n\ncaoyuan, could you revise this patch based on the feedback please? :)","accounts_in_message":[],"_revision_number":1},{"id":"ad7f63d32096ef343d7bc811c45b0d46d0b1cbb8","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-25 20:26:51.000000000","message":"Topic set to typo-fix","accounts_in_message":[],"_revision_number":1},{"id":"aa9e41f8cf548013469c4513de7df9cac8fe476c","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-26 09:54:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4f0813ddd8a1f9a7a0d6b49004012a48925f5c8d","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-26 09:59:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f4ddbfcc9427302ddd8dbb80a3d9873fb7460f64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 10:09:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/10/655010/3/check/test-release-openstack/74bb6b2/ : SUCCESS in 3m 27s\n- openstack-tox-docs http://logs.openstack.org/10/655010/3/check/openstack-tox-docs/f1e4c7d/html/ : SUCCESS in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/10/655010/3/check/openstack-tox-py36/ed6432e/ : SUCCESS in 4m 26s\n- openstack-tox-linters http://logs.openstack.org/10/655010/3/check/openstack-tox-linters/eb24e9f/ : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":3},{"id":"1323425c600ecc881ca578d4bcc172dc74a7604d","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-06 17:30:54.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"80032ad96153d7ff9412ace07dede384b6db2fde","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-27 02:54:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9bf16976274dc6e9e487f72b91d486c5c922bb50","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-27 02:57:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"41101931e42d3f6c5056170be3f52a164395fd92","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-27 03:05:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f40766c8cc6fa4cfb9b51621da16ae6e73550a8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 03:24:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/10/655010/6/check/test-release-openstack/1445ec4/ : SUCCESS in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/10/655010/6/check/openstack-tox-docs/847ece6/html/ : SUCCESS in 6m 29s\n- openstack-tox-py36 http://logs.openstack.org/10/655010/6/check/openstack-tox-py36/a932c7c/ : SUCCESS in 3m 59s\n- openstack-tox-linters http://logs.openstack.org/10/655010/6/check/openstack-tox-linters/faf496c/ : FAILURE in 6m 18s","accounts_in_message":[],"_revision_number":6},{"id":"4e4a5850339ec98aefacb0506931f4c83c016395","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-27 14:08:57.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"230f59e319613b3bcdfacf86538769a54cd8e5ee","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-29 00:02:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ed76760666a8b6e09574aa14ee60a6ec80357d6f","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-06-13 08:20:21.000000000","message":"Patch Set 6: Code-Review-1\n\nplease fix conflict","accounts_in_message":[],"_revision_number":6},{"id":"7567d2e09541dfdd6703b302c7f0f1642751b6ac","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-25 09:26:57.000000000","message":"Patch Set 6: Code-Review-1","accounts_in_message":[],"_revision_number":6},{"id":"4d2af4d3257d2d87a96e97534aae663fa170d196","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-06-25 09:31:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5df9b236d5436106230c62ee68a41cdc016265b6","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-06-25 09:32:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ed6355338aee13b5f9e4aeaa489b16692288cb52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 09:45:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/10/655010/8/check/test-release-openstack/eb884d5/ : SUCCESS in 4m 01s\n- openstack-tox-docs http://logs.openstack.org/10/655010/8/check/openstack-tox-docs/195b382/ : FAILURE in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/10/655010/8/check/openstack-tox-py36/48d8ff3/ : SUCCESS in 3m 28s\n- openstack-tox-linters http://logs.openstack.org/10/655010/8/check/openstack-tox-linters/af0d592/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":8},{"id":"714fa2e325af4167e25a3cec1e2dc7811eb25025","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-27 09:26:19.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"731b1e2fd33925161207667c9dba373a0403142f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 09:37:33.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/10/655010/9/check/test-release-openstack/ea088b9/ : SUCCESS in 2m 59s\n- openstack-tox-docs http://logs.openstack.org/10/655010/9/check/openstack-tox-docs/a2361c9/html/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/10/655010/9/check/openstack-tox-py36/a93e696/ : SUCCESS in 4m 24s\n- openstack-tox-linters http://logs.openstack.org/10/655010/9/check/openstack-tox-linters/a2c4840/ : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":9},{"id":"d69cb8b20152dc97c3c586d258e7fdd45dc8392d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-27 09:46:51.000000000","message":"Patch Set 9: Rollcall-Vote+1\n\nNot super-convinced this is necessary for old files (like old goals), but I guess it does not hurt","accounts_in_message":[],"_revision_number":9},{"id":"807235ef583514a7de284fbedb46f2be202d42b2","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-27 11:41:02.000000000","message":"Patch Set 9: Rollcall-Vote+1 Workflow+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"28297c4541b25ed7b73bf1fb362e9e0170c8b3a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:41:14.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"71dc52ba23da488cc0f82f68d0cd297fc6ca5a84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:49:50.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/10/655010/9/gate/test-release-openstack/0053e74/ : SUCCESS in 2m 09s\n- openstack-tox-docs http://logs.openstack.org/10/655010/9/gate/openstack-tox-docs/20a82cd/html/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/10/655010/9/gate/openstack-tox-py36/6f7cc72/ : SUCCESS in 3m 16s\n- openstack-tox-linters http://logs.openstack.org/10/655010/9/gate/openstack-tox-linters/bf00501/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":9},{"id":"3355e476ae566d7aaf5c3208b3dc6864ab931ed3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:49:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"77da85c3baa36dd811ea9eae1dfc1f227c3bc51e","revisions":{"d3039231fd2c2ed43884f06a8b818f9c44d08617":{"kind":"REWORK","_number":1,"created":"2019-04-23 11:38:15.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/10/655010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/10/655010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/10/655010/1"}}},"commit":{"parents":[{"commit":"dffd8467bda7c3c331f7e1db6f2b62f1459df6bf","subject":"Merge \"Elaborate on the business value of Designate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/dffd8467bda7c3c331f7e1db6f2b62f1459df6bf"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:46.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:38:07.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I64ca9232d168f6b8203b7b19cbf7280635dded52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/d3039231fd2c2ed43884f06a8b818f9c44d08617"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/d3039231fd2c2ed43884f06a8b818f9c44d08617"}]},"branch":"refs/heads/master"},"57dc2ef31b965d492d28773c3b6ef1344430d016":{"kind":"REWORK","_number":2,"created":"2019-04-26 09:54:09.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/10/655010/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/10/655010/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/10/655010/2"}}},"commit":{"parents":[{"commit":"dffd8467bda7c3c331f7e1db6f2b62f1459df6bf","subject":"Merge \"Elaborate on the business value of Designate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/dffd8467bda7c3c331f7e1db6f2b62f1459df6bf"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:46.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-26 09:53:57.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I64ca9232d168f6b8203b7b19cbf7280635dded52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/57dc2ef31b965d492d28773c3b6ef1344430d016"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/57dc2ef31b965d492d28773c3b6ef1344430d016"}]},"branch":"refs/heads/master"},"d2e942f0292abf34d540d4291446b6f0ba0c0ca4":{"kind":"REWORK","_number":3,"created":"2019-04-26 09:59:19.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/10/655010/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/10/655010/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/10/655010/3"}}},"commit":{"parents":[{"commit":"dffd8467bda7c3c331f7e1db6f2b62f1459df6bf","subject":"Merge \"Elaborate on the business value of Designate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/dffd8467bda7c3c331f7e1db6f2b62f1459df6bf"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:46.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-26 09:59:01.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I64ca9232d168f6b8203b7b19cbf7280635dded52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/d2e942f0292abf34d540d4291446b6f0ba0c0ca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/d2e942f0292abf34d540d4291446b6f0ba0c0ca4"}]},"branch":"refs/heads/master"},"ebc32a5c9316dfa3daae123d7f3b5c3286fe00f3":{"kind":"REWORK","_number":4,"created":"2019-05-27 02:54:42.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/10/655010/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/10/655010/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/10/655010/4"}}},"commit":{"parents":[{"commit":"aa8e0ba19723269e1369e274128f801bb43e67c0","subject":"Merge \"Remove source license from generated doc output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/aa8e0ba19723269e1369e274128f801bb43e67c0"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:46.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-05-27 02:54:28.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I64ca9232d168f6b8203b7b19cbf7280635dded52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/ebc32a5c9316dfa3daae123d7f3b5c3286fe00f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/ebc32a5c9316dfa3daae123d7f3b5c3286fe00f3"}]},"branch":"refs/heads/master"},"c2ef00904afcba81cdb9dcaadbbf118c34b2fe7e":{"kind":"REWORK","_number":5,"created":"2019-05-27 02:57:13.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/10/655010/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/10/655010/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/10/655010/5"}}},"commit":{"parents":[{"commit":"aa8e0ba19723269e1369e274128f801bb43e67c0","subject":"Merge \"Remove source license from generated doc output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/aa8e0ba19723269e1369e274128f801bb43e67c0"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:46.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-05-27 02:57:04.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I64ca9232d168f6b8203b7b19cbf7280635dded52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c2ef00904afcba81cdb9dcaadbbf118c34b2fe7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c2ef00904afcba81cdb9dcaadbbf118c34b2fe7e"}]},"branch":"refs/heads/master"},"6d4326798352d09f76c7901735564328e0289dfa":{"kind":"REWORK","_number":6,"created":"2019-05-27 03:05:39.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/10/655010/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/10/655010/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/10/655010/6"}}},"commit":{"parents":[{"commit":"aa8e0ba19723269e1369e274128f801bb43e67c0","subject":"Merge \"Remove source license from generated doc output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/aa8e0ba19723269e1369e274128f801bb43e67c0"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:46.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-05-27 03:05:29.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I64ca9232d168f6b8203b7b19cbf7280635dded52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/6d4326798352d09f76c7901735564328e0289dfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/6d4326798352d09f76c7901735564328e0289dfa"}]},"branch":"refs/heads/master"},"bb0469cbd18cdb673361cc6526080186fb80b88f":{"kind":"REWORK","_number":7,"created":"2019-06-25 09:31:25.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/10/655010/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/10/655010/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/10/655010/7"}}},"commit":{"parents":[{"commit":"f39811d1369798f94cf65c4c7cccff72d87a291f","subject":"Merge \"Retire xstatic-vis and xstatic-angular-ui-router projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f39811d1369798f94cf65c4c7cccff72d87a291f"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:46.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-06-25 09:31:11.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I64ca9232d168f6b8203b7b19cbf7280635dded52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/bb0469cbd18cdb673361cc6526080186fb80b88f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/bb0469cbd18cdb673361cc6526080186fb80b88f"}]},"branch":"refs/heads/master"},"e66c1c41bc61a9c9731128326ec524d83236b500":{"kind":"REWORK","_number":8,"created":"2019-06-25 09:32:48.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/10/655010/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/10/655010/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/10/655010/8"}}},"commit":{"parents":[{"commit":"f39811d1369798f94cf65c4c7cccff72d87a291f","subject":"Merge \"Retire xstatic-vis and xstatic-angular-ui-router projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f39811d1369798f94cf65c4c7cccff72d87a291f"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:46.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-06-25 09:32:40.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I64ca9232d168f6b8203b7b19cbf7280635dded52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/e66c1c41bc61a9c9731128326ec524d83236b500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/e66c1c41bc61a9c9731128326ec524d83236b500"}]},"branch":"refs/heads/master"},"77da85c3baa36dd811ea9eae1dfc1f227c3bc51e":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-06-27 09:26:19.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/10/655010/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/10/655010/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/10/655010/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/10/655010/9"}}},"commit":{"parents":[{"commit":"f91f1d9270c06b46aab52bb64625fd620fa07b47","subject":"Only get git timestamp for generated files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f91f1d9270c06b46aab52bb64625fd620fa07b47"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:46.000000000","tz":480},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-27 09:26:19.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I64ca9232d168f6b8203b7b19cbf7280635dded52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/77da85c3baa36dd811ea9eae1dfc1f227c3bc51e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/77da85c3baa36dd811ea9eae1dfc1f227c3bc51e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
