)]}'
{"id":"openstack%2Fgovernance~657141","triplet_id":"openstack%2Fgovernance~master~If7ad13546aae99d97562e99dda5b4e114e0659ac","project":"openstack/governance","branch":"master","topic":"formal-vote","hashtags":[],"change_id":"If7ad13546aae99d97562e99dda5b4e114e0659ac","subject":"Initiate process of migrating Documentation project to a SIG","status":"MERGED","created":"2019-05-04 17:24:37.000000000","updated":"2019-07-09 16:47:00.000000000","submitted":"2019-07-09 16:47:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657141-1562690820809-9c139318","meta_rev_id":"45b2b7934b857f146b114ed5c057500ba147d080","_number":657141,"virtual_id_number":657141,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-07-09 16:47:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":1,"date":"2019-06-28 17:20:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2019-06-28 17:20:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2019-06-28 17:41:54.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":1,"date":"2019-06-28 17:20:48.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2019-06-28 17:20:48.000000000","_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-06-28 17:20:48.000000000","_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":1,"date":"2019-06-28 17:20:48.000000000","_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2019-06-28 17:20:48.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-28 17:20:48.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-06-28 17:20:48.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-06-28 17:20:48.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-06-28 17:20:48.000000000","_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":1,"date":"2019-06-28 17:20:48.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-06-28 17:20:48.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-06-28 17:20:48.000000000","_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-04 17:27:36.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-05-06 17:57:15.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-05-08 15:29:04.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-09 15:40:03.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-05-13 10:05:10.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-13 12:25:08.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-05-13 14:02:13.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-05-20 07:27:07.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2019-06-07 08:58:54.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-06-13 15:34:45.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-13 15:35:24.000000000","updated_by":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"reviewer":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"state":"REVIEWER"},{"updated":"2019-06-24 18:37:17.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-06-25 06:57:30.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-06-28 17:41:54.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-07-09 16:47:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba27f5eaaa16515df5a8f3ffbd74fb097c59de84","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-04 17:24:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bce53381bef62cf3651a60c4ea380ee2471c2731","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-04 17:27:36.000000000","message":"Patch Set 1: Code-Review+1\n\nOslo is onboard with this.","accounts_in_message":[],"_revision_number":1},{"id":"85974dba4532bfd7e3d4104d8cedc45e1a654805","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 17:33:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/41/657141/1/check/openstack-tox-docs/fc77a7e/html/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/41/657141/1/check/openstack-tox-py36/473addd/ : SUCCESS in 7m 20s\n- openstack-tox-linters http://logs.openstack.org/41/657141/1/check/openstack-tox-linters/bb53df4/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e5fa248b7b89601c802831402f5465e17420fe85","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-06 17:57:15.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"66a8fc8686a0eab29f42d89bf574affc4e490f46","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-05-06 20:38:15.000000000","message":"Patch Set 1: Rollcall-Vote+1\n\nI agree that it makes sense for the Oslo team to get involved in maintaining these tools.","accounts_in_message":[],"_revision_number":1},{"id":"c4d2f2884927d917b7c14c4d1313b9217c2bf4fe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-08 14:50:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fe6f8de39510a82b485664255d99bfa7bcc5320b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 15:00:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/41/657141/2/check/openstack-tox-docs/2b57a65/html/ : SUCCESS in 6m 36s\n- openstack-tox-py36 http://logs.openstack.org/41/657141/2/check/openstack-tox-py36/a628888/ : SUCCESS in 6m 29s\n- openstack-tox-linters http://logs.openstack.org/41/657141/2/check/openstack-tox-linters/49deb15/ : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":2},{"id":"cf633309c61a2b21932c41d70015ff74707c3634","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-08 15:29:04.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\nOK with both PTLs, makes sense.","accounts_in_message":[],"_revision_number":2},{"id":"20a03b7b0f605f1ee10a31f66284491cb024e966","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-05-09 15:40:03.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"ca87114f392e1e5503caf032588a142f9363de05","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-13 10:05:10.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\nTo clarify: I agree on the transfer of ownership. For the migration to a SIG vs project team, I believe this has to be discussed in a different patch. It\u0027s not because this patch is accepted, that every patch in the series will be. I believe the docs are very important part of the delivery of an openstack release, and I want to ensure this is not eventually dropped into oblivion :D","accounts_in_message":[],"_revision_number":2},{"id":"29630978c4b0385a19a56b97e90ab95812a131cb","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-05-13 12:25:08.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\nJust worth noting the discussion here: https://review.opendev.org/#/c/657142/\n\nDo we want to begin the process of writing up something to the ML to get a good discussion going and make sure the commmunity is in agreement?\n\nPlease note, I\u0027m in favor of this change.","accounts_in_message":[],"_revision_number":2},{"id":"37a5c28895956056013c125f4114e7fcbf6d3b8f","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-05-13 14:02:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"da367bea4aba7cf471cc208cd4b37ed26659082f","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-05-20 07:27:07.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"b787059cb93954fea72ab80921e4597e388fb107","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-22 13:51:35.000000000","message":"Topic changed from docs-project-to-sig to formal-vote","accounts_in_message":[],"_revision_number":2},{"id":"5493ada9059f520cc1c22444ed0575214403ae59","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-07 08:58:54.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\nlgtm. thanks stephenfin and bnemec.","accounts_in_message":[],"_revision_number":2},{"id":"ba148f681c79184f6f9de482a99f6c8ef17418f0","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-06-10 07:37:25.000000000","message":"Patch Set 2:\n\nremember to register a SIG under https://opendev.org/openstack/governance-sigs/src/branch/master/sigs.yaml  unless Operation Docs SIG is the SIG you plan to added in. On the other hand, if Operation Docs SIG is not the SIG you plan to join, maybe there can be some discussion triggered in ML to collect feedbacks before creat a new one, like define what\u0027s different between two groups","accounts_in_message":[],"_revision_number":2},{"id":"2f13a8556b456a385d74ecc0bc2193db701a07cf","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-13 15:34:45.000000000","message":"Patch Set 2: Rollcall-Vote+1 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"699f8348464835331ce2461c86489d7252ed5caa","author":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"date":"2019-06-13 15:35:24.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"49cfc3578848c3f32bfcc8e0eaac89a7e3bee807","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-13 15:45:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"efccc0b87109da6bd6549818510285564d9b96f9","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-13 15:49:51.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"3649d91aec174fa82f65e31533a0653453289619","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-13 15:51:11.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"b3ee495437aa9067b4a8f6314ab9e96fc540274a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 16:08:08.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"c3024aa3bb6c6f726f60bc4cba7e25787e9141ea","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-06-24 18:37:17.000000000","message":"Patch Set 5: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":5},{"id":"68f0b6595b6c0d60babd3b63676279a0bb606c9b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-06-25 06:57:30.000000000","message":"Patch Set 5: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":5},{"id":"b9bb5352d28d6779ec3db3eba54831b0c54f7d8d","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-26 13:26:55.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"424fdaf8597e0b1c7245e006a54970a524a047ca","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-26 13:27:06.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"5d06260805e2eb681736596e122526ce95060341","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 13:42:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/41/657141/5/check/openstack-tox-docs/6ca34d1/ : FAILURE in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/41/657141/5/check/openstack-tox-py36/2bd2c74/ : SUCCESS in 3m 07s\n- openstack-tox-linters http://logs.openstack.org/41/657141/5/check/openstack-tox-linters/aa6d69f/ : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":5},{"id":"fcb5a60462ec5e23e6e7a2afc0345c68cffed481","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-26 21:55:25.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b0202d86e696f8af30f9cd90b8d502b6e7014f8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 22:13:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/41/657141/5/check/openstack-tox-docs/875f737/ : FAILURE in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/41/657141/5/check/openstack-tox-py36/814a845/ : SUCCESS in 3m 05s\n- openstack-tox-linters http://logs.openstack.org/41/657141/5/check/openstack-tox-linters/fc31f05/ : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":5},{"id":"f474272bac5b0b03f3aa0e61bd60136bcc93af80","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-28 17:20:48.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"fba68546a67098f8d2072cd936e2a52adffc328f","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-28 17:21:10.000000000","message":"Patch Set 6:\n\nrebased to place it in series properly with the previous patch","accounts_in_message":[],"_revision_number":6},{"id":"251b890b3f4f3f8ecbabbc07a4234c187c44edd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 17:30:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/41/657141/6/check/openstack-tox-docs/350133b/html/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/41/657141/6/check/openstack-tox-py36/519ad6a/ : SUCCESS in 3m 32s\n- openstack-tox-linters http://logs.openstack.org/41/657141/6/check/openstack-tox-linters/3a9b093/ : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":6},{"id":"ad77da6bb9ee2b09d949b2bd585d8db4d2786891","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-28 17:41:54.000000000","message":"Patch Set 6: Workflow+1\n\nRe-applying W+1 after rebasing with no content changes","accounts_in_message":[],"_revision_number":6},{"id":"1d656f7a41fb4088dee8ce11bcc251d991afdc02","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-28 17:42:31.000000000","message":"Topic changed from 657142 to formal-vote","accounts_in_message":[],"_revision_number":6},{"id":"5ed62221c1d08faf1dcee91401195d622aa6b90f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 16:35:17.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"536bdc66f6f3ad01db6d2c71f784c80fb48ac8d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 16:47:00.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/41/657141/6/gate/openstack-tox-docs/0c1cca9/html/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/41/657141/6/gate/openstack-tox-py36/285aaca/ : SUCCESS in 4m 24s\n- openstack-tox-linters http://logs.openstack.org/41/657141/6/gate/openstack-tox-linters/25c47ad/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":6},{"id":"45b2b7934b857f146b114ed5c057500ba147d080","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 16:47:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"0134af1b6c54fafb1d4743d20d9733171b1258b4","revisions":{"82d40f0201d3d4dbec5153dabf4036bf0ee5515c":{"kind":"REWORK","_number":1,"created":"2019-05-04 17:24:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/41/657141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/41/657141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/41/657141/1"}}},"commit":{"parents":[{"commit":"2a35b6e7815d2293b7830c520d62474b1eeeb9aa","subject":"Retire docs-specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/2a35b6e7815d2293b7830c520d62474b1eeeb9aa"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-04 16:35:39.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-04 17:24:33.000000000","tz":-360},"subject":"Initiate process of migrating Documentation project to a SIG","message":"Initiate process of migrating Documentation project to a SIG\n\nThe first step of this is moving the code deliverables to Oslo. This has\nalready been discussed with the Oslo PTL.\n\nChange-Id: If7ad13546aae99d97562e99dda5b4e114e0659ac\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/82d40f0201d3d4dbec5153dabf4036bf0ee5515c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/82d40f0201d3d4dbec5153dabf4036bf0ee5515c"}]},"branch":"refs/heads/master"},"16cd2e483c3cea8c3ee56a82b1239c6dce42b9eb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-08 14:50:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/41/657141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/41/657141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/41/657141/2"}}},"commit":{"parents":[{"commit":"d2ca9608b62ba98f1e213784d7ed9e5b0aeabd9b","subject":"Retire docs-specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/d2ca9608b62ba98f1e213784d7ed9e5b0aeabd9b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-04 16:35:39.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-08 14:50:10.000000000","tz":0},"subject":"Initiate process of migrating Documentation project to a SIG","message":"Initiate process of migrating Documentation project to a SIG\n\nThe first step of this is moving the code deliverables to Oslo. This has\nalready been discussed with the Oslo PTL.\n\nChange-Id: If7ad13546aae99d97562e99dda5b4e114e0659ac\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/16cd2e483c3cea8c3ee56a82b1239c6dce42b9eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/16cd2e483c3cea8c3ee56a82b1239c6dce42b9eb"}]},"branch":"refs/heads/master"},"2ac41cfe4f3e59c14f0310d05388791e42c14f19":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-13 15:45:40.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/41/657141/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/41/657141/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/41/657141/3"}}},"commit":{"parents":[{"commit":"46d0a4d827d25f3b3bf9a9495cff84d680b52828","subject":"Retire docs-specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/46d0a4d827d25f3b3bf9a9495cff84d680b52828"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-04 16:35:39.000000000","tz":-360},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2019-06-13 15:45:32.000000000","tz":-240},"subject":"Initiate process of migrating Documentation project to a SIG","message":"Initiate process of migrating Documentation project to a SIG\n\nThe first step of this is moving the code deliverables to Oslo. This has\nalready been discussed with the Oslo PTL.\n\nChange-Id: If7ad13546aae99d97562e99dda5b4e114e0659ac\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/2ac41cfe4f3e59c14f0310d05388791e42c14f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/2ac41cfe4f3e59c14f0310d05388791e42c14f19"}]},"branch":"refs/heads/master"},"81e4670abe8016c567f21362e666e7b7499e6b4f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-06-13 15:49:51.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/41/657141/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/41/657141/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/41/657141/4"}}},"commit":{"parents":[{"commit":"6e09dd39b25ed67ad58fe53987b83a6092055af3","subject":"Retire docs-specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/6e09dd39b25ed67ad58fe53987b83a6092055af3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-04 16:35:39.000000000","tz":-360},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2019-06-13 15:49:37.000000000","tz":-240},"subject":"Initiate process of migrating Documentation project to a SIG","message":"Initiate process of migrating Documentation project to a SIG\n\nThe first step of this is moving the code deliverables to Oslo. This has\nalready been discussed with the Oslo PTL.\n\nChange-Id: If7ad13546aae99d97562e99dda5b4e114e0659ac\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/81e4670abe8016c567f21362e666e7b7499e6b4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/81e4670abe8016c567f21362e666e7b7499e6b4f"}]},"branch":"refs/heads/master"},"cd26ec5eea82d27c6112b842e0f15ebc96ee7e04":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-13 15:51:11.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/41/657141/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/41/657141/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/41/657141/5"}}},"commit":{"parents":[{"commit":"4df29cea8a4e04faa3d32d2c3d406245b54ddf86","subject":"Retire docs-specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/4df29cea8a4e04faa3d32d2c3d406245b54ddf86"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-04 16:35:39.000000000","tz":-360},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2019-06-13 15:51:04.000000000","tz":-240},"subject":"Initiate process of migrating Documentation project to a SIG","message":"Initiate process of migrating Documentation project to a SIG\n\nThe first step of this is moving the code deliverables to Oslo. This has\nalready been discussed with the Oslo PTL.\n\nChange-Id: If7ad13546aae99d97562e99dda5b4e114e0659ac\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/cd26ec5eea82d27c6112b842e0f15ebc96ee7e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/cd26ec5eea82d27c6112b842e0f15ebc96ee7e04"}]},"branch":"refs/heads/master"},"0134af1b6c54fafb1d4743d20d9733171b1258b4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-06-28 17:20:48.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/41/657141/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/41/657141/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/41/657141/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/41/657141/6"}}},"commit":{"parents":[{"commit":"d28c309692bdb8fe86eba203bbc475d7bcfde223","subject":"Retire docs-specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/d28c309692bdb8fe86eba203bbc475d7bcfde223"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-04 16:35:39.000000000","tz":-360},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2019-06-28 17:20:44.000000000","tz":-240},"subject":"Initiate process of migrating Documentation project to a SIG","message":"Initiate process of migrating Documentation project to a SIG\n\nThe first step of this is moving the code deliverables to Oslo. This has\nalready been discussed with the Oslo PTL.\n\nChange-Id: If7ad13546aae99d97562e99dda5b4e114e0659ac\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/0134af1b6c54fafb1d4743d20d9733171b1258b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/0134af1b6c54fafb1d4743d20d9733171b1258b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
