)]}'
{"id":"openstack%2Fgovernance~657154","triplet_id":"openstack%2Fgovernance~master~I333393e0304844e30983f89c82eabcad11487be2","project":"openstack/governance","branch":"master","topic":"project-update","hashtags":[],"change_id":"I333393e0304844e30983f89c82eabcad11487be2","subject":"Add liberasurecode and pyeclib as Swift team deliverables","status":"MERGED","created":"2019-05-04 19:45:59.000000000","updated":"2019-05-31 15:47:49.000000000","submitted":"2019-05-30 15:22:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657154-1559229729361-45b1259d","meta_rev_id":"b3c7d73efef3bfd9079a10c5257db87e4ddc9dea","_number":657154,"virtual_id_number":657154,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-05-30 15:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":0,"date":"2019-05-28 21:20:11.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"date":"2019-05-29 14:02:47.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-05-30 15:10:43.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":1,"date":"2019-05-28 23:57:00.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2019-05-30 15:06:52.000000000","_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-05-30 15:14:10.000000000","_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":1,"date":"2019-05-29 19:36:38.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-29 17:41:32.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-05-30 12:16:12.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-05-30 15:06:18.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":1,"date":"2019-05-29 21:59:31.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-05-29 13:52:39.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-05-30 15:09:26.000000000","_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 08:31:13.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-05-28 23:57:00.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-05-29 13:52:39.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-05-29 14:02:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-29 17:41:32.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-29 19:36:38.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-05-29 21:59:31.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-05-30 15:06:18.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-05-30 15:06:52.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-30 15:09:26.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2019-05-30 15:10:43.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-30 15:14:10.000000000","updated_by":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"reviewer":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"state":"REVIEWER"},{"updated":"2019-05-30 15:22:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ca1e9770e3b461f7b3959671b43a25fe0d48be8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 19:45:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f014a2bcdb2feae22433a7c151cdeeec89c4cb7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 19:49:43.000000000","message":"Patch Set 1:\n\nSo I\u0027m not sure how the dependency should work, but the corresponding project-config change is at https://review.opendev.org/#/c/657155/","accounts_in_message":[],"_revision_number":1},{"id":"d348eba2bb3e51d7a2fe6cf684d365fff3825abb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 19:51:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/54/657154/1/check/openstack-tox-docs/c002246/html/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/54/657154/1/check/openstack-tox-py36/b7379f7/ : SUCCESS in 4m 15s\n- openstack-tox-linters http://logs.openstack.org/54/657154/1/check/openstack-tox-linters/fb00fcc/ : FAILURE in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"fd75a2be6b4880f948d44abb36cd63bc38c0359b","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-06 17:17:39.000000000","message":"Patch Set 1:\n\nYou should add a Depends-On in this patch for the project config change, but IIRC the CI will have to be rechecked once that passes anyway.","accounts_in_message":[],"_revision_number":1},{"id":"7452f9947ba58b18ab306fcc6b690f575ff5a385","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-06 23:49:04.000000000","message":"Patch Set 1:\n\nHuh. A reviewer for the project-config change suggested I set the dependency in the *other* direction... *shrug*","accounts_in_message":[],"_revision_number":1},{"id":"d937cdfab9280a22b1de5a0d8f6957bdf96a9cf5","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-07 13:26:55.000000000","message":"Patch Set 1:\n\nHmm, I was going by https://docs.openstack.org/infra/manual/creators.html#add-new-repository-to-the-governance-repository which is for new projects, but I don\u0027t know what the story is with renames - they\u0027ve only just now become a thing again. It\u0027s possible we have a circular dependency to break.","accounts_in_message":[],"_revision_number":1},{"id":"0dc6881a80ec593eb63ff352d2299237f8d0363f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-09 19:03:57.000000000","message":"Patch Set 1:\n\nWell, since the project-config change still couldn\u0027t get processed, couldn\u0027t hurt for me to try it. Will push up a new patch.","accounts_in_message":[],"_revision_number":1},{"id":"aed839897a12f00da1af340bc5c25f26d4ca41a1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-09 19:04:16.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f1be7f2f83cac3ff49a55d915d228d8d550f4c1f","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-05-10 08:31:13.000000000","message":"Patch Set 2:\n\nLGTM - thanks Tim. Waiting for the gates.","accounts_in_message":[],"_revision_number":2},{"id":"450a0719cb5c54b9db64893e6c7adff7f75157b3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-10 22:40:10.000000000","message":"Patch Set 2:\n\nrecheck\n\nI think the check job never triggered because of a circular dep. Dropped the depends-on in the project-config change, and we\u0027ll see how this behaves...","accounts_in_message":[],"_revision_number":2},{"id":"2efecc7c8b63650e31e43f89dfd232641faa79c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 22:40:29.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project \"openstack/pyeclib\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/pyeclib\n      templates:\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 4350, column 3","accounts_in_message":[],"_revision_number":2},{"id":"1562586ce297f68c6ae191322076bd5f8fa5da65","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-13 09:48:00.000000000","message":"Patch Set 2:\n\n\u003e recheck\n \u003e \n \u003e I think the check job never triggered because of a circular dep.\n \u003e Dropped the depends-on in the project-config change, and we\u0027ll see\n \u003e how this behaves...\n\nNo clue. it would indeed be weird if there are things that need removal, like the zuul files, for re-adding them later...","accounts_in_message":[],"_revision_number":2},{"id":"12d2394000b27e1742b5976f33eee770fc7a659a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-13 12:07:47.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2502383cb90574439a4fd0f4c671c928462583e3","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-13 12:08:21.000000000","message":"Patch Set 2: -Rollcall-Vote","accounts_in_message":[],"_revision_number":2},{"id":"c8a609ce01e22cb44417a822abe9fcd5512a18a9","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-22 13:46:51.000000000","message":"Topic set to project-update","accounts_in_message":[],"_revision_number":2},{"id":"37757903f86c76d2a09e103ef74d6577e03b2fc9","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-27 08:18:59.000000000","message":"Patch Set 2:\n\nI suggest moving them in with x/ and once the repo is renamed by infra (after this lands), you can propose another change that updates the repo name.","accounts_in_message":[],"_revision_number":2},{"id":"badf778146fca59520ffc3343a1186115de4a55a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-27 08:19:03.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"f4c6f929ee26787dd9da06a2317d055a521fe596","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-28 14:29:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"517ccd7c619ff4ff84e67f23801a5583ddd378af","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-28 14:30:28.000000000","message":"Patch Set 3:\n\nI\u0027ll leave the ordering up to you guys. Done.","accounts_in_message":[],"_revision_number":3},{"id":"caaf8b71c97a665f957bd18975877515c4797d3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 14:50:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/54/657154/3/check/openstack-tox-docs/640cd8a/html/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/54/657154/3/check/openstack-tox-py36/2ad85e8/ : SUCCESS in 4m 35s\n- openstack-tox-linters http://logs.openstack.org/54/657154/3/check/openstack-tox-linters/b39177e/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":3},{"id":"6811c49e9a5868faf6de6392297ca931bba561fe","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-05-28 15:06:22.000000000","message":"Patch Set 3:\n\nFood for thought... liberasurecode is written predominantly in C with typical GNU Automake build toolchain, and pyeclib is non-pure-Python with C extensions (unsurprisingly I suppose, it links liberasurecode). Don\u0027t get me wrong, I think this is great and I don\u0027t think these are reasons not to have them under the scope of OpenStack governance, but these are firsts for us so I want to make sure we think through the policy implications.\n\nFirst, I assume we build and test these (because if they\u0027re not tested they\u0027re broken!), but for other languages we document a standard set of project testing interfaces and supported runtimes:\n\nhttps://governance.openstack.org/tc/reference/project-testing-interface.html\n\nDo we need to amend that information, perhaps documenting whatever these two projects do as our initial \"standard\" for C libraries and Python libraries with C extensions?\n\nSecond, do we expect to distribute either of these in compiled form or only in source form? If we\u0027re shipping builds of either, then our 2017 resolution on binary artifact release management is probably relevant:\n\nhttps://governance.openstack.org/tc/resolutions/20170530-binary-artifacts.html\n\nIt was written with things like container images and distro packages in mind, but most of the same concerns likely apply here unless these projects are being distributed as source only. It would be good for the maintainers to review that resolution and let us know whether the guidelines are applicable or whether we need a new resolution superseding it to also provide appropriate guidance for their situation.","accounts_in_message":[],"_revision_number":3},{"id":"c54a06ee159de7044ca379a55136d56b873cefb1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-28 21:20:11.000000000","message":"Patch Set 3:\n\nJust to establish/highlight my perspective, my first priority is (must always be) the consumers of the code we\u0027ve written. That covers end-users, operators, deployers, packagers, other developers, and everyone in between -- and for their sake, I owe them at minimum the answers to a few questions like\n\n- Where do I go to get help?\n- Where do I report bugs?\n- Where can I find your latest release?\n\nWhen these projects were first brought under the openstack namespace [1], these answers changed. This was disruptive (there were still new issues getting filed against the old projects for nearly a year after the transition [2]), but worth it to move away from the original authors\u0027 personal bitbucket repositories to something more community-oriented.\n\nThese were brought in as unofficial projects deliberately, for a few reasons:\n\n- We knew they could be useful beyond OpenStack.\n- We were hopeful that an independent community could form around them.\n- We did not want to push the Technical Committee to establish broad new policies based on a single project.\n\nWith the OpenDev migration [3] and subsequent halting of GitHub mirroring for unofficial projects [4], our answers for those key questions are changing again, which is less than desirable. Just last week I had a packager wondering why he wasn\u0027t seeing the latest tag where he expected [5].\n\nMy favored resolution to that is to move them back to the openstack namespace; it has been the canonical location for the past three years and it should continue to be. The simplest way to accomplish that (in my mind) is to reflect reality: the Swift community owns and maintains these projects. While we have had some interest from other developers, I would not say an independent community has formed. So, make the projects official Swift deliverables, rename the projects, and then the names and URLs from two months ago work again.\n\nPerhaps this is the wrong approach -- I wasn\u0027t intending to create a lot of governance work or change how we test and deliver code. Certainly, I would hesitate to recommend OpenStack-wide policies based upon a single sample. Still, in light of the contributor history and lacking a better organization under which these should live, returning to the status quo seemed to make sense. Again, I just want to ensure existing links work and accurately reflect the community that\u0027s driving development.\n\nRegarding the called-out policy implications:\n\nYes, these are tested. liberasurecode has two make targets (test and valgrind-test) that are run sequentially in the gate on both ubuntu and centos 7. pyeclib uses tox to run tests with latest distro packages of liberasurecode under py27, py35, py36, and py37; there are further tests that run against tip-of-master liberasurecode.\n\nAs far as release artifacts, liberasurecode just has tags in a git repository -- we let distro packagers handle builds for various architectures. pyeclib uploads sdists and x86_64 wheels to tarballs.openstack.org [6], as well as sdists to PyPI [7]. We needed to specifically exclude the wheels from being uploaded to PyPI [8] which would reject them. At first look, I don\u0027t see any divergences from the binary-artifacts resolution for those wheels, but if it would simplify matters, I\u0027m fairly certain we\u0027d be content to just stop producing them -- the recommended ways to install pyeclib are (in rough order of freshness):\n\n- from distro packages,\n- via pip (which will download and compile the sdist from PyPI), or\n- from source, typically as tip-of-master.\n\n[1] https://review.opendev.org/#/c/314724/\n[2] https://bitbucket.org/kmgreen2/pyeclib/issues/\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/005011.html\n[4] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/005629.html\n[5] http://eavesdrop.openstack.org/irclogs/%23openstack-swift/%23openstack-swift.2019-05-24.log.html#t2019-05-24T16:59:34\n[6] http://tarballs.openstack.org/pyeclib/\n[7] https://pypi.org/project/pyeclib/#files\n[8] https://review.opendev.org/#/c/649455/","accounts_in_message":[],"_revision_number":3},{"id":"d1126d74816f543ea93d4f368cd47fb5fc00aec1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-05-28 21:49:13.000000000","message":"Patch Set 3:\n\nThanks! Policy wise, it sounds to me like the most we\u0027d need to do to accommodate these is to simply document that we use the GCC/Automake toolchain versions for our target distros in the PTI as our tested runtime, and I don\u0027t even feel like that needs to be done before we accept these project additions (but I\u0027ve no doubt some other TC members will weigh in on that too).\n\nIt wasn\u0027t immediately apparent that the loss of GitHub mirroring from the OpenDev migration was the initial motivation for this change, and I feel compelled to point out that there are solutions. Projects like https://github.com/openstack/ara-web/ and https://github.com/openstack/airship-shipyard/ worked with the OpenDev sysadmin team to \"transfer\" them to a different GH org where they have admin rights, and each adds a single $project-upload-git-mirror zuul job to their post pipelines (I\u0027d recommend the release pipeline too) which updates that mirror continuously.","accounts_in_message":[],"_revision_number":3},{"id":"8ab51b6a5f68a05309352c9da0d288e887ad1b95","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-28 23:57:00.000000000","message":"Patch Set 3: Rollcall-Vote+1\n\nPolicy-wise, I think pycelib is a no-brainer: you install it with pip; it\u0027s a Python project for all intents and purposes.\n\nliberasurecode is slightly different. We depend on lots of C libraries in OpenStack, but in most cases we don\u0027t actually maintain them. In this case, however, the Swift team is actually the owner of this project. I\u0027m a believer in governance documents reflecting reality on the ground, so if this is the path the Swift team want to take I have no objection - in particular I\u0027m not concerned that people will take this as precedent for rewriting OpenStack in C ;)\n\nOf course if the Swift team wanted to continue to pursue the path of trying to form a standalone community around this by creating a separate namespace on OpenDev (and mirroring it to GitHub, with appropriate redirects from the old openstack namespace), then I\u0027d be fine with that as well.","accounts_in_message":[],"_revision_number":3},{"id":"75c18290527c33fde892d019f01552a30bcd14d8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-05-29 13:52:39.000000000","message":"Patch Set 3: Rollcall-Vote+1\n\nAgree we might want to adjust some documents to document how to do similar cases in the future, but otherwise in this case it sounds like a more accurate reflection of the current reality (that those repos are maintained by the Swift team, as a part of Swift), so +1","accounts_in_message":[],"_revision_number":3},{"id":"4e9f23af3ed3deb2b6fa86cc80348b563dc5378f","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-29 17:41:32.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"a2a37fd50556a08f3363c3eda6adf2d21ff87b13","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-05-29 19:36:38.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"ef54cc81d99f3b7533789d40ed8bd1cf14ff5014","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-05-29 21:59:31.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"41f37cb96a6f97bdfaf827430ebc71a89f78e703","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-30 12:16:12.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"0c002aa5f8d0c1cb6e7a6be56859e90c631863ee","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-05-30 15:06:18.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"adda9aa881b2f436ce2017c6c2a112b43719d5ab","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-30 15:06:52.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"f2d3d2463f1cc40e952e844fea86623e6bc90ca4","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-05-30 15:09:26.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"79a8a686442c67d631b4fc68b922b5366a2943aa","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-30 15:10:43.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f7bc54d085b8943751f8678fffc14dc0fa404904","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 15:10:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a5ef00e35d1e563cb82b3ca5b84500d73e58f6b","author":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"date":"2019-05-30 15:14:10.000000000","message":"Patch Set 3: Rollcall-Vote+1\n\nAgree this reflects reality :)","accounts_in_message":[],"_revision_number":3},{"id":"ab87f322128cc95a255eee923a58c37c09383866","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 15:22:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/54/657154/3/gate/openstack-tox-docs/2db0ac9/html/ : SUCCESS in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/54/657154/3/gate/openstack-tox-py36/1494cd0/ : SUCCESS in 4m 29s\n- openstack-tox-linters http://logs.openstack.org/54/657154/3/gate/openstack-tox-linters/6da612f/ : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":3},{"id":"c268c58a8192ddad94bffa36cffffc46d9cb65a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 15:22:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b3c7d73efef3bfd9079a10c5257db87e4ddc9dea","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-31 15:47:49.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"37c376d400947f26bdcaed124e28ce3367c413dd","revisions":{"736891e30edf4d8313d0c00752f8ee920b875f47":{"kind":"REWORK","_number":1,"created":"2019-05-04 19:45:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/657154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/54/657154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/54/657154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/54/657154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/54/657154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/54/657154/1"}}},"commit":{"parents":[{"commit":"173df219ac1f3d384d1c593e35d2a4acabcd7ebf","subject":"Merge \"Remove stable:follows-policy tag for Trove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/173df219ac1f3d384d1c593e35d2a4acabcd7ebf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-04 19:40:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-04 19:44:50.000000000","tz":-420},"subject":"Add liberasurecode and pyeclib as Swift team deliverables","message":"Add liberasurecode and pyeclib as Swift team deliverables\n\nThis more-accurately reflects the realities that:\n\n- the primary developers for these projects are Swift developers and\n- the primary consumer of these projects is Swift\n\nChange-Id: I333393e0304844e30983f89c82eabcad11487be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/736891e30edf4d8313d0c00752f8ee920b875f47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/736891e30edf4d8313d0c00752f8ee920b875f47"}]},"branch":"refs/heads/master"},"095a6ed52dd14fb0b107bc14f0a77638b79bb654":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-09 19:04:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/657154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/54/657154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/54/657154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/54/657154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/54/657154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/54/657154/2"}}},"commit":{"parents":[{"commit":"173df219ac1f3d384d1c593e35d2a4acabcd7ebf","subject":"Merge \"Remove stable:follows-policy tag for Trove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/173df219ac1f3d384d1c593e35d2a4acabcd7ebf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-04 19:40:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-09 19:02:30.000000000","tz":-420},"subject":"Add liberasurecode and pyeclib as Swift team deliverables","message":"Add liberasurecode and pyeclib as Swift team deliverables\n\nThis more-accurately reflects the realities that:\n\n- the primary developers for these projects are Swift developers and\n- the primary consumer of these projects is Swift\n\nChange-Id: I333393e0304844e30983f89c82eabcad11487be2\nDepends-On: I9ec488f04ca119b048b0f47235fc9ffaa4ab7417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/095a6ed52dd14fb0b107bc14f0a77638b79bb654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/095a6ed52dd14fb0b107bc14f0a77638b79bb654"}]},"branch":"refs/heads/master"},"37c376d400947f26bdcaed124e28ce3367c413dd":{"kind":"REWORK","_number":3,"created":"2019-05-28 14:29:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/657154/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/54/657154/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/54/657154/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/54/657154/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/54/657154/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/54/657154/3"}}},"commit":{"parents":[{"commit":"173df219ac1f3d384d1c593e35d2a4acabcd7ebf","subject":"Merge \"Remove stable:follows-policy tag for Trove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/173df219ac1f3d384d1c593e35d2a4acabcd7ebf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-04 19:40:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-28 14:28:31.000000000","tz":-420},"subject":"Add liberasurecode and pyeclib as Swift team deliverables","message":"Add liberasurecode and pyeclib as Swift team deliverables\n\nThis more-accurately reflects the realities that:\n\n- the primary developers for these projects are Swift developers and\n- the primary consumer of these projects is Swift\n\nChange-Id: I333393e0304844e30983f89c82eabcad11487be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/37c376d400947f26bdcaed124e28ce3367c413dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/37c376d400947f26bdcaed124e28ce3367c413dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
