)]}'
{"id":"openstack%2Fgovernance~660448","triplet_id":"openstack%2Fgovernance~master~Id8b5400f27c4d61c2f0feff2abe12c1d940ea72b","project":"openstack/governance","branch":"master","topic":"project-update","hashtags":[],"change_id":"Id8b5400f27c4d61c2f0feff2abe12c1d940ea72b","subject":"Retire ansible-role-k8s-* repos","status":"MERGED","created":"2019-05-21 16:24:24.000000000","updated":"2019-05-31 12:30:49.000000000","submitted":"2019-05-31 12:30:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660448-1559305849835-26cc25ca","meta_rev_id":"f341e1916c4e9293bbc7729c0aeb30156306a969","_number":660448,"virtual_id_number":660448,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"date":"2019-05-21 19:07:51.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-05-31 12:30:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-05-21 17:05:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-05-31 12:18:54.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-05-22 13:33:50.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2019-05-30 15:12:39.000000000","_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-05-22 09:06:16.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-05-27 19:57:42.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-05-21 17:26:55.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-05-27 08:06:33.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 17:05:30.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-05-21 17:26:55.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-22 09:06:16.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-05-22 13:33:50.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-05-27 19:57:42.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-05-30 15:12:39.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-31 12:18:54.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-31 12:30:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e09a9e1d918cc9027c28af4e3a3a342329e4c9ee","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-21 16:24:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0482c54d69bc41b41f780bb4381dd363000f125","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 16:40:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/48/660448/1/check/openstack-tox-docs/f107dbd/html/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/48/660448/1/check/openstack-tox-py36/50e1311/ : SUCCESS in 4m 19s\n- openstack-tox-linters http://logs.openstack.org/48/660448/1/check/openstack-tox-linters/364a553/ : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":1},{"id":"78fcb5ffb58ebaa3a1b8816a01737e5b1bd1f952","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 16:54:37.000000000","message":"Patch Set 1:\n\nI have noticed you have noop jobs (so step 1 and step 2 done), but I think it would be nice if you would put a message in those repos (i.e. follow step 3 [1]) and adapt things in infra (step 4), before processing this.\n\n[1]: https://docs.openstack.org/infra/manual/drivers.html#step-3-remove-project-content.","accounts_in_message":[],"_revision_number":1},{"id":"dca846a7cc4530a7e979a98f9bbced5fd56b3d2a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-05-21 17:05:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"92594e3daa6ed2acd8e47450fb4ac32b2b929c02","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-21 17:10:09.000000000","message":"Patch Set 1:\n\nI have. See depends-on chain.","accounts_in_message":[],"_revision_number":1},{"id":"ca2a09e46f7df27f47f233f5299b7eb8569850d2","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 17:26:55.000000000","message":"Patch Set 1: Rollcall-Vote+1\n\n\u003e I have. See depends-on chain.\n\nI am obviously blind at this end of day.","accounts_in_message":[],"_revision_number":1},{"id":"e7695aa29ae80f0f5b201dda20a19698a2ea6489","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-21 19:07:51.000000000","message":"Patch Set 1:\n\nrecheck :D","accounts_in_message":[],"_revision_number":1},{"id":"325652a4dd1dd60d6bc7f73b41eaa1c504f54095","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 19:19:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/48/660448/1/check/openstack-tox-docs/a8d3096/html/ : SUCCESS in 5m 01s\n- openstack-tox-py36 http://logs.openstack.org/48/660448/1/check/openstack-tox-py36/d4dacbf/ : SUCCESS in 4m 43s\n- openstack-tox-linters http://logs.openstack.org/48/660448/1/check/openstack-tox-linters/f253fdb/ : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a9110292e91cba4080df264c11b55a029cc5b5d3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-05-22 09:06:16.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"3288139052bf1f2e1e4e5401781319ec69873e44","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-22 13:33:01.000000000","message":"Topic changed from retire-ansible-role-k8s to project-update","accounts_in_message":[],"_revision_number":1},{"id":"b836a5e1171f740c19d2363030d3f987c70430d8","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-05-22 13:33:50.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"c1b691a80d2850149709765698eb8574ce29962d","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-27 08:06:33.000000000","message":"Patch Set 1: Rollcall-Vote+1\n\nThis can merge on 2019-05-29.","accounts_in_message":[],"_revision_number":1},{"id":"20fa43f1b0b79ed4a3f7c4589dc9b0205bae5fa1","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-05-27 19:57:42.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"d0e3a9d2da88d4474fd7469073ad769f7654217b","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-30 15:12:39.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"b6319c7e565b82fca57907b6d64b54280dfa9672","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-31 12:18:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2760e1946d9ecf4016437bc80169dcef566c2acc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 12:19:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4018c3bb948f1739c647243fc93a9187c7432de1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 12:30:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f341e1916c4e9293bbc7729c0aeb30156306a969","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 12:30:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/48/660448/1/gate/openstack-tox-docs/68ae2dc/html/ : SUCCESS in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/48/660448/1/gate/openstack-tox-py36/4fea464/ : SUCCESS in 3m 47s\n- openstack-tox-linters http://logs.openstack.org/48/660448/1/gate/openstack-tox-linters/28cb0fe/ : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0f1047dfaa1fdd4331f2d6feda3188e4319308e","revisions":{"a0f1047dfaa1fdd4331f2d6feda3188e4319308e":{"kind":"REWORK","_number":1,"created":"2019-05-21 16:24:24.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/48/660448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/48/660448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/48/660448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/48/660448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/48/660448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/48/660448/1"}}},"commit":{"parents":[{"commit":"173df219ac1f3d384d1c593e35d2a4acabcd7ebf","subject":"Merge \"Remove stable:follows-policy tag for Trove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/173df219ac1f3d384d1c593e35d2a4acabcd7ebf"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-21 16:24:03.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-21 16:24:03.000000000","tz":-360},"subject":"Retire ansible-role-k8s-* repos","message":"Retire ansible-role-k8s-* repos\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-May/006458.html\n\nChange-Id: Id8b5400f27c4d61c2f0feff2abe12c1d940ea72b\nDepends-On: https://review.opendev.org/#/c/660446/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/a0f1047dfaa1fdd4331f2d6feda3188e4319308e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/a0f1047dfaa1fdd4331f2d6feda3188e4319308e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
