)]}'
{"id":"openstack%2Fgovernance~668151","triplet_id":"openstack%2Fgovernance~master~I960c680f1e2bae93fea047a4ef805e6e748dc5b9","project":"openstack/governance","branch":"master","topic":"code-change","hashtags":[],"change_id":"I960c680f1e2bae93fea047a4ef805e6e748dc5b9","subject":"Allow yaml checks to work with mktemp in a different path","status":"MERGED","created":"2019-06-28 11:38:09.000000000","updated":"2019-06-28 18:22:56.000000000","submitted":"2019-06-28 18:22:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668151-1561746176826-cfcade68","meta_rev_id":"b5003b0091a359f53a4c9e01151f8a0f8f9ef6d3","_number":668151,"virtual_id_number":668151,"owner":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-28 18:22:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":1,"date":"2019-06-28 18:12:15.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":1,"date":"2019-06-28 17:04:52.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":1,"date":"2019-06-28 18:12:15.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 17:04:52.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-06-28 18:12:15.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-28 18:22:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57f6bd40067eecc089895a0161244c01771ebadc","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-28 11:38:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80ed71bf7e9441a0ace37320a38c2a7d5b745f4f","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-28 11:41:05.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"40325f24a2e838924da347e854deb8e635461d60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 11:56:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/668151/2/check/openstack-tox-docs/07025e3/html/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/51/668151/2/check/openstack-tox-py36/b411ecf/ : SUCCESS in 3m 31s\n- openstack-tox-linters http://logs.openstack.org/51/668151/2/check/openstack-tox-linters/4926879/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"2b20219190e2c3801c103152e3fa7cc1a04757e0","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-28 17:04:52.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"8db6449f8e212474afecf47f9fb1f76a9841b426","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-28 17:45:59.000000000","message":"Topic changed from linter-update to code-change","accounts_in_message":[],"_revision_number":2},{"id":"76fe69c244b929a0423a99030e6f36eebc777648","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-28 18:12:15.000000000","message":"Patch Set 2: Rollcall-Vote+1 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f8f28c44c3ceba68603ad52589dd88ca2e10132d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 18:12:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d4fb24fa6c59a95f261dd29e373a1a89caf0f01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 18:22:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/668151/2/gate/openstack-tox-docs/73215ce/html/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/51/668151/2/gate/openstack-tox-py36/a57aada/ : SUCCESS in 4m 32s\n- openstack-tox-linters http://logs.openstack.org/51/668151/2/gate/openstack-tox-linters/e293cce/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"b5003b0091a359f53a4c9e01151f8a0f8f9ef6d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 18:22:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aaef9cc8f099111fcbc831c2cc62c64cd33924e6","revisions":{"c461be0eb82dc2edbe3d5fec65af707655ecfd1c":{"kind":"REWORK","_number":1,"created":"2019-06-28 11:38:09.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/51/668151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/51/668151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/51/668151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/51/668151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/51/668151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/51/668151/1"}}},"commit":{"parents":[{"commit":"49c91fcf33e464a34e611b45a89ad6381b4e7452","subject":"Inital random assignment of liasons","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/49c91fcf33e464a34e611b45a89ad6381b4e7452"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-06-28 11:37:19.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-06-28 11:37:19.000000000","tz":60},"subject":"Allow yaml checks to work with mktemp in a different path","message":"Allow yaml checks to work with mktemp in a different path\n\nChange-Id: I960c680f1e2bae93fea047a4ef805e6e748dc5b9\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c461be0eb82dc2edbe3d5fec65af707655ecfd1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c461be0eb82dc2edbe3d5fec65af707655ecfd1c"}]},"branch":"refs/heads/master"},"aaef9cc8f099111fcbc831c2cc62c64cd33924e6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-28 11:41:05.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/51/668151/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/51/668151/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/51/668151/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/51/668151/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/51/668151/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/51/668151/2"}}},"commit":{"parents":[{"commit":"4bff6be36a84372c68e6f186f1beb26186b22539","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/4bff6be36a84372c68e6f186f1beb26186b22539"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-06-28 11:37:19.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-06-28 11:40:52.000000000","tz":60},"subject":"Allow yaml checks to work with mktemp in a different path","message":"Allow yaml checks to work with mktemp in a different path\n\nChange-Id: I960c680f1e2bae93fea047a4ef805e6e748dc5b9\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/aaef9cc8f099111fcbc831c2cc62c64cd33924e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/aaef9cc8f099111fcbc831c2cc62c64cd33924e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
