)]}'
{"id":"openstack%2Fgovernance~668188","triplet_id":"openstack%2Fgovernance~master~Icf454704a1503d27c0c526863bfaf103a7d7f8e1","project":"openstack/governance","branch":"master","topic":"DMN-test-display-of-liasons","hashtags":[],"change_id":"Icf454704a1503d27c0c526863bfaf103a7d7f8e1","subject":"DNM: Display table of TC Liasons and projects","status":"ABANDONED","created":"2019-06-28 15:00:34.000000000","updated":"2019-07-01 10:40:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1123d5cd9a66bf9ba208552ead27383462c7857e","_number":668188,"virtual_id_number":668188,"owner":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-28 15:17:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 15:17:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4282b093eb39a47c9a3070e71c1192e65d314a20","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-28 15:00:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85f90be5aba64e33f02c1db69f777165cbb67a71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 15:17:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/668188/1/check/openstack-tox-docs/461886f/html/ : SUCCESS in 4m 19s\n- openstack-tox-linters http://logs.openstack.org/88/668188/1/check/openstack-tox-linters/41a3bf7/ : FAILURE in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"61065fdf1b23149579cb50ca302b3aff29359cfc","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-07-01 10:40:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9261e2477742f3bd2a7d93fff589852f75051be5","revisions":{"9261e2477742f3bd2a7d93fff589852f75051be5":{"kind":"REWORK","_number":1,"created":"2019-06-28 15:00:34.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/88/668188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/88/668188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/88/668188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/88/668188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/88/668188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/88/668188/1"}}},"commit":{"parents":[{"commit":"41aa061a6e7c63fc46f1720f515d42a78099025d","subject":"Inital random assignment of liasons","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/41aa061a6e7c63fc46f1720f515d42a78099025d"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-06-28 13:36:49.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-06-28 14:59:44.000000000","tz":60},"subject":"DNM: Display table of TC Liasons and projects","message":"DNM: Display table of TC Liasons and projects\n\nChange-Id: Icf454704a1503d27c0c526863bfaf103a7d7f8e1\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/9261e2477742f3bd2a7d93fff589852f75051be5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/9261e2477742f3bd2a7d93fff589852f75051be5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
