)]}'
{"id":"openstack%2Fgovernance~674049","triplet_id":"openstack%2Fgovernance~master~I5c062c3e3d08abce884734332a2ca53f214d1f68","project":"openstack/governance","branch":"master","topic":"formal-vote","hashtags":[],"change_id":"I5c062c3e3d08abce884734332a2ca53f214d1f68","subject":"I18n Extra-ATCs for Train","status":"MERGED","created":"2019-08-01 14:28:32.000000000","updated":"2019-08-20 18:56:38.000000000","submitted":"2019-08-20 18:56:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674049-1566327398059-462ff7de","meta_rev_id":"0f04ea21c8a0c2d63e7176cc672708d693bf56b6","_number":674049,"virtual_id_number":674049,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-08-20 18:56:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-20 18:45:55.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":1,"date":"2019-08-13 15:47:58.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2019-08-15 16:00:52.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-08-16 05:10:20.000000000","_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":1,"date":"2019-08-12 16:13:42.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-12 16:50:52.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-08-16 03:01:54.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-08-16 14:34:39.000000000","_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 14:30:23.000000000","updated_by":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"reviewer":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"state":"REVIEWER"},{"updated":"2019-08-05 12:34:45.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-08-13 15:47:58.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-08-15 16:00:52.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-16 03:01:54.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-08-16 05:10:20.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2019-08-16 14:34:39.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-08-20 18:45:55.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-20 18:56:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16fa648fa5619cafe29999f93bb8d7288099fb5f","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2019-08-01 14:28:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62a72dc0fc1a5bf52a790c977220f5cf2ff64ef6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 14:43:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/49/674049/1/check/openstack-tox-docs/bd03c46/html/ : SUCCESS in 4m 51s\n- openstack-tox-py36 https://logs.opendev.org/49/674049/1/check/openstack-tox-py36/4cc3531/ : SUCCESS in 4m 08s\n- openstack-tox-py37 https://logs.opendev.org/49/674049/1/check/openstack-tox-py37/8316032/ : SUCCESS in 4m 31s\n- openstack-tox-linters https://logs.opendev.org/49/674049/1/check/openstack-tox-linters/d167620/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"526beb9e78ce330c37c3397ad4eaef933807626a","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-08-01 15:08:03.000000000","message":"Topic changed from i18n-atc to formal-vote","accounts_in_message":[],"_revision_number":1},{"id":"4147d12daac4de4e9ab74d023180643377d39c56","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-08-01 15:08:24.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"42880e1751122072cdcd391d57750528c10ef032","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-08-01 15:14:13.000000000","message":"Patch Set 1:\n\n(6 comments)\n\n@Ian - Several people on the ATC list are not showing up in the Foundation community member database on openstack.org. I believe that means that even as ATCs, they will not be able to vote. They may also lose other privileges, but I\u0027m not sure. I have linked to the log output for each person who isn\u0027t in the database so you don\u0027t have to scan the logs yourself.","accounts_in_message":[],"_revision_number":1},{"id":"3c32e563aa380af5e53caaee024ffcffbec4b49c","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2019-08-01 15:25:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fb5eece1a37b22dc5bb621e2524d70dbf5afee2c","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2019-08-01 15:29:51.000000000","message":"Patch Set 2:\n\n@Doug: Thanks a lot for pointing out this and I appreciate TC\u0027s implementation on such an additional validation! Actually I relied on an Ansible implementation managed in I18n repo [1] but seems that the implementation needs to be revised to see json output from Foundation member check calls.\n\n \u003e @Ian - Several people on the ATC list are not showing up in the\n \u003e Foundation community member database on openstack.org. I believe\n\n[1] https://opendev.org/openstack/i18n/src/branch/master/playbooks/generate_atc.yml#L77","accounts_in_message":[],"_revision_number":2},{"id":"d64cdfca92456839ec4c70ad9aa732ad75e0e1cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 15:44:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/49/674049/2/check/openstack-tox-docs/a080b72/html/ : SUCCESS in 3m 52s\n- openstack-tox-py36 https://logs.opendev.org/49/674049/2/check/openstack-tox-py36/edb4f69/ : SUCCESS in 4m 20s\n- openstack-tox-py37 https://logs.opendev.org/49/674049/2/check/openstack-tox-py37/40ab6f2/ : SUCCESS in 4m 14s\n- openstack-tox-linters https://logs.opendev.org/49/674049/2/check/openstack-tox-linters/602380e/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"90de76c7b1aa87922e4db5dbf87ad6655d404320","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-08-05 12:34:45.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"66d0b81b0101ab8df7c0d4314ea59c39add28c41","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-08-05 12:36:14.000000000","message":"Patch Set 2:\n\n\u003e @Doug: Thanks a lot for pointing out this and I appreciate TC\u0027s\n \u003e implementation on such an additional validation! Actually I relied\n \u003e on an Ansible implementation managed in I18n repo [1] but seems\n \u003e that the implementation needs to be revised to see json output from\n \u003e Foundation member check calls.\n \u003e \n \u003e \u003e @Ian - Several people on the ATC list are not showing up in the\n \u003e \u003e Foundation community member database on openstack.org. I believe\n \u003e \n \u003e [1] https://opendev.org/openstack/i18n/src/branch/master/playbooks/generate_atc.yml#L77\n\nI wasn\u0027t able to make the URL linked there give me useful query results, but I didn\u0027t spend much time on it. You might want to confirm with the Foundation that that API works the way it\u0027s being used. Alternatively, you could use the openstack_governance library as an API to do the validation. If we don\u0027t have a command line tool for it already, it should be straightforward to add one to make the ansible integration easier.","accounts_in_message":[],"_revision_number":2},{"id":"ee879f70a0d9cf59ac8f815337e7978b853be032","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-08-05 18:14:51.000000000","message":"Patch Set 2:\n\n\u003e I believe that means that even as ATCs, they will not be able to vote.\n\nBy definition an ATC must be a member of the Foundation; I don\u0027t believe we can approve ATCs who are not. See appendix 4, section 3(a)(ii) of the bylaws: https://www.openstack.org/legal/technical-committee-member-policy/","accounts_in_message":[],"_revision_number":2},{"id":"5c368ff66d06e6ce1fbfa48e1c982b99b7c345e6","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-12 13:05:26.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"ea151930c5f7dd52c7e0e77fa973cabf9d8f2cb1","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-12 13:06:11.000000000","message":"Patch Set 2: -Rollcall-Vote","accounts_in_message":[],"_revision_number":2},{"id":"56b94c00a4a36ceee5f269ff3d90210fbf6de828","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-12 14:05:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"72217d4f2d4e946fa82e58317c891ac707532f32","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-12 14:09:00.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"a84becf1546d27be7b453a35a356014d49d7b0d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 14:23:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/49/674049/4/check/openstack-tox-docs/5481c26/html/ : SUCCESS in 3m 45s\n- openstack-tox-py36 https://logs.opendev.org/49/674049/4/check/openstack-tox-py36/a349f76/ : SUCCESS in 4m 22s\n- openstack-tox-py37 https://logs.opendev.org/49/674049/4/check/openstack-tox-py37/771ae4d/ : SUCCESS in 3m 49s\n- openstack-tox-linters https://logs.opendev.org/49/674049/4/check/openstack-tox-linters/8d4097c/ : FAILURE in 2m 53s","accounts_in_message":[],"_revision_number":4},{"id":"b2ce8e175e37c6d13631a617b0c763d95d558bf3","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-12 16:13:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"4834cb3d8adca22dbef3cbc61f5421c5e2ec37fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 16:29:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f708ec489454c65a8ee4ad8dacc6aec : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdf3e82a0ff44e71afede7735ee02283 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de69d669af994548a072e701e71c3367 : SUCCESS in 3m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b145940083854085a6849842096272f4 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":5},{"id":"e53fd9414867f951b7c95089f22c7eb1c64e3811","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-12 16:50:52.000000000","message":"Patch Set 5: Rollcall-Vote+1\n\nI think we\u0027re good in this case because the CI test passed fine so at this point all the ATCs are valid foundation members.","accounts_in_message":[],"_revision_number":5},{"id":"2340cfde27093cd8f1556be2c450ba1ca8579c7a","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-08-13 15:47:58.000000000","message":"Patch Set 5: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":5},{"id":"322ccc2b46d706d577201542734e14e406b2718f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-15 16:00:52.000000000","message":"Patch Set 5: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":5},{"id":"9e2458aeedc3721104cf97f630baa6cc7a3a43a1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-16 03:01:54.000000000","message":"Patch Set 5: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":5},{"id":"fcd8090bcfe19bd932a93d67187195f3f937ad1b","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-08-16 05:10:20.000000000","message":"Patch Set 5: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":5},{"id":"ea3d81292c3859698b03ab666ee084361f92db39","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-16 14:34:39.000000000","message":"Patch Set 5: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":5},{"id":"67ffdcf2b225e65501159b062807275c313e66bb","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-20 18:45:55.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"12e4cffa1a4f3695ded3f81af3e68ca133ae23a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:46:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b072a22b7a26d079616aa7ae16da7dbf84adb5f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:56:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82ef20427cad48cf80d7765bd2925f7d : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66608edf9f2449f99ea9d648a060bb98 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b3ca9b504024a09b85caccadddd7c19 : SUCCESS in 4m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9fa3914f3cff4400b40038cd18e7891f : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":5},{"id":"0f04ea21c8a0c2d63e7176cc672708d693bf56b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:56:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"87868e4fc9899117a86149d3f76bc36a924c2879","revisions":{"6f349aa6d4d2b328ae804c7912edc174ffdad469":{"kind":"REWORK","_number":1,"created":"2019-08-01 14:28:32.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/49/674049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/49/674049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/49/674049/1"}}},"commit":{"parents":[{"commit":"b24d958367fe9e3a69a0ace813bf92fbc99b3d1d","subject":"Merge \"Retire puppet-crane\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/b24d958367fe9e3a69a0ace813bf92fbc99b3d1d"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2019-08-01 14:10:19.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2019-08-01 14:10:19.000000000","tz":540},"subject":"I18n Extra-ATCs for Train","message":"I18n Extra-ATCs for Train\n\nList of active translators from 2019-01-26 to 2019-07-25\nidentified by Zanata Stats API.\nFull stats under https://wiki.openstack.org/wiki/I18nTeam/ATC_statistics#Train_cycle_.282019-01-26_to_2019-07-25.29\n\nChange-Id: I5c062c3e3d08abce884734332a2ca53f214d1f68\nCo-Authored-By: Frank Kloeker \u003cf.kloeker@telekom.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/6f349aa6d4d2b328ae804c7912edc174ffdad469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/6f349aa6d4d2b328ae804c7912edc174ffdad469"}]},"branch":"refs/heads/master"},"b9667780c77abfab92792c6d0590425613a4b62a":{"kind":"REWORK","_number":2,"created":"2019-08-01 15:25:59.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/49/674049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/49/674049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/49/674049/2"}}},"commit":{"parents":[{"commit":"b24d958367fe9e3a69a0ace813bf92fbc99b3d1d","subject":"Merge \"Retire puppet-crane\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/b24d958367fe9e3a69a0ace813bf92fbc99b3d1d"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2019-08-01 14:10:19.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2019-08-01 15:25:14.000000000","tz":540},"subject":"I18n Extra-ATCs for Train","message":"I18n Extra-ATCs for Train\n\nList of active translators from 2019-01-26 to 2019-07-25\nidentified by Zanata Stats API.\nFull stats under https://wiki.openstack.org/wiki/I18nTeam/ATC_statistics#Train_cycle_.282019-01-26_to_2019-07-25.29\n\nChange-Id: I5c062c3e3d08abce884734332a2ca53f214d1f68\nCo-Authored-By: Frank Kloeker \u003cf.kloeker@telekom.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/b9667780c77abfab92792c6d0590425613a4b62a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/b9667780c77abfab92792c6d0590425613a4b62a"}]},"branch":"refs/heads/master"},"306285e9f99d07fd86c267eda0d91f20d7b55f79":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-12 14:05:50.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/49/674049/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/49/674049/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/49/674049/3"}}},"commit":{"parents":[{"commit":"17c89cebd89335f361855c595e1b55147a858134","subject":"linters: enforce ATC membership after January 2020","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/17c89cebd89335f361855c595e1b55147a858134"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2019-08-01 14:10:19.000000000","tz":540},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-08-12 14:05:50.000000000","tz":0},"subject":"I18n Extra-ATCs for Train","message":"I18n Extra-ATCs for Train\n\nList of active translators from 2019-01-26 to 2019-07-25\nidentified by Zanata Stats API.\nFull stats under https://wiki.openstack.org/wiki/I18nTeam/ATC_statistics#Train_cycle_.282019-01-26_to_2019-07-25.29\n\nChange-Id: I5c062c3e3d08abce884734332a2ca53f214d1f68\nCo-Authored-By: Frank Kloeker \u003cf.kloeker@telekom.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/306285e9f99d07fd86c267eda0d91f20d7b55f79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/306285e9f99d07fd86c267eda0d91f20d7b55f79"}]},"branch":"refs/heads/master"},"70ef81bd8dab689c62a1496b76140959feb65377":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-12 14:09:00.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/49/674049/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/49/674049/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/49/674049/4"}}},"commit":{"parents":[{"commit":"ec4bbf93d8e21745e6f46c8ca7c32849f3c6b970","subject":"linters: enforce ATC membership after January 2020","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/ec4bbf93d8e21745e6f46c8ca7c32849f3c6b970"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2019-08-01 14:10:19.000000000","tz":540},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-08-12 14:09:00.000000000","tz":0},"subject":"I18n Extra-ATCs for Train","message":"I18n Extra-ATCs for Train\n\nList of active translators from 2019-01-26 to 2019-07-25\nidentified by Zanata Stats API.\nFull stats under https://wiki.openstack.org/wiki/I18nTeam/ATC_statistics#Train_cycle_.282019-01-26_to_2019-07-25.29\n\nChange-Id: I5c062c3e3d08abce884734332a2ca53f214d1f68\nCo-Authored-By: Frank Kloeker \u003cf.kloeker@telekom.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/70ef81bd8dab689c62a1496b76140959feb65377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/70ef81bd8dab689c62a1496b76140959feb65377"}]},"branch":"refs/heads/master"},"87868e4fc9899117a86149d3f76bc36a924c2879":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-12 16:13:42.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/49/674049/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/49/674049/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/674049/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/49/674049/5"}}},"commit":{"parents":[{"commit":"346d3d2f062dce8fcc714d78d56aa948f5598e3b","subject":"linters: enforce ATC membership after January 2020","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/346d3d2f062dce8fcc714d78d56aa948f5598e3b"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2019-08-01 14:10:19.000000000","tz":540},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-08-12 16:13:42.000000000","tz":0},"subject":"I18n Extra-ATCs for Train","message":"I18n Extra-ATCs for Train\n\nList of active translators from 2019-01-26 to 2019-07-25\nidentified by Zanata Stats API.\nFull stats under https://wiki.openstack.org/wiki/I18nTeam/ATC_statistics#Train_cycle_.282019-01-26_to_2019-07-25.29\n\nChange-Id: I5c062c3e3d08abce884734332a2ca53f214d1f68\nCo-Authored-By: Frank Kloeker \u003cf.kloeker@telekom.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/87868e4fc9899117a86149d3f76bc36a924c2879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/87868e4fc9899117a86149d3f76bc36a924c2879"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
