)]}'
{"id":"openstack%2Fgovernance~677771","triplet_id":"openstack%2Fgovernance~master~I4a4b17bbbe2e3168c70b67594639946895ff4d1f","project":"openstack/governance","branch":"master","topic":"formal-vote","hashtags":[],"change_id":"I4a4b17bbbe2e3168c70b67594639946895ff4d1f","subject":"Align release naming process with practice","status":"MERGED","created":"2019-08-21 15:53:02.000000000","updated":"2019-09-24 17:45:25.000000000","submitted":"2019-09-24 17:42:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677771-1569346976595-dadc74e4","meta_rev_id":"8659ba5db5bcbfea4019b168487d8a601313697c","_number":677771,"virtual_id_number":677771,"owner":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 17:42:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":-1,"date":"2019-08-21 20:40:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":-1,"date":"2019-08-21 20:40:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":-1,"date":"2019-08-25 06:57:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-09-24 17:35:54.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":1,"date":"2019-08-21 20:40:30.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-08-21 20:40:30.000000000","_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-08-22 08:59:45.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-09-24 17:20:46.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-08-21 20:40:30.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-09-23 21:31:13.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-21 20:40:30.000000000","_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 15:58:28.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-08-21 16:01:30.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-08-21 16:03:04.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-21 16:30:05.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-08-21 17:19:34.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-21 18:37:59.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-08-22 08:59:45.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-25 06:57:06.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-09-24 17:20:46.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-09-24 17:35:54.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-24 17:42:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab71c3ed3e6c4adb154f5a4617f3df03661a5414","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-08-21 15:53:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9698f7d2d5f6769067d81d4817e8cf17c4fb9a17","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-21 15:54:49.000000000","message":"Topic set to formal-vote","accounts_in_message":[],"_revision_number":1},{"id":"59d612bddf2134a937c114a740b8e05d64deefa3","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-21 15:58:28.000000000","message":"Patch Set 1: Rollcall-Vote+1\n\nI like that. Yes. Good clarification.","accounts_in_message":[],"_revision_number":1},{"id":"b2cee589fb47dec11ae0dc9dca6cfbf9348c28b8","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-21 16:01:30.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"c087100dd6b90c1d4c1be590b385bec591c0db49","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-21 16:03:04.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"90885df350174f40ec7fe2fe3dd3143503609688","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 16:05:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e154957077d4db891870b68e0a451f6 : SUCCESS in 7m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3ccfcf866e6341da8dd826236281fd74 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c145eeebfe1b5c7e54a56faaec1cbe1a69f780db","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-08-21 16:18:43.000000000","message":"Patch Set 1:\n\nWhile this is a good addition, it explains only one of the *seven* names which was removed from the poll this time.","accounts_in_message":[],"_revision_number":1},{"id":"57a11360a5bb7dd69e26f9d0d4cc5dd4099aac9d","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-08-21 16:30:05.000000000","message":"Patch Set 1: Code-Review-1\n\nLet me clarify that a bit.  While I think this is a good addition if we stick with our current process, in practice I think this is ultimately harmful.  Rather than reducing contention, it will increase it, and will force us to have discussions about things which are offensive to members of our very large community.  This is why I propose that we scrap the process entirely in favor of something that is far less likely to generate controversy or distraction.  Let\u0027s pick a new process which won\u0027t risk alienating people.","accounts_in_message":[],"_revision_number":1},{"id":"1e12c836ce51bb6e626c852c7d673bf7ed127065","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-08-21 16:38:19.000000000","message":"Patch Set 1: Rollcall-Vote+1\n\n\u003e will force us to have discussions about things which are offensive to members of our very large community.\n\nIt doesn\u0027t force us to have the discussion in public, unlike what happens if those things win the poll (see: M release).\n\n\u003e Let\u0027s pick a new process which won\u0027t risk alienating people.\n\nThat\u0027s a fine discussion to have also, but I don\u0027t think it should block us from making sure our current documentation accurately reflects our de facto practice and the knowledge we\u0027ve won from past experience.","accounts_in_message":[],"_revision_number":1},{"id":"bf1f3483cb1b69c71b6cc9bec1a41bf79a4bdd18","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-08-21 16:57:57.000000000","message":"Patch Set 1:\n\nZane Bitter wrote:\n\n\u003e \u003e will force us to have discussions about things which are offensive to members of our very large community.\n\u003e\n\u003e It doesn\u0027t force us to have the discussion in public, unlike what\n\u003e happens if those things win the poll (see: M release).\n\nAre you suggesting that people who find names inconsiderate should contact the coordinator privately, and the coordinator should provide no public explanation for why they were removed?  Should they automatically remove words, or use their own judgement if someone says \"the word \u0027Train\u0027 is offensive\"?\n\nIf so, that is all worth including in the document.\n\nI still think it\u0027s a bad idea and will result in more contention rather than less.","accounts_in_message":[],"_revision_number":1},{"id":"049c1a27cb5266f17b3b2b30bbbd40deca04194c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-21 17:19:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fce787e27435046591a746203076f897d2f24e7d","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-08-21 19:45:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31de53d17d48501d4622c0d5e6ff6dccc2e2a5c0","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-08-21 20:40:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"315529dbcf6e0871efc6e919089803b678eefe6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 20:49:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d79fd98cfb574038a6e8c52276ea55d1 : SUCCESS in 7m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bda2bd18c72042b1a4fc76297c740f5e : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"452c6f04e69713943ed0cc256985058e9f2f80ba","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-22 08:59:45.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\nI agree that we should remove abusive words from the poll options even before having the poll. But I also agree that it increases the subjectivity of the exercise.\n\nThis accurately describes the process we currently follow, so I\u0027m +2 on merging this. That said, I think it reinforces the fact that the process is broken and needs to be reformed, ideally in favor of a deterministic process that removes subjectivity and sidesteps cultural issues.","accounts_in_message":[],"_revision_number":2},{"id":"d65fcfc830ee99f4cb8bb89956f37e31ae5c44e4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-25 06:57:06.000000000","message":"Patch Set 2: Code-Review-1\n\n\u003e Zane Bitter wrote:\n \u003e \n \u003e \u003e \u003e will force us to have discussions about things which are\n \u003e offensive to members of our very large community.\n \u003e \u003e\n \u003e \u003e It doesn\u0027t force us to have the discussion in public, unlike what\n \u003e \u003e happens if those things win the poll (see: M release).\n \u003e \n \u003e Are you suggesting that people who find names inconsiderate should\n \u003e contact the coordinator privately, and the coordinator should\n \u003e provide no public explanation for why they were removed?  Should\n \u003e they automatically remove words, or use their own judgement if\n \u003e someone says \"the word \u0027Train\u0027 is offensive\"?\n \u003e \n \u003e If so, that is all worth including in the document.\n \u003e \n \u003e I still think it\u0027s a bad idea and will result in more contention\n \u003e rather than less.\n\nYeah, James has good point here. IMO also before removing (when TC says yes to removal candidate) it should be made public with appropriate reasoning. If there are very strict objection from the community above removal that can be discussed. If there is no conclusion from community then TC can have the power to take the final decision.  \n\nit is great to add the removal process but I am -1 to include/discuss with the community also for removal.","accounts_in_message":[],"_revision_number":2},{"id":"4640e65a654ac9b0522a103cee34f5d3be57bb7c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-25 06:59:39.000000000","message":"Patch Set 2:\n\n\u003e \u003e Zane Bitter wrote:\n \u003e \u003e\n \u003e \u003e \u003e \u003e will force us to have discussions about things which are\n \u003e \u003e offensive to members of our very large community.\n \u003e \u003e \u003e\n \u003e \u003e \u003e It doesn\u0027t force us to have the discussion in public, unlike\n \u003e what\n \u003e \u003e \u003e happens if those things win the poll (see: M release).\n \u003e \u003e\n \u003e \u003e Are you suggesting that people who find names inconsiderate\n \u003e should\n \u003e \u003e contact the coordinator privately, and the coordinator should\n \u003e \u003e provide no public explanation for why they were removed?  Should\n \u003e \u003e they automatically remove words, or use their own judgement if\n \u003e \u003e someone says \"the word \u0027Train\u0027 is offensive\"?\n \u003e \u003e\n \u003e \u003e If so, that is all worth including in the document.\n \u003e \u003e\n \u003e \u003e I still think it\u0027s a bad idea and will result in more contention\n \u003e \u003e rather than less.\n \u003e \n \u003e Yeah, James has good point here. IMO also before removing (when TC\n \u003e says yes to removal candidate) it should be made public with\n \u003e appropriate reasoning. If there are very strict objection from the\n \u003e community above removal that can be discussed. If there is no\n \u003e conclusion from community then TC can have the power to take the\n \u003e final decision.\n \u003e \n \u003e it is great to add the removal process but I am -1 to\n \u003e include/discuss with the community also for removal.\n\nClarity about my wording: I am -1 *because we should* include/discuss with the community also for removal.","accounts_in_message":[],"_revision_number":2},{"id":"0e4755b5445fc5215f12bb022e3597ba63fec09f","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-09-04 21:28:26.000000000","message":"Patch Set 2:\n\n@gmann, I invite you to read through https://en.wikipedia.org/wiki/Place_names_considered_unusual#Profanity,_humorous_words_or_highly_charged_words and provide a list of the names you\u0027d be comfortable leading a public discussion about. I\u0027ll wait ;)","accounts_in_message":[],"_revision_number":2},{"id":"04e6d43b4430f83216687805df75d6297c2005a1","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 21:31:13.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"48b28b15d58fc2e09fb262e1f871f580152cbf93","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-09-24 17:20:46.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"d8d555b5fbca16170c267681de2de562500beaa7","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-24 17:35:54.000000000","message":"Patch Set 2: Workflow+1\n\nThis has reached more than the necessary votes by reaching a majority.\n\nI agree with the concerns raises by Jim, Andreas, and Ghanshyam.\nHowever, I think we can adjust this later.\nIt respects the votes the TC members also have done through CIVS and published on the ML. Let\u0027s move forward.","accounts_in_message":[],"_revision_number":2},{"id":"f282997ef90ee78ee720818d000d9d9793b461f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 17:36:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8400efbca66ff5d4886c71e38e5ab8c3e6ff5c34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 17:42:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"da8174af044787b76694b26fab327dd1718bb965","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 17:42:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6984a09d344e4089b95a83ea7c55c8e2 : SUCCESS in 5m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2fcd99987681416e8c645d931d06e2ca : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8659ba5db5bcbfea4019b168487d8a601313697c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 17:45:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/3f56635bc0d84a8ca57164f727bc5bcc : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f18d311586bafb7a36f223e924c6656f316bf843","revisions":{"28b8ed291ad6a542d8a06c1af27beb113b7212ee":{"kind":"REWORK","_number":1,"created":"2019-08-21 15:53:02.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/71/677771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/71/677771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/71/677771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/71/677771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/71/677771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/71/677771/1"}}},"commit":{"parents":[{"commit":"b4f340600cb963347e40c1eeb8110449b1047ebe","subject":"Merge \"I18n Extra-ATCs for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/b4f340600cb963347e40c1eeb8110449b1047ebe"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-08-21 15:34:54.000000000","tz":-240},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-08-21 15:50:04.000000000","tz":-240},"subject":"Align release naming process with practice","message":"Align release naming process with practice\n\nThe actual conduct of the release naming process has diverged from that\nwhich is documented. Bring them back into alignment by giving the naming\nco-ordinator and the TC an explicit role in removing names that aren\u0027t\nin line with the Code of Conduct\u0027s injunction to be considerate of our\nglobal audience, prior to their inclusion in the poll.\n\nChange-Id: I4a4b17bbbe2e3168c70b67594639946895ff4d1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/28b8ed291ad6a542d8a06c1af27beb113b7212ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/28b8ed291ad6a542d8a06c1af27beb113b7212ee"}]},"branch":"refs/heads/master"},"f18d311586bafb7a36f223e924c6656f316bf843":{"kind":"NO_CHANGE","_number":2,"created":"2019-08-21 20:40:30.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/71/677771/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/71/677771/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/71/677771/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/71/677771/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/71/677771/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/71/677771/2"}}},"commit":{"parents":[{"commit":"b4f340600cb963347e40c1eeb8110449b1047ebe","subject":"Merge \"I18n Extra-ATCs for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/b4f340600cb963347e40c1eeb8110449b1047ebe"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-08-21 15:34:54.000000000","tz":-240},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-08-21 20:40:17.000000000","tz":-240},"subject":"Align release naming process with practice","message":"Align release naming process with practice\n\nThe actual conduct of the release naming process has diverged from that\nwhich is documented. Bring them back into alignment by giving the naming\nco-ordinator and the TC an explicit role in removing names that aren\u0027t\nin line with the Code of Conduct\u0027s injunction to be considerate of our\nglobal audience, prior to their inclusion in the poll.\n\nChange-Id: I4a4b17bbbe2e3168c70b67594639946895ff4d1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f18d311586bafb7a36f223e924c6656f316bf843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f18d311586bafb7a36f223e924c6656f316bf843"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
