)]}'
{"id":"openstack%2Fgovernance~677772","triplet_id":"openstack%2Fgovernance~master~I158a32b76aee559c194a4bcafa1482e0426e87f3","project":"openstack/governance","branch":"master","topic":"formal-vote","hashtags":[],"change_id":"I158a32b76aee559c194a4bcafa1482e0426e87f3","subject":"Clarify proscription of generic release names","status":"MERGED","created":"2019-08-21 15:53:02.000000000","updated":"2019-09-24 17:50:02.000000000","submitted":"2019-09-24 17:48:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677772-1569347307500-c425b5ec","meta_rev_id":"b3c2e6a58c92f7009d219c1c379c9e2948811206","_number":677772,"virtual_id_number":677772,"owner":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 17:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-08-21 20:47:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-09-23 18:51:31.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-09-04 15:00:09.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":1,"date":"2019-08-22 13:57:23.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-08-22 09:02:59.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":1,"date":"2019-08-25 07:00:10.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-08-26 10:58:29.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-09-04 15:00:09.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-22 09:35:45.000000000","_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 16:01:57.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-08-21 16:46:44.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-08-21 17:13:34.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2019-08-21 20:47:42.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-08-22 09:02:59.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-08-22 09:35:45.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-08-25 07:00:10.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-09-04 15:00:09.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-09-23 18:51:31.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-24 17:48:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d7195068d1d31e06fe6d61350034ab69e66b8b73","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-08-21 15:53:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"667d74a76fafa3f7844411664afc1a69383e798a","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-21 15:54:44.000000000","message":"Topic set to formal-vote","accounts_in_message":[],"_revision_number":1},{"id":"d2d962bb84c33e452a537b163ba20a34284bf7b7","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-21 16:01:57.000000000","message":"Patch Set 1: Rollcall-Vote+1\n\nIf we keep what we are currently doing, this is a useful clarification","accounts_in_message":[],"_revision_number":1},{"id":"0a24fbff03a93c6df007680654744f1469fab243","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 16:13:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d08d3df359d473196ea3d237a076abc : SUCCESS in 6m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/125d16e431c84ae78d46f708782d5702 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1cf0c7c50054546174b1c831fa3ae7e22661e65a","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-21 16:21:34.000000000","message":"Patch Set 1: Rollcall-Vote+1\n\nAgreed, useful.","accounts_in_message":[],"_revision_number":1},{"id":"d720a4f18162311f5dcdac505af37fc095b226c9","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-08-21 16:22:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ed99a85fec9fd6483bfe6a811ff3a54f87c56fd","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-08-21 16:27:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcdc108302e603a7c2e461ef30208ce453dd3c32","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-08-21 16:46:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bc78377a07917d1180d1175831d926f24d9f7da","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2019-08-21 17:13:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e979dc59e166d59d466c2f0267abc5f170fad8b4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-21 18:04:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1ace89a4aa03c16bf507321dca60546608d4e738","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-08-21 20:42:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"157380e62c88216ab3e30c2921c6f81f60bb35af","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-21 20:47:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1aec328c5d8b583220d36f869bcd1e72c9fd6005","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 20:53:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea8b92a32e745d7b13f4640e5d05bad : SUCCESS in 6m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/01332a9ac4dd41b48f4140ec3584a397 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"2b2fa050fea18854fd7229974dc3c50e468c7f3e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-22 09:02:59.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\nThis accurately describes the process we *currently* follow, so I\u0027m +2 on merging this. I still think we should replace our current process with one of jeblair\u0027s suggestions though.","accounts_in_message":[],"_revision_number":2},{"id":"e9c187c537c80773d08022b43ebfd0746c3692e4","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-22 09:35:45.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"273f013356c794075a41f25baedc2152a849b8af","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-08-22 13:57:23.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"2cc4dc45ba4925d50503900020e4227156209c33","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-25 07:00:10.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\nthanks. lgtm","accounts_in_message":[],"_revision_number":2},{"id":"0eb5d50a037fc0352145752581f17c31abfeead2","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-26 10:58:29.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"39d31556388b9b6fa6a94edbd548073a1adbf177","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-09-04 15:00:09.000000000","message":"Patch Set 2: Rollcall-Vote+1 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c39e7530fd4ae770d2a0de0ff870595cb52b815c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 18:31:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ef39bcddd8b797d1bc069c00436eb1f4d2606538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 18:40:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86ffbd242241437498ceb5b7950e1d25 : SUCCESS in 6m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e2df8a12df0c4396a296c429d89ae396 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"526ddd3538bb43cb58fe5dd7977096b9f537babc","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 18:51:31.000000000","message":"Patch Set 2: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"062cbb850f50a550c5b4f9abbecb67fd23937a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 19:03:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d683514decf4aa3a1b01c150cc8293c : SUCCESS in 8m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f2e41a7e1ee24783b622a891998dfb60 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"3f8da5a5c09b9670f2a400b2225c34f1b496ce95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 17:36:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ed4d279725f4b6de1b19c61c3f1caa3790b82228","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 17:48:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848eab281916461e837ee383cb6a8df2 : SUCCESS in 5m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/94c80d21206743c18348d2c9cdb9b244 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"15ea8a8fb77b6097b0970679f793b7c7a662dc53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 17:48:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b3c2e6a58c92f7009d219c1c379c9e2948811206","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 17:50:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/944912a21e8c47b4a721d04e71d63cde : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"97c9e86a51546f7aa0e7b473d208bff7ae079748","revisions":{"900fd50621dc415d55d378f5362b11940ee70f50":{"kind":"REWORK","_number":1,"created":"2019-08-21 15:53:02.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/72/677772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/72/677772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/72/677772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/72/677772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/72/677772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/72/677772/1"}}},"commit":{"parents":[{"commit":"28b8ed291ad6a542d8a06c1af27beb113b7212ee","subject":"Align release naming process with practice","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/28b8ed291ad6a542d8a06c1af27beb113b7212ee"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-08-21 15:43:47.000000000","tz":-240},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-08-21 15:50:04.000000000","tz":-240},"subject":"Clarify proscription of generic release names","message":"Clarify proscription of generic release names\n\nThe fourth release name criterion arguably implies that generic names of\ntypes of geographic features are not eligible, but is ambiguous and has\nbeen the source of confusion. Avoid this by stating it explicitly.\n\nChange-Id: I158a32b76aee559c194a4bcafa1482e0426e87f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/900fd50621dc415d55d378f5362b11940ee70f50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/900fd50621dc415d55d378f5362b11940ee70f50"}]},"branch":"refs/heads/master"},"97c9e86a51546f7aa0e7b473d208bff7ae079748":{"kind":"REWORK","_number":2,"created":"2019-08-21 20:42:20.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/72/677772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/72/677772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/72/677772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/72/677772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/72/677772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/72/677772/2"}}},"commit":{"parents":[{"commit":"f18d311586bafb7a36f223e924c6656f316bf843","subject":"Align release naming process with practice","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f18d311586bafb7a36f223e924c6656f316bf843"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-08-21 15:43:47.000000000","tz":-240},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-08-21 20:40:48.000000000","tz":-240},"subject":"Clarify proscription of generic release names","message":"Clarify proscription of generic release names\n\nThe fourth release name criterion arguably implies that generic names of\ntypes of geographic features are not eligible, but is ambiguous and has\nbeen the source of confusion. Reduce this by stating it explicitly.\n\nArguably this rule would have required an exception to be granted for\nPike, and thus could be considered a change in policy (though\ninterpretations on this seem to differ based on whether you are from\nBoston or not).\n\nChange-Id: I158a32b76aee559c194a4bcafa1482e0426e87f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/97c9e86a51546f7aa0e7b473d208bff7ae079748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/97c9e86a51546f7aa0e7b473d208bff7ae079748"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
