)]}'
{"id":"openstack%2Fgovernance~680407","triplet_id":"openstack%2Fgovernance~master~I0a6842e27dc376134ffa3c393cd90fdb1158ca52","project":"openstack/governance","branch":"master","topic":"formal-vote","hashtags":[],"change_id":"I0a6842e27dc376134ffa3c393cd90fdb1158ca52","subject":"Add Tempest and Plugins req in Train tested Runtime","status":"ABANDONED","created":"2019-09-05 13:21:21.000000000","updated":"2019-12-04 12:05:35.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b6f281eaa87862ee8d0434944b5158164a561ef7","_number":680407,"virtual_id_number":680407,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"date":"2019-09-17 14:48:46.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-09-05 13:35:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-1,"date":"2019-09-12 13:17:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":-1,"date":"2019-09-18 02:32:09.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2019-09-12 09:09:43.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":-1,"date":"2019-10-15 14:05:15.000000000","_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":-1,"date":"2019-10-18 17:07:10.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":1,"date":"2019-11-15 18:05:38.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":-1,"date":"2019-09-12 12:14:27.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-09 11:45:45.000000000","_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-05 13:35:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-09 11:45:45.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-09-12 09:09:43.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-12 12:14:27.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-12 13:17:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-18 02:32:09.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-10-15 14:05:15.000000000","updated_by":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"reviewer":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"state":"REVIEWER"},{"updated":"2019-10-18 17:07:10.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-11-15 18:05:38.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"}],"messages":[{"id":"97394d6a2d14fef3a297eb1594bfe55ff4bafcc4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-05 13:21:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57628ad2428f7f075dea288d21c3807a22987caf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-05 13:30:43.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nSeems a little late to be changing guidance on this for train at this point.\n\nDoes tempest really need to stick with 3.5? The services need to run under 3.5 for those older branches, but tempest itself shouldn\u0027t really matter. Or is the concern that if we don\u0027t continue to support 3.5, the user may need to install a newer version of tempest to be able to run the tempest tests?","accounts_in_message":[],"_revision_number":1},{"id":"14cd6e77e2f8d77e3ea132c6162ca44d91246a37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 13:35:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e62c46ac37944fe8dc655bcfd18bc4e : SUCCESS in 4m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5ab3e1d87af44815aee5183f25666784 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5979a1296977957a6d26de96d875e3c22e143d36","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-05 14:00:26.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n \u003e \n \u003e Seems a little late to be changing guidance on this for train at\n \u003e this point.\n \u003e \n \u003e Does tempest really need to stick with 3.5? The services need to\n \u003e run under 3.5 for those older branches, but tempest itself\n \u003e shouldn\u0027t really matter. Or is the concern that if we don\u0027t\n \u003e continue to support 3.5, the user may need to install a newer\n \u003e version of tempest to be able to run the tempest tests?\n\nIt actually needs to support otherwise we may break stable branch testing. It is not a new addition, it is to mention explicitly. While dropping the support of py3.5 in Stein and updating the Train runtime we have not removed the py3.5 support from Tempest or any of its plugin.","accounts_in_message":[],"_revision_number":1},{"id":"bc1311c91bdc45c7837f9492140f6ad00e2de783","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-09-09 11:45:13.000000000","message":"Topic set to documentation-update","accounts_in_message":[],"_revision_number":1},{"id":"a2763685c2b60dff8e55a8627c022ddfd7db16f7","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-09-09 11:45:24.000000000","message":"Topic changed from documentation-update to documentation-change","accounts_in_message":[],"_revision_number":1},{"id":"09489849bccd6730d1b99bfb19d13d3b85044234","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-09-09 11:45:45.000000000","message":"Patch Set 1: Rollcall-Vote+1\n\nOther than Sean\u0027s nit, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"0015c5554df4147266e46cba4259dcf79f8d9b3c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-12 09:09:43.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"198c66b82bfcb67949a08b5d207d0fb1df52aa72","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-12 12:14:27.000000000","message":"Patch Set 1: Rollcall-Vote-1\n\nTempest is testing. For all that matters, it could be in go or rust. That shouldn\u0027t have any impact on the ecosystem as a whole like service, clients, libraries, or deployment  projects.\n\nI am confused why we want an explicit TC overrule to say that tempest is an exception.\n\nTempest is master of its destiny. If tempest is using python, then it should follow the python guidelines. But it doesn\u0027t prevent ensuring stability for older releases if the project decides so.","accounts_in_message":[],"_revision_number":1},{"id":"ca4ef193ca34f0f67aacc0d7932ca01ba2c23462","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-12 13:17:23.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Tempest is testing. For all that matters, it could be in go or\n \u003e rust. That shouldn\u0027t have any impact on the ecosystem as a whole\n \u003e like service, clients, libraries, or deployment  projects.\n \u003e \n \u003e I am confused why we want an explicit TC overrule to say that\n \u003e tempest is an exception.\n \u003e \n \u003e Tempest is master of its destiny. If tempest is using python, then\n \u003e it should follow the python guidelines. But it doesn\u0027t prevent\n \u003e ensuring stability for older releases if the project decides so.\n\nI agree with this position. We don\u0027t need to have governance level declarations for every project variation.","accounts_in_message":[],"_revision_number":1},{"id":"caf0c712a67e8af724df75c79f3cde1a86014fd0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-17 14:47:41.000000000","message":"Patch Set 1:\n\n\u003e Tempest is testing. For all that matters, it could be in go or\n \u003e rust. That shouldn\u0027t have any impact on the ecosystem as a whole\n \u003e like service, clients, libraries, or deployment  projects.\n \u003e \n \u003e I am confused why we want an explicit TC overrule to say that\n \u003e tempest is an exception.\n \u003e \n \u003e Tempest is master of its destiny. If tempest is using python, then\n \u003e it should follow the python guidelines. But it doesn\u0027t prevent\n \u003e ensuring stability for older releases if the project decides so.\n\nAlong with OpenStack services official supported python version, if we add the officially supported version of OpenStack testing tools then it will help Tempest users to understand that Train version of Tempest can be run against python2.7 cloud.\n\nI am not seeing any harm to say about officially supported python version for the testing tool because it is not same as OpenStack services.","accounts_in_message":[],"_revision_number":1},{"id":"b7776cad2985284710409c4ac6f7eb5a31af386a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-17 14:48:46.000000000","message":"Patch Set 1:\n\n\u003e \u003e Tempest is testing. For all that matters, it could be in go or\n \u003e \u003e rust. That shouldn\u0027t have any impact on the ecosystem as a whole\n \u003e \u003e like service, clients, libraries, or deployment  projects.\n \u003e \u003e\n \u003e \u003e I am confused why we want an explicit TC overrule to say that\n \u003e \u003e tempest is an exception.\n \u003e \u003e\n \u003e \u003e Tempest is master of its destiny. If tempest is using python,\n \u003e then\n \u003e \u003e it should follow the python guidelines. But it doesn\u0027t prevent\n \u003e \u003e ensuring stability for older releases if the project decides so.\n \u003e \n \u003e I agree with this position. We don\u0027t need to have governance level\n \u003e declarations for every project variation.\n\nThis is not each project variation this is about OpenStack testing tool supported versions along with services supported version.","accounts_in_message":[],"_revision_number":1},{"id":"77e525cc759684a15651b1db9312ad3009c037ea","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-09-18 02:32:09.000000000","message":"Patch Set 1: Rollcall-Vote-1\n\nThis document contains the minimum set of Python versions that each project should gate against. There\u0027s nothing stopping projects from running other tests according to their own requirements, and no need to document those here IMHO.","accounts_in_message":[],"_revision_number":1},{"id":"aaebefd0feadc01b29e38f1d58cbc636487e03b2","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-01 11:33:38.000000000","message":"Topic changed from documentation-change to formal-vote","accounts_in_message":[],"_revision_number":1},{"id":"79147e015f03e6c6ab72bf924116a3e24e8a692c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-01 11:34:02.000000000","message":"Patch Set 1:\n\nI changed the topic, as this is effectively changing the policy, not clarifying it.","accounts_in_message":[],"_revision_number":1},{"id":"e37553be92c0bf7db8d9e272e74aa41da22a5f3a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-01 11:45:46.000000000","message":"Patch Set 1:\n\nI think we should abandon the topic until https://review.opendev.org/681203 is settled.","accounts_in_message":[],"_revision_number":1},{"id":"c4e2b61fabff916e4c26fa2b8a5e9f856b601108","author":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"date":"2019-10-15 14:05:15.000000000","message":"Patch Set 1: Rollcall-Vote-1\n\n\u003e I think we should abandon the topic until https://review.opendev.org/681203\n \u003e is settled.\n\n++.\n\nI\u0027m -1 until we settle that out.","accounts_in_message":[],"_revision_number":1},{"id":"60cfa49812fc5a863912711330318cfc385dbb19","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-10-18 17:07:10.000000000","message":"Patch Set 1: Rollcall-Vote-1\n\n\u003e I think we should abandon the topic until https://review.opendev.org/681203\n \u003e is settled.\n\nI concur.","accounts_in_message":[],"_revision_number":1},{"id":"63a1858353b164e906da657df7d28c87ecfa8388","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-11-15 18:05:38.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"b6f281eaa87862ee8d0434944b5158164a561ef7","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-12-04 12:05:35.000000000","message":"Abandoned\n\nThis has too many dissenting votes. I am abandoning it, as we won\u0027t be able to go forward with this proposition.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"47e359f3bac4d235a287e1d83a4258d38b0ae868","revisions":{"47e359f3bac4d235a287e1d83a4258d38b0ae868":{"kind":"REWORK","_number":1,"created":"2019-09-05 13:21:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/07/680407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/07/680407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/07/680407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/07/680407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/07/680407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/07/680407/1"}}},"commit":{"parents":[{"commit":"fae520ace09987f2823ab0f1edb28f03179ac4ac","subject":"Merge \"Add os_murano project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/fae520ace09987f2823ab0f1edb28f03179ac4ac"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-05 13:17:47.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-05 13:21:17.000000000","tz":0},"subject":"Add Tempest and Plugins req in Train tested Runtime","message":"Add Tempest and Plugins req in Train tested Runtime\n\nTempest and Temepst Plugins are branchless and their\nmaster code test the supported stable branches. This\nmakes them to keep testing the older Python version\nalso. For exmaple, they need to keep testing py3.5 also.\n\nLet\u0027s mention that explicitly to make it clear.\n\nChange-Id: I0a6842e27dc376134ffa3c393cd90fdb1158ca52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/47e359f3bac4d235a287e1d83a4258d38b0ae868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/47e359f3bac4d235a287e1d83a4258d38b0ae868"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
