)]}'
{"id":"openstack%2Fgovernance~694480","triplet_id":"openstack%2Fgovernance~master~I0200b500c4a89c6326d843b8628dfd87dc6eb778","project":"openstack/governance","branch":"master","topic":"project-update","hashtags":[],"change_id":"I0200b500c4a89c6326d843b8628dfd87dc6eb778","subject":"Retire neutron-interconnection from openstack/","status":"MERGED","created":"2019-11-15 08:52:39.000000000","updated":"2019-12-03 15:11:57.000000000","submitted":"2019-12-03 15:11:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694480-1575385917764-232123cc","meta_rev_id":"ae580d7ea29994ac61af067051eb306033b806c4","_number":694480,"virtual_id_number":694480,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-20 11:40:37.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-03 15:11:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-11-18 15:57:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2019-11-15 16:08:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-11-29 16:44:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2019-12-03 14:08:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":1,"date":"2019-12-03 14:59:06.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-11-15 17:47:48.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":1,"date":"2019-11-15 16:08:13.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-12-03 14:31:07.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-18 16:25:13.000000000","_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"value":0,"date":"2019-12-03 14:07:30.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-11-21 11:10:43.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2019-11-21 14:59:38.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 16:08:30.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-11-15 17:47:48.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-11-18 15:57:34.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-11-18 16:25:13.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-11-21 11:10:43.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-11-21 14:59:38.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-11-29 16:44:21.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-12-03 14:07:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-12-03 14:08:19.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-12-03 14:31:07.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-12-03 14:59:06.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-12-03 15:11:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e02c22b7fac849d70749df1cffee90f311a3f0cb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 08:52:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd87afcd79f8b285dcda1f501aab5b506890fc7c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-15 08:58:00.000000000","message":"Patch Set 1: Code-Review-1\n\nyou need to add to the legacy file for reference","accounts_in_message":[],"_revision_number":1},{"id":"27d69fd3bfe0dd683b6cc340459278bac60d6418","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 09:15:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d9381cde71cacbdb1ce6a1e48debcc4152457547","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 09:15:17.000000000","message":"Patch Set 1:\n\n\u003e you need to add to the legacy file for reference\n\nDone","accounts_in_message":[],"_revision_number":1},{"id":"063099c22e179d2be487689a813c23248d0130e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 09:25:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7af9fec15fc4ebf9f1ae15137aed387 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ebc9bc019f44145817ab69bf7ae2efe : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9a96468945e4123a7b993028b3e5361 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd6aa9e3e6bb4c3c80c6e3ba2d6f9f0d : SUCCESS in 5m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d6f62f8cbb524d69b7688b7f838d1b3b : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b5a7902a02c12b8a7ed1ee36c546b49d8b204263","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 09:37:38.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1f1ea5dfa3de3bdcbfebf4676eb38f28a52cc07a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 09:54:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6851353976c447d7a0a4aee23dd15b1e : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3061584b2b464e9ce66552a6be13cd : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fd438ca1bb045eb97023853e9ef5786 : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94fc01da9816411fb8b7512577e0f380 : SUCCESS in 3m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/97e4ce1e7a1f4bd28eef14f260918625 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"565ca6857844df80fa6793aaa71d3e6558099d3b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-15 11:37:16.000000000","message":"Patch Set 3:\n\nI feel like the rename should depend on the governance change, rather than the other way around. That will allow to unblock this while we wait for the next rename window...","accounts_in_message":[],"_revision_number":3},{"id":"01db13ab2b318e955395bb1d5313707079b4ef23","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-15 12:24:15.000000000","message":"Patch Set 3:\n\nThierry, based on https://governance.openstack.org/tc/resolutions/20190711-mandatory-repository-retirement.html - it will not be a rename but instead a retirement and create.\n\nThat can be done without a maintenance window.","accounts_in_message":[],"_revision_number":3},{"id":"72afa91d48f52626ce12dd64381143c594778f71","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-11-15 15:26:05.000000000","message":"Topic changed from project-rename to project-update","accounts_in_message":[],"_revision_number":3},{"id":"f14e6119b9221eee67315df4fc3c3fceba2acda4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 15:51:02.000000000","message":"Topic changed from project-update to neutron-interconnection-retire","accounts_in_message":[],"_revision_number":3},{"id":"b6f884165594d63918a6e562ae27cc44784b748d","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-11-15 16:08:13.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"dd0955112f3ef28be110454529d2f55bacaa2f58","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-15 16:08:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"893eddbf1f64eb4af856deeeeaccf81e5eac1388","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-11-15 17:47:48.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"1d68c465c61cae928829022a358d0e5fcf5078b4","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-11-16 15:45:33.000000000","message":"Patch Set 3: Code-Review-1\n\nAs said on the mailing list:\n\nI understand the need to adapt the project status to the lack of activity on the project in the past year.\nDuring this time, my time as a reviewer, preventing the code submitted to be merged, has been taken by other activites (still OpenStack related!).\n\nHaving been at the origin of the project, I have to apologize for the lack of communication of where we were about this project.\nMy apologies for that.\n\nWe still would like to have a place to let the proposal exit, code be reviewed and tested.\nHosting under \"x/\" would work for us.\n\nHope that this can work like this...\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"c30e8e26eb0d916b431b04adb76a9b30601742d6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-16 16:30:01.000000000","message":"Patch Set 3:\n\nYou can host it under \"x\" namespace, nothing here stops you. Just create a new review to import it. That is indepentend from this change.","accounts_in_message":[],"_revision_number":3},{"id":"176951ada7eb23f6d97e0cfa8a093e39d1e88f01","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-18 09:56:19.000000000","message":"Patch Set 3:\n\nAlso if the goal is to keep it around, you could opt to give it a more descriptive org name than the default \"x\" name.","accounts_in_message":[],"_revision_number":3},{"id":"6da57dd210c863114660c18d39a6333af9124a6c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-11-18 15:57:34.000000000","message":"Patch Set 3: Code-Review+1\n\nThomas, this change is just related to the neutron governance. It is unrelated to x/ namespace repository.","accounts_in_message":[],"_revision_number":3},{"id":"796b1d361980a3220ecb741188130a2c7a975050","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-11-18 16:25:13.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"669c2d5d73b1d49b9fc50e0cf59c8b36418d42a5","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-11-20 07:37:20.000000000","message":"Topic changed from neutron-interconnection-retire to project-update","accounts_in_message":[],"_revision_number":3},{"id":"fab56e890b1467446b6a0e6d495968caef448427","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-20 11:40:37.000000000","message":"Patch Set 3:\n\nSo, according to comments from Thomas and others, can we proceed with this patch now and later Thomas will propose to create new repo in x/ namespace if they will need that? Or should I do something else still here?","accounts_in_message":[],"_revision_number":3},{"id":"c12a7ab6b5487919b57372b68810e40c5e3de3bf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-20 14:09:20.000000000","message":"Patch Set 3:\n\nYes, you should be able to continue with this one - and Thomas can propose any time a new change for \"x\" namespace","accounts_in_message":[],"_revision_number":3},{"id":"90ae3cc2e22920290eab717a9d3830beabd5d09f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-21 11:10:43.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"c89d357fc40aef96eb00d330a55b8ce8e1a571c6","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-11-21 14:59:38.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"aeca5327da49f430f6f415fd62099143bf475747","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-11-29 16:44:21.000000000","message":"Patch Set 3: Code-Review+1\n\nOk, neutron-interconnection has now moved to \"x/\" namespace.","accounts_in_message":[],"_revision_number":3},{"id":"197d398a3f158064603946cbfe83bf478c9396c3","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-12-03 14:08:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"772221ab1d027e93782d1b490c1c69b60aed2e97","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-03 14:31:07.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"6d3c318623d3ec3e3292619f7e5ab8e3bf5b9aa7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-12-03 14:59:06.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0c16c4df937a87b5bedce533c3c0e75117e12415","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 14:59:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b6eaa4ac547ec03e7a4fa2b7c647b1077a73dcd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 15:11:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9f5915f74f743f0b3d3dcdeae77f434 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad45fb9f45b4282ba389151a47aaac4 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e66ffc8a1975448a8c7d5f64aacfc877 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1dca144d997e4b26b14fbbfc3dff6109 : SUCCESS in 4m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/be7a1d45292442f5b4ed5f961f1a1464 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":3},{"id":"ae580d7ea29994ac61af067051eb306033b806c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 15:11:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bac3be754f6002003b3315f79b9156bb9949d31d","revisions":{"fd35addd84ea1cc1d285569910be9d72a6b1a90d":{"kind":"REWORK","_number":1,"created":"2019-11-15 08:52:39.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/80/694480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/80/694480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/80/694480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/80/694480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/80/694480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/80/694480/1"}}},"commit":{"parents":[{"commit":"c5b61b68f5fe60b2b3fe9cfeae019b7fd8b2a714","subject":"Merge \"Add goal to drop Python 2.7 Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c5b61b68f5fe60b2b3fe9cfeae019b7fd8b2a714"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 08:51:10.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 08:52:36.000000000","tz":60},"subject":"Rename neutron-interconnection from openstack/ to x/ namespace","message":"Rename neutron-interconnection from openstack/ to x/ namespace\n\nProject neutron-interconnection was officialy in Neutron stadium but as\nthere was no activity on this project for quite long time and project\nactually never got any real implementation done, during the Shanghai PTG\nwe decided to move it out of stadium to x/ namespace.\n\nAlso api-ref for this project is removed from neutron-lib with commit\n[1].\n\nProject rename patch is proposed at [2]\n\n[1] https://review.opendev.org/#/c/694466/\n[2] https://review.opendev.org/#/c/694478/\n\nDepends-On: https://review.opendev.org/#/c/694478/\nChange-Id: I0200b500c4a89c6326d843b8628dfd87dc6eb778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/fd35addd84ea1cc1d285569910be9d72a6b1a90d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/fd35addd84ea1cc1d285569910be9d72a6b1a90d"}]},"branch":"refs/heads/master"},"de41220312c62bd70d240d078977874bd2068ae7":{"kind":"REWORK","_number":2,"created":"2019-11-15 09:15:10.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/80/694480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/80/694480/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/80/694480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/80/694480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/80/694480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/80/694480/2"}}},"commit":{"parents":[{"commit":"c5b61b68f5fe60b2b3fe9cfeae019b7fd8b2a714","subject":"Merge \"Add goal to drop Python 2.7 Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c5b61b68f5fe60b2b3fe9cfeae019b7fd8b2a714"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 08:51:10.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 09:14:55.000000000","tz":60},"subject":"Rename neutron-interconnection from openstack/ to x/ namespace","message":"Rename neutron-interconnection from openstack/ to x/ namespace\n\nProject neutron-interconnection was officialy in Neutron stadium but as\nthere was no activity on this project for quite long time and project\nactually never got any real implementation done, during the Shanghai PTG\nwe decided to move it out of stadium to x/ namespace.\n\nAlso api-ref for this project is removed from neutron-lib with commit\n[1].\n\nProject rename patch is proposed at [2]\n\n[1] https://review.opendev.org/#/c/694466/\n[2] https://review.opendev.org/#/c/694478/\n\nDepends-On: https://review.opendev.org/#/c/694478/\nChange-Id: I0200b500c4a89c6326d843b8628dfd87dc6eb778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/de41220312c62bd70d240d078977874bd2068ae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/de41220312c62bd70d240d078977874bd2068ae7"}]},"branch":"refs/heads/master"},"bac3be754f6002003b3315f79b9156bb9949d31d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-11-15 09:37:38.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/80/694480/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/80/694480/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/80/694480/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/80/694480/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/80/694480/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/80/694480/3"}}},"commit":{"parents":[{"commit":"c5b61b68f5fe60b2b3fe9cfeae019b7fd8b2a714","subject":"Merge \"Add goal to drop Python 2.7 Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c5b61b68f5fe60b2b3fe9cfeae019b7fd8b2a714"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 08:51:10.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 09:37:31.000000000","tz":0},"subject":"Retire neutron-interconnection from openstack/","message":"Retire neutron-interconnection from openstack/\n\nProject neutron-interconnection was officialy in Neutron stadium but as\nthere was no activity on this project for quite long time and project\nactually never got any real implementation done, during the Shanghai PTG\nwe decided to move it out of stadium to x/ namespace.\n\nAlso api-ref for this project is removed from neutron-lib with commit\n[1].\n\nProject rename patch is proposed at [2].\n\nPatch which adds README file to the exising project is available at [3].\n\n[1] https://review.opendev.org/694466/\n[2] https://review.opendev.org/694478/\n[3] https://review.opendev.org/694490/\n\nDepends-On: https://review.opendev.org/#/c/694478/\nChange-Id: I0200b500c4a89c6326d843b8628dfd87dc6eb778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/bac3be754f6002003b3315f79b9156bb9949d31d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/bac3be754f6002003b3315f79b9156bb9949d31d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
