)]}'
{"id":"openstack%2Fgovernance~701486","triplet_id":"openstack%2Fgovernance~master~I31684292f69fdd157c14d06ea1d6b1b31eb0cfb5","project":"openstack/governance","branch":"master","topic":"typo-fix","hashtags":[],"change_id":"I31684292f69fdd157c14d06ea1d6b1b31eb0cfb5","subject":"Fix some typos","status":"MERGED","created":"2020-01-08 01:28:42.000000000","updated":"2020-01-08 11:58:22.000000000","submitted":"2020-01-08 11:56:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701486-1578484560487-893a25b6","meta_rev_id":"2858781173fa427ec4d6991f7f5b15fd6d658e55","_number":701486,"virtual_id_number":701486,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-08 11:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":1,"date":"2020-01-08 11:15:10.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":1,"date":"2020-01-08 11:14:54.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-08 11:15:10.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-01-08 11:56:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1315baac3658f6d9f0c0cb4d06344db8eb073a8b","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-01-08 01:28:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef491646146d78554cd00b08be2542bca8c08e30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 01:34:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34eab3fed2f4deba0ec03fbadb7ff9e : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23095f96d5234f459cf36196d28cdedd : SUCCESS in 3m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8a3a877db1f346a4a9fe96a5ad651e3a : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2bec9b9618e8c6f103fe0d0823e08cac984519b8","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-08 11:04:38.000000000","message":"Topic changed from typo to typo-fix","accounts_in_message":[],"_revision_number":1},{"id":"14f24204ddd79e1b7201cb5d3e91b2425a4eed33","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-08 11:14:54.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"9d5e94654f7e64211ee9fd3f9098f7a998831805","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-08 11:15:10.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53a5f81acc2b43dee658317de644d8928673a374","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 11:15:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bca46aff570deae948fbbbd8afe72f2ab3e481fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 11:56:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01c569289c09497d8240ce3a0749612f : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c792e635984e8aba4fe3bd0e61abba : SUCCESS in 4m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/475897d322f4432b833059f667ff7269 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2a2fad1f8c960653d04cd9fa97ff257bd8f418c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 11:56:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2858781173fa427ec4d6991f7f5b15fd6d658e55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 11:58:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-tc-static https://zuul.opendev.org/t/openstack/build/986f3dc2c2fd42a58ce714c93f661448 : SUCCESS in 1m 49s\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/3e56617f9aa642c883502c41db82773e : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ea2ec0cd49eadfa1dac1c503d707fdbc57d136f","revisions":{"3ea2ec0cd49eadfa1dac1c503d707fdbc57d136f":{"kind":"REWORK","_number":1,"created":"2020-01-08 01:28:42.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/86/701486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/86/701486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/86/701486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/86/701486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/86/701486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/86/701486/1"}}},"commit":{"parents":[{"commit":"188abad6334cd7c8a1dabb2250df1a5d5be1d2af","subject":"Merge \"Add tripleo-operator-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/188abad6334cd7c8a1dabb2250df1a5d5be1d2af"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-01-08 01:02:33.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-01-08 01:02:33.000000000","tz":480},"subject":"Fix some typos","message":"Fix some typos\n\nThis is to fix the following typos:\n  fullfill\n  engourage\n  reponsible\n  peforms\n  conisistency\n  decsion\n  ceilomter\n  evalation\n  does\u0027t\n\nChange-Id: I31684292f69fdd157c14d06ea1d6b1b31eb0cfb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/3ea2ec0cd49eadfa1dac1c503d707fdbc57d136f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/3ea2ec0cd49eadfa1dac1c503d707fdbc57d136f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
