)]}'
{"id":"openstack%2Fgovernance~720107","triplet_id":"openstack%2Fgovernance~master~Ic6d6cc4fa1a2bc939c17e8aa63ac67aabc4403c9","project":"openstack/governance","branch":"master","topic":"formal-vote","hashtags":[],"change_id":"Ic6d6cc4fa1a2bc939c17e8aa63ac67aabc4403c9","subject":"goals: add container-images","status":"ABANDONED","created":"2020-04-15 00:43:18.000000000","updated":"2020-05-13 15:27:04.000000000","total_comment_count":43,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f6f63886b949005574b86582b27134956d9495c3","_number":720107,"virtual_id_number":720107,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-04-16 19:33:25.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"date":"2020-04-24 14:32:55.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"date":"2020-04-24 07:49:00.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-15 01:18:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"date":"2020-04-23 23:32:15.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-04-16 10:14:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2020-04-18 22:29:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":-1,"date":"2020-04-26 17:20:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":-1,"date":"2020-04-20 22:06:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":-1,"date":"2020-04-16 12:10:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"all":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"date":"2020-04-23 19:39:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"date":"2020-04-16 09:09:02.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"date":"2020-04-30 15:54:15.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-15 01:05:18.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-04-15 01:18:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-16 09:09:02.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2020-04-16 10:14:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-16 12:10:11.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-16 19:33:25.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-04-18 22:29:06.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-04-20 22:06:16.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-04-23 19:39:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-23 23:32:15.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-04-24 07:49:00.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-04-26 17:20:16.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-04-30 15:54:15.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"c830f27853df328aa14484d23d1290d13d8ae805","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-15 00:43:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccb1d28febfe03baff86067e2a40bda628405472","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-15 00:47:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"82c48d60435dd012a91a019d54c4cd8da9fc173c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 01:02:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a38265fa5e74f2cb9effab0f54a94f9 : FAILURE in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8d2a070c12a4c2c9edcd38ad9ba83bc : SUCCESS in 5m 19s\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/9c2b01d7707e4c24b703040e7cb293a1 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e46dfb93d54d46a8afc1363be178c26c : SUCCESS in 4m 28s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/goals/selected/victoria/container-images.rst in the commit\u0027\n  Comments left for invalid file doc/source/goals/selected/victoria/container-images.rst","accounts_in_message":[],"_revision_number":2},{"id":"0cc65f104d281d033d83779448babd8c11277180","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-15 01:05:18.000000000","message":"Patch Set 2: Code-Review-1\n\nThanks Naseer for this. we need to add it in goal/proposed (instead of goal/selected/victoria) folder first and then we select one out of those for the V cycle. I have added this goal entry for V cycle goal etherpad:\n\nhttps://etherpad.opendev.org/p/YVR-v-series-goals","accounts_in_message":[],"_revision_number":2},{"id":"dd271be78b1638b1a0ba5ff57374209e63c3f597","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-15 01:08:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7b0eb87b11e0e7e56905e07325367ab3bac3b8fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 01:18:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc3142cf3e09452fa291165f32f6c78c : FAILURE in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe1b0c9fa9154d428e85f11930d84ab6 : SUCCESS in 3m 33s\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/fbc69d4cb21f4e98aa0144a0474be405 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a9fe8cc0afe3495eb424e0112eddeaac : SUCCESS in 3m 48s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/goals/proposed/container-images.rst in the commit\u0027\n  Comments left for invalid file doc/source/goals/proposed/container-images.rst","accounts_in_message":[],"_revision_number":3},{"id":"2ddc64102c4975be7a8c5b58ca47bd27912e9667","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-16 09:24:47.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"394b83b3edfbdeb6b7442b417578016dbfbf647d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-04-16 10:14:47.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"21fe76edee8f2eb12d71a163af47d33c7c453365","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-16 12:10:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nI think it needs a bit more preparatory work to run it as a goal. While I agree with the main theme (and like the bindep part of PoC), I have some concerns (please see the inline comment).\n\nAlso, what about Ubuntu Charms? And OpenStack-Helm? They also containerize OpenStack. Obviously they are not image-building-oriented but they certainly achieve the goal of having services in containers. Is there a way for us to coordinate the discussion and share the experience?","accounts_in_message":[],"_revision_number":3},{"id":"878ca038396c5193ffc067de2c40ebcb93985773","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-04-16 12:47:46.000000000","message":"Topic set to formal-vote","accounts_in_message":[],"_revision_number":3},{"id":"b43630235898179ce674e53e5ed414ecc8290198","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-16 14:23:22.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThanks for your comments.  I plan to hack on what the images all look like with mordred at some point.\n\nUbuntu Charms uses snaps I think?  Helm relies on Loci images but I think it\u0027s possible for them to easily re-use those images with almost no impact IMHO.\n\nI do think we should share all of our experience there, there\u0027s a lot of different drivers in here.","accounts_in_message":[],"_revision_number":3},{"id":"79a41f96ae7ad00b445f529583a67c1b5d175474","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-16 14:52:37.000000000","message":"Patch Set 3:\n\n(1 comment)\n\n\u003e Thanks for your comments.  I plan to hack on what the images all\n \u003e look like with mordred at some point.\n\nThanks for good discussion. Ack, count me in technical stuff.\n\n \u003e Ubuntu Charms uses snaps I think?  Helm relies on Loci images but I\n \u003e think it\u0027s possible for them to easily re-use those images with\n \u003e almost no impact IMHO.\n\nHonestly no idea. :-) As for Helm, they did use Kolla at some point (when dinosaurs were around), then LOCI but nowadays I\u0027m not even sure about that.\n\n \u003e I do think we should share all of our experience there, there\u0027s a\n \u003e lot of different drivers in here.\n\nFor sure, looking forward to kollaboration (kolla pun intended).","accounts_in_message":[],"_revision_number":3},{"id":"6da205c96b806bf093ea210e0d80759479454dfa","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-16 14:57:33.000000000","message":"Removed Code-Review+1 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"538fc70912c6547413891e4d91dc844288abfce8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-16 14:59:57.000000000","message":"Patch Set 3:\n\nI have read your comments. The idea of simple, single container per openstack component may work for some projects.\n\nBut once distro dependencies are involved hell will open. Try to get Nova running without libvirt and qemu. Or handle x86-64/aarch64/ppc64le in sane way.\n\nI will watch goes it goes.","accounts_in_message":[],"_revision_number":3},{"id":"42cf8006c2c665442b3f8163ffcc49b57e674dc6","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-16 15:11:17.000000000","message":"Patch Set 3:\n\n(1 comment)\n\n\u003e I have read your comments. The idea of simple, single container per\n \u003e openstack component may work for some projects.\n \u003e \n \u003e But once distro dependencies are involved hell will open. Try to\n \u003e get Nova running without libvirt and qemu. Or handle\n \u003e x86-64/aarch64/ppc64le in sane way.\n\nYou do bring up a good point here and we can probably use multiarch images with manifests so that it automatically loads the right architecture \n\n \u003e I will watch goes it goes.\n\nI think the idea is that libvirt will *not* be running inside the container, neither will be shipping a libvirt image.  The idea is that you would run your own libvirt, you\u0027d take care of doing bind mounts, you can containerize libvirt or run it on the host, but that\u0027s *entirely* up to the user, think of this as shipping pypi wheels (with binary dependencies), nothing more.","accounts_in_message":[],"_revision_number":3},{"id":"168bec1a99f2dcc9af0ae356eb35c3df0505eb21","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-16 15:20:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d2e5ad5d90fa2ec766da3cf1b7a04abb1042f27a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-16 17:26:38.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThank you for your comments Mark!","accounts_in_message":[],"_revision_number":3},{"id":"01694d3897858d32c8b427439f4c0bc5e14ecbb2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-16 18:40:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a944fbe6936b5bae1ed0acfe3b1b411f47ebe62","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-16 19:30:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1988bd62c85aabc4aaf5a488ab58544ef94711cd","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-16 19:33:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1b54995e2b9f4d6c434e7bc0f96fdbddf57e3fbe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-16 20:00:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ff2a97c39ff167a91e5bc84a137183e60127f4ab","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-16 20:56:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"44c4486e8a9d19952158b1aa25332fd3bee96546","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-18 15:26:42.000000000","message":"Patch Set 3:\n\nA bit more on Kolla (as promised):\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-April/014255.html","accounts_in_message":[],"_revision_number":3},{"id":"4c2130ec483f47170787643f42794801d9e5474e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-18 22:29:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nI like the idea of \u0027image per project\u0027. But it just does not make sense for so many projects where you need all those distro dependencies to get it running.","accounts_in_message":[],"_revision_number":3},{"id":"fbc4fca37684ef655fdd537ac24a8fcac15647fe","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-20 19:53:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a4d546f80f7e50c9f669a02ebeb1582795969b2","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-04-20 22:06:16.000000000","message":"Patch Set 3: Code-Review-1\n\n(9 comments)\n\nin general in not really a fan of this goal as written as i dont think it will be useful for either casual end-users trying to get something working quick or for production.\n\ndoing this at the project level will put an increased burden on the core teams to mainain a deployment solution that will not be useful in there day to day dev workflow and will also either need ot be hevely customisable to support all user of the project or highly opinionated alienating others.\n\nwith that said i do think there is a related goal to this\n\ne.g. having a quick and easy way to get started with each project.\nmaybe using containers to help achive that but i think loci and kolla but do that better today.","accounts_in_message":[],"_revision_number":3},{"id":"d51d1a9c7c6c5a116c30d2276e7e182111651b49","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-04-20 22:47:05.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"65f6cbe095e53c30c9a04995f5a18b03e0827685","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-21 13:14:48.000000000","message":"Patch Set 3:\n\n(10 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d0b890e7ffbe66b54684231182722ab0da715ef5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-21 13:30:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b8633648707671a6741db0788b63236d227ff817","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-21 19:08:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"55b035645225117689db692da50b722615a4703c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-04-21 19:30:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b405fea96822fc3c305815e97ae014df39e2979b","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-04-23 15:34:31.000000000","message":"Patch Set 3: Rollcall-Vote+1\n\nAdding my two cents here.  I am not much of an expert with containers  so not going to dig into technical details.  It kind of looks like from the discussion in here that there is a lot of bikeshedding going on.  I think the idea is good and would hate to see perfect be the enemy of good here and prevent any progress.\n\nIf we can start by getting something simple in place for each of the core projects and then iterate from there I think it would be great.","accounts_in_message":[],"_revision_number":3},{"id":"be1cdefbc4199ce008c4fdb5ec3aa0cd1df1b3da","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 19:39:17.000000000","message":"Patch Set 3:\n\n\u003e Adding my two cents here.  I am not much of an expert with\n \u003e containers  so not going to dig into technical details.  It kind of\n \u003e looks like from the discussion in here that there is a lot of\n \u003e bikeshedding going on.  I think the idea is good and would hate to\n \u003e see perfect be the enemy of good here and prevent any progress.\n \u003e \n \u003e If we can start by getting something simple in place for each of\n \u003e the core projects and then iterate from there I think it would be\n \u003e great.\n\nThere is some bikeshedding for sure, but there is also some politics. We already have two projects that create containers for OpenStack - Kolla and LOCI. Let\u0027s not add a third set of images lightly.\n\nI\u0027m not here to block this proposal (even if I could), but I do want to make sure that the facts are presented, and that the reasons for doing this are valid. Through discussion we\u0027ve unearthed some misconceptions about Kolla, and dug into the reasons behind this proposal that are not covered in the text.\n\nIf this was a new project proposal I would be in more of a mindset to let it compete, and let it play out. However a goal is something all projects are strongly encouraged to implement, and has a certain \u0027blessed\u0027 nature.\n\nThis proposal has triggered some good discussion, and I think there\u0027s probably a solution where we meet in the middle. I would like to see us come to agreement on the direction, or at least agree to disagree :)","accounts_in_message":[],"_revision_number":3},{"id":"a62fa6a355739e5de046a8d293e6d1e4d907cc55","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-04-23 23:32:15.000000000","message":"Patch Set 3:\n\nI\u0027m not going to weigh in on whether or not this makes sense for OpenStack, but did want to try and provide a summary for how this setup is different from what OpenStack is already doing. I put this in IRC and realize now a few hours later it is worth putting on the change.\n\nThe first thing is using the speculative building, testing, gating, and promoting of container images that Zuul has built out. This is basically a framework of jobs and tools (like the zuul registry) that allows you to easily test containers in the same way Zuul tests git commits. Kolla and Loci and whoever else should be able to easily reconsume this set of tooling so this isn\u0027t unique, but I think it is an important piece of properly testing and publishing images.\n\nNext is the system OpenDev and Zuul has built doesn\u0027t try to provide images for every distro or any distro. It uses the python-slim images published to docker since we are building software for python. This drastically reduces the problem space, simplifies testing, and reduces debugging overhead. It will use whatever distro python has determined is best for python. I think this is a fairly big departure from the existing container projects.\n\nThe system OpenDev and Zuul have built assumes \"well formed\" python projects that publish bindep dependencies and can be pip installed. What this allows us to do is use a standard set of base images to build all of the python artifacts, stage them on a throwaway container, then install them into the final image that is actually published. This helps keep image size down. This also means you don\u0027t have to write a bunch of custom docker code for each python project. They are expected to drop in and just work.\n\nFinally, the end result of this is more like having a package of each project and not a full deployment. End users are expected to provide configuration and orchestration and whatever else is necessary to actually operate the software.","accounts_in_message":[],"_revision_number":3},{"id":"9599e6a3b69f614bf8fb3ba712e6f54bb2e9b1a1","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-04-24 07:45:53.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8fe4b41b3959e07a9c9f69545ed829f56e2a02e8","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-04-24 07:49:00.000000000","message":"Patch Set 3:\n\n\u003e I\u0027m not going to weigh in on whether or not this makes sense for\n \u003e OpenStack, but did want to try and provide a summary for how this\n \u003e setup is different from what OpenStack is already doing. I put this\n \u003e in IRC and realize now a few hours later it is worth putting on the\n \u003e change.\n \u003e \n \u003e The first thing is using the speculative building, testing, gating,\n \u003e and promoting of container images that Zuul has built out. This is\n \u003e basically a framework of jobs and tools (like the zuul registry)\n \u003e that allows you to easily test containers in the same way Zuul\n \u003e tests git commits. Kolla and Loci and whoever else should be able\n \u003e to easily reconsume this set of tooling so this isn\u0027t unique, but I\n \u003e think it is an important piece of properly testing and publishing\n \u003e images.\n \u003e \n \u003e Next is the system OpenDev and Zuul has built doesn\u0027t try to\n \u003e provide images for every distro or any distro. It uses the\n \u003e python-slim images published to docker since we are building\n \u003e software for python. This drastically reduces the problem space,\n \u003e simplifies testing, and reduces debugging overhead. It will use\n \u003e whatever distro python has determined is best for python. I think\n \u003e this is a fairly big departure from the existing container\n \u003e projects.\n \u003e \n \u003e The system OpenDev and Zuul have built assumes \"well formed\" python\n \u003e projects that publish bindep dependencies and can be pip installed.\n \u003e What this allows us to do is use a standard set of base images to\n \u003e build all of the python artifacts, stage them on a throwaway\n \u003e container, then install them into the final image that is actually\n \u003e published. This helps keep image size down. This also means you\n \u003e don\u0027t have to write a bunch of custom docker code for each python\n \u003e project. They are expected to drop in and just work.\n \u003e \n \u003e Finally, the end result of this is more like having a package of\n \u003e each project and not a full deployment. End users are expected to\n \u003e provide configuration and orchestration and whatever else is\n \u003e necessary to actually operate the software.\n\nI think that summary is clear and noble. I like it. I just would have prefered a staged approach:\n1) We clarify the bindep\u0027s scope, and make sure it\u0027s consistent across projects\n2) We use this tooling to build and publish images\n3) We work as testing those together\n1-2-3\u0027) We try the crazy idea of Sean in parallel :)","accounts_in_message":[],"_revision_number":3},{"id":"5e6076644fc6f007d66771b27e2fd85677dd5296","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-24 08:16:19.000000000","message":"Patch Set 3:\n\nOK, then all interested parties agree that better bindep is key (considering discussions both here and on the ml). Making bindep right and consistent would help both Kolla and any new endeavor related to packaging/containerization. However, bindep really should consider the different distros and their supported versions, rather than being usable only by python images based on debian buster. Hence, I would best see this community goal as making bindep shine and using it (more intensively) for target systems (DevStack, Kolla, OpenStack-Helm/LOCI, new containers, traditional packaging etc.).","accounts_in_message":[],"_revision_number":3},{"id":"8765e110126604b7b5008a92f4eb8a952966dcf9","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-24 14:32:55.000000000","message":"Patch Set 3:\n\nI think the concern here is how to test bindep -- especially for things like build-time dependencies and run-time dependencies.  With DevStack, we run everything in one VM so it\u0027s impossible to track that accurately.","accounts_in_message":[],"_revision_number":3},{"id":"3a85bb169c7d7e8d82adfda976ac9fba1257b412","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-24 14:42:00.000000000","message":"Patch Set 3:\n\nIMHO DevStack could abandon some layers of extra package installation code if we did bindep just right. Do remember we might want to touch bindep as a tool as well. This traditional usage could happen like this: collect all services to install (in terms of repos, DevStack does it), collect bindep from them all, install what is required for building (based on whatever extra filters we imagine), build and install services, run bindep again, now for runtime, run services. This is still better than hacks we have and tests the coverage somehow. Still, in containerised that would be more apparent and accurate. I\u0027m just emphasizing that bindep is the thing we might want to tackle as a community which brings the most benefits to plans of us all.","accounts_in_message":[],"_revision_number":3},{"id":"0c767d98ada7503920ca9796b1bb9d99526668e5","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-04-26 17:20:16.000000000","message":"Patch Set 3: Code-Review-1\n\nFrom Zun\u0027s perspective, we don\u0027t want to spend time to implement this goal. First, there are extra overhead to maintain such container images. Second, the use cases of this goal doesn\u0027t apply to Zun project. Itt is not clear how this goal will benefit the Zun project (or the potential benefit is not clear to me).","accounts_in_message":[],"_revision_number":3},{"id":"77337dbb81e41b4c758a96025dc19f6819e0b1a7","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-04-30 15:54:15.000000000","message":"Patch Set 3: -Rollcall-Vote\n\nHave gotten feedback from the community that they have concerns with this.  Think this needs more conversation before becoming a goal.","accounts_in_message":[],"_revision_number":3},{"id":"f0334139a24048f76b21a2a3d5df919867e44756","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-05-10 17:50:44.000000000","message":"Abandoned\n\ni give up :)\n\ni\u0027ll just submit patches and whatever projects accept it can accept it, and those that don\u0027t, meh.\n\ni\u0027ll maybe repropose this once we have a few projects implemented it and we show how this works.","accounts_in_message":[],"_revision_number":3},{"id":"07ce12abcf6cd2025211f87fdc74ff12a89758d6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-11 07:09:15.000000000","message":"Patch Set 3:\n\n\u003e Abandoned\n \u003e \n \u003e i give up :)\n \u003e \n \u003e i\u0027ll just submit patches and whatever projects accept it can accept\n \u003e it, and those that don\u0027t, meh.\n \u003e \n \u003e i\u0027ll maybe repropose this once we have a few projects implemented\n \u003e it and we show how this works.\n\nSorry to see it being abandoned but it seems right to do a more realistic poc to validate the benefits before going community goal. Anyways, please let me know if you need help and feel welcome to add me as a reviewer to any related patches. I feel bad for not being able to follow up on the common ideas we have but time constraints are always an issue.","accounts_in_message":[],"_revision_number":3},{"id":"f6f63886b949005574b86582b27134956d9495c3","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2020-05-13 15:27:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"748d44ead3edc8de7017fdd0c7d5dec9d8e8c1ea","revisions":{"dd2c708d1e199ca3fb82225f4df2d8cb0f0133b1":{"kind":"REWORK","_number":1,"created":"2020-04-15 00:43:18.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/07/720107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/07/720107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/07/720107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/07/720107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/07/720107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/07/720107/1"}}},"commit":{"parents":[{"commit":"ed73db7af1479151b1654245d308488eef50e5c4","subject":"Merge \"OpenStack-Helm does not require release management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/ed73db7af1479151b1654245d308488eef50e5c4"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-04-15 00:42:41.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-04-15 00:42:41.000000000","tz":-240},"subject":"goals: add container-images","message":"goals: add container-images\n\nThis patch proposes a goal for Victoria to provide container images\nusing the system we use inside OpenDev.\n\nChange-Id: Ic6d6cc4fa1a2bc939c17e8aa63ac67aabc4403c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/dd2c708d1e199ca3fb82225f4df2d8cb0f0133b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/dd2c708d1e199ca3fb82225f4df2d8cb0f0133b1"}]},"branch":"refs/heads/master"},"8c696b1787497956ed8e7e42933d111772cbf7a4":{"kind":"REWORK","_number":2,"created":"2020-04-15 00:47:04.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/07/720107/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/07/720107/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/07/720107/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/07/720107/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/07/720107/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/07/720107/2"}}},"commit":{"parents":[{"commit":"ed73db7af1479151b1654245d308488eef50e5c4","subject":"Merge \"OpenStack-Helm does not require release management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/ed73db7af1479151b1654245d308488eef50e5c4"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-04-15 00:42:41.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-04-15 00:46:58.000000000","tz":-240},"subject":"goals: add container-images","message":"goals: add container-images\n\nThis patch proposes a goal for Victoria to provide container images\nusing the system we use inside OpenDev.\n\nChange-Id: Ic6d6cc4fa1a2bc939c17e8aa63ac67aabc4403c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/8c696b1787497956ed8e7e42933d111772cbf7a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/8c696b1787497956ed8e7e42933d111772cbf7a4"}]},"branch":"refs/heads/master"},"748d44ead3edc8de7017fdd0c7d5dec9d8e8c1ea":{"kind":"REWORK","_number":3,"created":"2020-04-15 01:08:37.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/07/720107/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/07/720107/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/07/720107/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/07/720107/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/07/720107/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/07/720107/3"}}},"commit":{"parents":[{"commit":"ed73db7af1479151b1654245d308488eef50e5c4","subject":"Merge \"OpenStack-Helm does not require release management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/ed73db7af1479151b1654245d308488eef50e5c4"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-04-15 00:42:41.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-04-15 01:08:31.000000000","tz":-240},"subject":"goals: add container-images","message":"goals: add container-images\n\nThis patch proposes a goal for Victoria to provide container images\nusing the system we use inside OpenDev.\n\nChange-Id: Ic6d6cc4fa1a2bc939c17e8aa63ac67aabc4403c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/748d44ead3edc8de7017fdd0c7d5dec9d8e8c1ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/748d44ead3edc8de7017fdd0c7d5dec9d8e8c1ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
