)]}'
{"id":"openstack%2Fgovernance~720109","triplet_id":"openstack%2Fgovernance~master~I4cc0bd52e1029931f5257e645f11226957d23d82","project":"openstack/governance","branch":"master","topic":"election-results","hashtags":[],"change_id":"I4cc0bd52e1029931f5257e645f11226957d23d82","subject":"Update Reference with Victoria Election Results","status":"MERGED","created":"2020-04-15 00:49:30.000000000","updated":"2020-04-16 15:15:38.000000000","submitted":"2020-04-16 15:12:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720109-1587049958207-2391f812","meta_rev_id":"0c83f90fd5bd2aaf7bf2f4e5f35f0ae76eac154f","_number":720109,"virtual_id_number":720109,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-16 15:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2020-04-16 15:01:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2020-04-16 15:03:02.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2020-04-16 14:57:50.000000000","_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2020-04-16 14:57:03.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2020-04-16 15:03:02.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 04:44:16.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2020-04-16 14:57:03.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-04-16 14:57:50.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2020-04-16 15:01:35.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-04-16 15:03:02.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-04-16 15:12:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ccf7f6d7cf3afa5a9faee66f2004f6a0a9211cb","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-04-15 00:49:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25b26053ec2e7d9dc802e03a98a5b44f870aadd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 01:08:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b60aaeaea6094d70895d0678224f9030 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f2d75e46db44a9a9ffc129e594e6e0f : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87ff2053f2ca4feaa00b4fb7b3ec039a : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1326940ca4784e118c637e6c0a23a037 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/093bd731ee5443cabf369cdb45a60c41 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/95f840630ab34da29fd717b5feac7e5d : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cdf69d886e2e4e2a8f6868c0c6140107 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"467e3e766c020924d8c960086d15c103f02c9874","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-04-15 11:10:19.000000000","message":"Topic changed from v_tc_election_results to election-results","accounts_in_message":[],"_revision_number":1},{"id":"0440a1c8bc292e1ac4d90de731a21a22518e2ae8","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-04-15 11:19:56.000000000","message":"Patch Set 1: Rollcall-Vote+1\n\nSeems to reflect the results of https://civs.cs.cornell.edu/cgi-bin/results.pl?id\u003dE_95cb11614fb23566 .","accounts_in_message":[],"_revision_number":1},{"id":"f4276cd60aaa3323d1f99c8a4c0c44eee51a08e5","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-04-15 12:21:54.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"7f0d3ab62acd9a0bb17c3ef6e1cc8d275f81f25f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-15 15:11:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b682f2d5ac5368e553ff8b41de83a909571722f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2020-04-16 04:44:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"295469be530961d3f7a288582a035b08de7067c5","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-04-16 07:38:39.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n\nAgreed with Tony there.\nGmann, do you mind removing your -1, and I will take care of the extra update/code-change to update this?\n\nIf you\u0027re okay, we can proceed ahead, and validate the TC members, which sounds the higher priority to get the whole ball running (chair stuff, etc.)","accounts_in_message":[],"_revision_number":1},{"id":"e9693004bf381ae87f45eda6faf4be7c8e8fa047","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-16 13:42:05.000000000","message":"Patch Set 1:\n\n\u003e \u003e (1 comment)\n \u003e \n \u003e Agreed with Tony there.\n \u003e Gmann, do you mind removing your -1, and I will take care of the\n \u003e extra update/code-change to update this?\n \u003e \n \u003e If you\u0027re okay, we can proceed ahead, and validate the TC members,\n \u003e which sounds the higher priority to get the whole ball running\n \u003e (chair stuff, etc.)\n\nok changing table and all is ok for me to do in followup but let\u0027s leave \u0027term\u0027 blank in this patch because conveying the term which is going to change can be confusing for many people.","accounts_in_message":[],"_revision_number":1},{"id":"1f02de1b4d7d4adda80b755d74584649ba39dd08","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-04-16 14:45:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"701086e4213a0b8413a3dcb1b9f645819d8084cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 14:54:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6b6fd2d3d7544928976fbddd5eb4ac6 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/653077c5d3bf4b09a88515df10152220 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b784f8bec9784fafae9349dd996fc22e : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd46d9cc353a4ad68c40bb048881848f : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e41da2fa8216454dbcbf527218301621 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/1f904c843da6411787bc6a3137ef75aa : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b86fd829f9334d17ad93827fe60d1771 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"49bf05b2ba37d2fcf62a66496b3992b4a7b90762","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-16 14:57:03.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\nThanks, lgtm now and let\u0027s work on better structure of this table in https://review.opendev.org/#/c/720477/1","accounts_in_message":[],"_revision_number":2},{"id":"825a9bd038b0752929d0426ac5712d984b6b8f65","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-04-16 14:57:50.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"106f8d5bd6acd9068405192bd0c50c01cd311b10","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-04-16 15:01:35.000000000","message":"Patch Set 2: Code-Review+1\n\nAs an election official, I support the current patch to move forward.","accounts_in_message":[],"_revision_number":2},{"id":"e526be3deb0b68a54e6b4c839c82f1bff3b28f54","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-04-16 15:03:02.000000000","message":"Patch Set 2: Rollcall-Vote+1 Workflow+1\n\nLet\u0027s merge this!","accounts_in_message":[],"_revision_number":2},{"id":"475b0dcc8dcbd13ccd3f3f25f865a3dc8e9122e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 15:03:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bba76bcb978858fb0193fd6f17852aaec125801f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 15:12:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4bea2715774df2b94af84bc059c478 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/212af227e0764eb98068c90518937fed : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/810c3bd606a7458eacf0ff6919ae01c6 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7da5ac67bafd4817a939539f194dfb89 : SUCCESS in 5m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/48b1e3870ab2435994cf2c47a7abf550 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"ce393745d524de2a6186eada109900874510915f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 15:12:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0c83f90fd5bd2aaf7bf2f4e5f35f0ae76eac154f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 15:15:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/e4e88f22161e49af8a250e61fc91f9cd : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ae1667d456ab052d57025fd8d121542f79c790fb","revisions":{"5f7292477fe2c294fd180c8d1bbb494ea92fcf48":{"kind":"REWORK","_number":1,"created":"2020-04-15 00:49:30.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/09/720109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/09/720109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/09/720109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/09/720109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/09/720109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/09/720109/1"}}},"commit":{"parents":[{"commit":"6c91b455c5430f2d2614d9614aac095f1af68246","subject":"Merge \"Handling the OpenStack name in external services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/6c91b455c5430f2d2614d9614aac095f1af68246"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-04-15 00:47:25.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-04-15 00:49:24.000000000","tz":-420},"subject":"Update Reference with Victoria Election Results","message":"Update Reference with Victoria Election Results\n\nChange-Id: I4cc0bd52e1029931f5257e645f11226957d23d82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/5f7292477fe2c294fd180c8d1bbb494ea92fcf48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/5f7292477fe2c294fd180c8d1bbb494ea92fcf48"}]},"branch":"refs/heads/master"},"ae1667d456ab052d57025fd8d121542f79c790fb":{"kind":"REWORK","_number":2,"created":"2020-04-16 14:45:41.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/09/720109/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/09/720109/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/09/720109/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/09/720109/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/09/720109/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/09/720109/2"}}},"commit":{"parents":[{"commit":"6c91b455c5430f2d2614d9614aac095f1af68246","subject":"Merge \"Handling the OpenStack name in external services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/6c91b455c5430f2d2614d9614aac095f1af68246"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-04-15 00:47:25.000000000","tz":-420},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-04-16 14:44:44.000000000","tz":120},"subject":"Update Reference with Victoria Election Results","message":"Update Reference with Victoria Election Results\n\nChange-Id: I4cc0bd52e1029931f5257e645f11226957d23d82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/ae1667d456ab052d57025fd8d121542f79c790fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/ae1667d456ab052d57025fd8d121542f79c790fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
