)]}'
{"id":"openstack%2Fgovernance~720545","triplet_id":"openstack%2Fgovernance~master~I03903891fcf83151d6b0010900a382984fca2e81","project":"openstack/governance","branch":"master","topic":"election-results","hashtags":[],"change_id":"I03903891fcf83151d6b0010900a382984fca2e81","subject":"Add nomination for TC chair","status":"MERGED","created":"2020-04-16 16:19:48.000000000","updated":"2020-04-21 08:59:53.000000000","submitted":"2020-04-21 08:58:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720545-1587459504172-6b5ca73b","meta_rev_id":"288a09f238698a92e8d87e9bd4f373435bdbd26a","_number":720545,"virtual_id_number":720545,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-21 08:58:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2020-04-21 08:50:23.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"all":[{"value":1,"date":"2020-04-21 08:46:45.000000000","_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2020-04-21 07:48:50.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-21 08:46:45.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2020-04-21 08:50:23.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-04-21 08:58:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfaeedb6f4de954fa2cf0255d4d9c553467e6a01","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-16 16:19:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ac82291a43e19fb5fa3c0850102229d9d3eff20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 16:32:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a3ab9b2dc2d4b3eaed9a1a09a9474b3 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a029727fb98d470b8d49f67a9989f2a6 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e29dc68ee7f49a394ad089555d9f8c6 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/341fa86d22f748f5bdf3a272e58e3375 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef29f915ad8d4d2ab6cdac73a8d061dc : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/33f4ec13b39c4f5180eb89cbbf7a1b90 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/399aea638c3144d68dda4afe404f7940 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c42f9ad234de6228beac7ab9a96a1b6886d2994a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-04-21 07:48:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e39f015907fc9b1af455b4f7791dc4df3989e03","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-04-21 07:48:44.000000000","message":"Topic set to election-results","accounts_in_message":[],"_revision_number":2},{"id":"00700efc83e7f07a2c63790a585dbc18d5ad50bd","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-04-21 07:48:50.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"65044543343b9df539389accf77946b2edffe7cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 07:58:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9f9c8b57f3a453285a3b19b94dbd4f0 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a319901b6d24299b224a645793ef614 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7562f8d48adc428394892cb43212ba41 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7fafd250ea0a40f5b320a18a1924c259 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8602bf667be040328aca295af101c724 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/4c34c00472484947b22cbdcc1076c46f : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2790bbd9687d42bb8ad40e0ff544527a : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":2},{"id":"68c31a4c9ab00ad4cc5d5816c7cb20b958df0721","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-04-21 08:46:45.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"ff3ec05804807672d0788eea1235bc4ec7e2f149","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-04-21 08:50:23.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"72da27025b76078ed12505ad9f51309fed36e1af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 08:50:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"afc6506e115535f6c754533636e3c9ec37262e3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 08:58:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e11424e093db485b9fe6adcea44e9cfb : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57c0771c7eef4c4c99397a028995345f : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b04f06f03e924525b4321817ff73dc1d : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b4f13737b6b421ea4ee68b4e703d51e : SUCCESS in 5m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/943cf2fa4f544986a9dae2b97b865e0f : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"fb860a7531e3d5eb1bd5be9501bf57fca88dbb70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 08:58:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"288a09f238698a92e8d87e9bd4f373435bdbd26a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 08:59:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/e7abf84d399e427ba53ec51e7c3d7fcb : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ffe9886325f9c307280f7132da1ef6b9f0d08f78","revisions":{"6dca7e821b1b588860898acf8bd75733f8a602dc":{"kind":"REWORK","_number":1,"created":"2020-04-16 16:19:48.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/45/720545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/45/720545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/720545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/720545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/720545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/45/720545/1"}}},"commit":{"parents":[{"commit":"727f6435516f36564f37b5acbcf4be39db35d0d7","subject":"Merge \"Remove projects\u0027 TC liaisons\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/727f6435516f36564f37b5acbcf4be39db35d0d7"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-04-16 16:17:22.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-04-16 16:17:22.000000000","tz":-240},"subject":"Add nomination for TC chair","message":"Add nomination for TC chair\n\nI\u0027d like to run for the chair of the technical commitee for the upcoming\ncycle, I\u0027ve served in the past and helped drive discussion and keep up\nwith the progress of the technical commitee, thanks to the outreach and\nhelp from Doug Hellman at the time.\n\nI plan on continuing to push us to innovate more and drive more changes\nthrough the TC, while simplifying our process.  I intend to go back to\nmy weekly updates that I would post to our mailing list to keep our\ncommunity updated.\n\nI personally also think that there is benefit in being a board member as\nwell so I can push feedback directly from the TC into the board\nmeetings.\n\nChange-Id: I03903891fcf83151d6b0010900a382984fca2e81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/6dca7e821b1b588860898acf8bd75733f8a602dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/6dca7e821b1b588860898acf8bd75733f8a602dc"}]},"branch":"refs/heads/master"},"ffe9886325f9c307280f7132da1ef6b9f0d08f78":{"kind":"REWORK","_number":2,"created":"2020-04-21 07:48:20.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/45/720545/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/45/720545/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/720545/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/720545/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/720545/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/45/720545/2"}}},"commit":{"parents":[{"commit":"b5290de2f6738b60ce433542ebd9f01fd6c7c8ec","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/b5290de2f6738b60ce433542ebd9f01fd6c7c8ec"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-04-16 16:17:22.000000000","tz":-240},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-04-21 07:48:03.000000000","tz":120},"subject":"Add nomination for TC chair","message":"Add nomination for TC chair\n\nI\u0027d like to run for the chair of the technical commitee for the upcoming\ncycle, I\u0027ve served in the past and helped drive discussion and keep up\nwith the progress of the technical commitee, thanks to the outreach and\nhelp from Doug Hellman at the time.\n\nI plan on continuing to push us to innovate more and drive more changes\nthrough the TC, while simplifying our process.  I intend to go back to\nmy weekly updates that I would post to our mailing list to keep our\ncommunity updated.\n\nI personally also think that there is benefit in being a board member as\nwell so I can push feedback directly from the TC into the board\nmeetings.\n\nChange-Id: I03903891fcf83151d6b0010900a382984fca2e81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/ffe9886325f9c307280f7132da1ef6b9f0d08f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/ffe9886325f9c307280f7132da1ef6b9f0d08f78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
