)]}'
{"id":"openstack%2Fgovernance~720552","triplet_id":"openstack%2Fgovernance~master~I7116e848f113a9842785c01530fec018570490dd","project":"openstack/governance","branch":"master","topic":"mugsie-tc-chair-nomination","hashtags":[],"change_id":"I7116e848f113a9842785c01530fec018570490dd","subject":"Add mugsie nomination for TC Chair","status":"ABANDONED","created":"2020-04-16 16:48:03.000000000","updated":"2020-04-22 10:10:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"20e9bce3140cdef2e807dc2b97cf8952cae28b6b","_number":720552,"virtual_id_number":720552,"owner":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-16 17:00:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 17:00:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f304c6221bdc91436a992aff9418d4b283b4b77","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-04-16 16:48:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe04504dbd5d966948fd912a0944ae28cc3886c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 17:00:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed2baccba45a4927ad41b4e32018e0a4 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4e6771ca074541a02b347c506a83f1 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbf6f924bd1d4139a04f8df58f8529a6 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29abe5c00bd24848babff1cc1d7fadab : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7a0226d68304b2392e732e7c15026bd : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/1d3e1c6741154fa08c1bb212d82f885f : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7e50bae312aa41b7b1e118e557806952 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2915b42d0baf6634206c0b97d442326b4e7c2eba","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-04-22 10:10:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"101760ae87937beaa66b70d6f30c2e8c102277f9","revisions":{"101760ae87937beaa66b70d6f30c2e8c102277f9":{"kind":"REWORK","_number":1,"created":"2020-04-16 16:48:03.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/52/720552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/52/720552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/52/720552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/52/720552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/52/720552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/52/720552/1"}}},"commit":{"parents":[{"commit":"727f6435516f36564f37b5acbcf4be39db35d0d7","subject":"Merge \"Remove projects\u0027 TC liaisons\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/727f6435516f36564f37b5acbcf4be39db35d0d7"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2020-04-16 16:37:36.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2020-04-16 16:37:36.000000000","tz":60},"subject":"Add mugsie nomination for TC Chair","message":"Add mugsie nomination for TC Chair\n\nI would like the TC to consider me for Chair.\n\nI have been involved in the TC, both as a member and concerned community\nmember for a long time, and would like to think I have helped drive us forward.\n\nI think the TC Update that Chris Dent used to do was important - I don\u0027t think\nit is useful as a weekly, but as a bi-weekly or longer *regular* update it\nallows the TC to have a chunk of work that people can read about.\n\nI have been dealing with the board for my time on the TC, and I think I can\nbe a representive conduit for the TC and the OpenStack community to push\nissues, ideas, and feedback to the Board and other OpenStack Foundation\nProjects.\n\nThanks for reading this, and your consideration.\n\n- Graham\n\nChange-Id: I7116e848f113a9842785c01530fec018570490dd\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/101760ae87937beaa66b70d6f30c2e8c102277f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/101760ae87937beaa66b70d6f30c2e8c102277f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
