)]}'
{"id":"openstack%2Fgovernance~733901","triplet_id":"openstack%2Fgovernance~master~Ibc25efd3ba201cadfb3677d2c6d7219cb430dadd","project":"openstack/governance","branch":"master","topic":"project-update","hashtags":[],"change_id":"Ibc25efd3ba201cadfb3677d2c6d7219cb430dadd","subject":"Retire swift-specs","status":"MERGED","created":"2020-06-05 17:43:16.000000000","updated":"2020-06-15 19:30:10.000000000","submitted":"2020-06-15 19:23:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733901-1592249029185-035ca625","meta_rev_id":"4fbeea48e7b68ef87733528dc9fc7db8389fe43d","_number":733901,"virtual_id_number":733901,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-15 19:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-15 19:11:12.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-12 23:37:46.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2020-06-15 14:08:02.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 17:44:53.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-15 14:08:02.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-06-15 19:11:12.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-06-15 19:23:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26a7588013ba9b7807d910bff674df68e5bbb98c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-05 17:43:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37a95e2d1fdbaed621c43684a3aa6c5158acf9f5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-05 17:44:53.000000000","message":"Patch Set 1: Code-Review-1\n\nThis is half of the change - and ok.\n\nUpdate legacy.html to add swift-specs there","accounts_in_message":[],"_revision_number":1},{"id":"b8300de29156e2bf3e3f15db492ccd181a153b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 17:55:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36da783cc4ec463dad4cb0b82be89ecc : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b938c906cf6f4d5789a68f3405f9cf11 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4ac0b91d84e491eb1012b7d538bed94 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e07cf714d9b45aabb31d8f54d88cd82 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b481ad673d68449f8bd9e1b2b36cc19a : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/d2d19e393bca46aba7a891ca1cc796d5 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5f65d06c0a924c0b8869446896e0a3b3 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4e5fe6f872cee1727dac3d6a34735e8dd367b9cb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-09 21:48:11.000000000","message":"Patch Set 1:\n\n\u003e Update legacy.html to add swift-specs there\n\nThat feels unnecessary, given the stated purpose that\n\n\u003e We track these temporarily to manage voter rolls.\n\nI\u0027m fairly certain we\u0027ve already met the criteria for removing swift-specs from legacy.yaml, so it seems silly to add it. The last merge (excluding the one to remove all content) was in 2016, and even taking a broader view and looking at all *proposed* patches, there\u0027s only been one in the last year: https://review.opendev.org/#/c/732744/","accounts_in_message":[],"_revision_number":1},{"id":"7cba2365dad02aa94cae34de725e0c6b78cf7576","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-12 23:37:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"15f581e76781d97c449edd22520056c58d80e3ce","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-12 23:37:46.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"1ad92b691f2635d444b7b01114889f91982cb307","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 23:43:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd0ced603bd94817a75170479067cbf5 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e75dc9f9794a42a63ce8e020e4f422 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1026fb28a2a74c5a85f36d05bdfdb5a7 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57454c879fa545399c175838826bf813 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43227f96f0af4fd18afbd697c7b20e23 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/03a4ecd632e648d28584ae0933072941 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/087042fe151c4f7d9f8b2e9976e9831b : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a11e6d2ae7d5cf3d765b551e06e1169a40f85850","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-13 00:00:46.000000000","message":"Topic set to project-update","accounts_in_message":[],"_revision_number":2},{"id":"98ff3efe04a68519c16334348c9929f51fade790","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-06-15 14:08:02.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\nMakes sense to retire it if they aren\u0027t using it.","accounts_in_message":[],"_revision_number":2},{"id":"3a1102b19ceb6145c4c2b0faa61c61ae1d30e7f6","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-15 19:11:12.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"06a5713ae5ecc1814103711357709890dff9e873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:12:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b4b9cc02e308ecd829f0ce2865b47cf33ae93259","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:23:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e3d5d3f20b416394141b64bff0a795 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ba666ec84fb432cb2ade01a1589d022 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a6a30c4755a40989044419693bb0677 : SUCCESS in 7m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e5e5b3d40474b718112ca7365dd14d9 : SUCCESS in 10m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3ebc7957afcb40f881d9333b605609c5 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":2},{"id":"27cce2ba5cd9e2c5a78083a318835433cbccd132","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:23:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4fbeea48e7b68ef87733528dc9fc7db8389fe43d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:30:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/afd9e2c894d0404d833fddf147bf3bb7 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e2007870bca8194d19d3c07ca8edcaf218ba021","revisions":{"3658c5ae4550e726cccd52f896e52185b2bd667d":{"kind":"REWORK","_number":1,"created":"2020-06-05 17:43:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/733901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/01/733901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/01/733901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/01/733901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/01/733901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/01/733901/1"}}},"commit":{"parents":[{"commit":"2a6832fae60f154497510bdb31d5a32a96bf3deb","subject":"Merge \"Move i18n into a SIG\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/2a6832fae60f154497510bdb31d5a32a96bf3deb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-05 17:42:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-05 17:42:34.000000000","tz":-420},"subject":"Retire swift-specs","message":"Retire swift-specs\n\nThe Swift team hasn\u0027t used swift-specs for about four years; see\nhttp://lists.openstack.org/pipermail/openstack-dev/2016-May/094026.html\n\nChange-Id: Ibc25efd3ba201cadfb3677d2c6d7219cb430dadd\nRelated-Change: https://review.opendev.org/#/c/732999/\nRelated-Change: https://review.opendev.org/#/c/733001/\nRelated-Change: https://review.opendev.org/#/c/733003/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/3658c5ae4550e726cccd52f896e52185b2bd667d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/3658c5ae4550e726cccd52f896e52185b2bd667d"}]},"branch":"refs/heads/master"},"1e2007870bca8194d19d3c07ca8edcaf218ba021":{"kind":"REWORK","_number":2,"created":"2020-06-12 23:37:13.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/01/733901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/01/733901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/01/733901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/01/733901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/01/733901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/01/733901/2"}}},"commit":{"parents":[{"commit":"2a6832fae60f154497510bdb31d5a32a96bf3deb","subject":"Merge \"Move i18n into a SIG\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/2a6832fae60f154497510bdb31d5a32a96bf3deb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-05 17:42:18.000000000","tz":-420},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-06-12 23:37:09.000000000","tz":-240},"subject":"Retire swift-specs","message":"Retire swift-specs\n\nThe Swift team hasn\u0027t used swift-specs for about four years; see\nhttp://lists.openstack.org/pipermail/openstack-dev/2016-May/094026.html\n\nChange-Id: Ibc25efd3ba201cadfb3677d2c6d7219cb430dadd\nRelated-Change: https://review.opendev.org/#/c/732999/\nRelated-Change: https://review.opendev.org/#/c/733001/\nRelated-Change: https://review.opendev.org/#/c/733003/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/1e2007870bca8194d19d3c07ca8edcaf218ba021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/1e2007870bca8194d19d3c07ca8edcaf218ba021"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
