)]}'
{"id":"openstack%2Fgovernance~736737","triplet_id":"openstack%2Fgovernance~master~I74583fb696356a00cef898647b135c2476936305","project":"openstack/governance","branch":"master","topic":"code-change","hashtags":[],"change_id":"I74583fb696356a00cef898647b135c2476936305","subject":"tools: update check-review-status to drop grace period","status":"MERGED","created":"2020-06-18 14:46:26.000000000","updated":"2020-06-22 12:36:35.000000000","submitted":"2020-06-22 12:33:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736737-1592829204030-7f2da1d8","meta_rev_id":"e8ce0144595820a39f0f5c98ef6e7caae15baac9","_number":736737,"virtual_id_number":736737,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-22 12:33:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-19 18:00:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2020-06-19 19:47:56.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2020-06-19 19:47:56.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2020-06-22 07:59:07.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-19 18:00:56.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 18:00:56.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-06-22 07:59:07.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-06-22 12:33:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"960a483aee007f790f7ac71a3fc094817a1a6ebf","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-18 14:46:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48d93834b9259b349806f3805bac5e492d160917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 15:02:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7267261dcf0b47b6b4ff4c3e4eaa2e3b : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69f45b6fa49a4c239a7f932e03402c71 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98efc0899fa94ae589c527bb0988353e : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/015e03905d9f4f5eb41c775397fc8f01 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c66aed34db02451ebad19beee27523a9 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/64a223b910504e09b4f1137d326c7eb4 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c9c57599738d47e780b30c5999869cda : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0d091376cb52c00545349b4c2ef913b916d82dbc","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-06-19 18:00:56.000000000","message":"Patch Set 1: Rollcall-Vote+1 Code-Review+1\n\nThanks Mohammed.","accounts_in_message":[],"_revision_number":1},{"id":"8a2f4a3e7792ae96fcb3c3bbaab34a050dc1b044","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-19 19:47:56.000000000","message":"Patch Set 1: Rollcall-Vote+1 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f20cbfe931f4a8832f308bd86cf634826387a260","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-06-22 07:59:07.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"e35d8a4311c1f73ed69b2223832d84d2ff6b046a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 12:25:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c524ed20477130a9d82d488d1660d03d2e60cf41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 12:33:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ed31fbd6874530bd5268333078ae79 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f9eec6d9554f52930abc11275de26c : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d6be05ae5e0480d901379317a063ce9 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51af1f83a1714e1c86b595821fdb75e2 : SUCCESS in 6m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/128e2359a71c4b40b67a901d10ff19bc : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f251d9bb7f6998c4a76787e91d4faa3972bcde93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 12:33:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e282f0d5ed734a112c43974323563b62d95750e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 12:36:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/4999a7ea16a2413186bc7c7c423233eb : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79fd6a86f35f1a88349dd48cd93d027a2856f8c9","revisions":{"79fd6a86f35f1a88349dd48cd93d027a2856f8c9":{"kind":"REWORK","_number":1,"created":"2020-06-18 14:46:26.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/37/736737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/37/736737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/37/736737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/37/736737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/37/736737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/37/736737/1"}}},"commit":{"parents":[{"commit":"c0aab03233ca0e7dcc9da2d30e73fe2b6fa5b684","subject":"Add openstack-ansible-os_adjutant repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c0aab03233ca0e7dcc9da2d30e73fe2b6fa5b684"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-06-18 14:44:57.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-06-18 14:44:57.000000000","tz":-240},"subject":"tools: update check-review-status to drop grace period","message":"tools: update check-review-status to drop grace period\n\nWe\u0027ve decided that we\u0027re removing grace period on resolutions and this\npatch updates the tooling to reflect that.\n\nDepends-On: https://review.opendev.org/735361\nChange-Id: I74583fb696356a00cef898647b135c2476936305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/79fd6a86f35f1a88349dd48cd93d027a2856f8c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/79fd6a86f35f1a88349dd48cd93d027a2856f8c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
