)]}'
{"id":"openstack%2Fgovernance~770285","triplet_id":"openstack%2Fgovernance~master~I6656e85dc07282e545ca5fdd51f23a872f916c8d","project":"openstack/governance","branch":"master","hashtags":[],"change_id":"I6656e85dc07282e545ca5fdd51f23a872f916c8d","subject":"WIP NO MERGE Move os-*-config to Heat project governance","status":"ABANDONED","created":"2021-01-12 12:36:19.000000000","updated":"2021-01-26 16:55:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c1446034de3732bef40d082f85c6ca0ad15155ec","_number":770285,"virtual_id_number":770285,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-12 12:48:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":-1,"date":"2021-01-12 15:35:33.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 12:48:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98e8252681a0d3a2957a5103a16c6e8205fffd4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-12 12:36:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"447e829132771452aafc81ee9568e4e8663cb67e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 12:48:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fd7e267bf1642b2b428a22267eac69d : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6069ad9ed5084e929867caedafba22bd : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8684b9d3ceb4e5fa486c5b0c7b176b7 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e3ab477828a4364a21acafd9847f659 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86905f8850f748eaa184e15e68bd616e : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/3b8b22be0cd846e381299babf86708f2 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/133c61ab23224f00bec04a29a1f1d8a6 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"400b5745a84f34f346ec836e5e894cb8f72144f8","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-12 15:35:33.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c1446034de3732bef40d082f85c6ca0ad15155ec","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-26 16:55:54.000000000","message":"Abandoned\n\ninstead we are moving them independent  https://review.opendev.org/c/openstack/releases/+/772570 http://lists.openstack.org/pipermail/openstack-discuss/2021-January/020030.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20e91643311efa49285e68f220f5a5040a6b81e7","revisions":{"20e91643311efa49285e68f220f5a5040a6b81e7":{"kind":"REWORK","_number":1,"created":"2021-01-12 12:36:19.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/85/770285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/85/770285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/85/770285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/85/770285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/85/770285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/85/770285/1"}}},"commit":{"parents":[{"commit":"3aaa73e3c36276ae47e213cc8e136d0a64d9dcd5","subject":"Merge \"Improve check-review-status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/3aaa73e3c36276ae47e213cc8e136d0a64d9dcd5"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-01-12 12:12:30.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-01-12 12:17:22.000000000","tz":120},"subject":"WIP NO MERGE Move os-*-config to Heat project governance","message":"WIP NO MERGE Move os-*-config to Heat project governance\n\nAs discussed at [1] the TripleO project is no longer using the\nos-*-config projects for a few cycles now. The TripleO team is\nproposing to move these to independent as it doesn\u0027t make sense to\ncreate tagged releases for these projects at least for TripleO.\nUltimately we may even move these to EOL.\n\nAs pointed out in [1] Heat may still be using these? In which case\nit makes sense for Heat to own and manage the release of these\nprojects\n\nSetting WIP and posting for the discussion.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019751.html\n\nChange-Id: I6656e85dc07282e545ca5fdd51f23a872f916c8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/20e91643311efa49285e68f220f5a5040a6b81e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/20e91643311efa49285e68f220f5a5040a6b81e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
