)]}'
{"id":"openstack%2Fgovernance~772597","triplet_id":"openstack%2Fgovernance~master~I8f968adcc827cc7171c8310a585d3c79aba12940","project":"openstack/governance","branch":"master","topic":"project-update","hashtags":[],"change_id":"I8f968adcc827cc7171c8310a585d3c79aba12940","subject":"Add rbd-iscsi-client to cinder project","status":"MERGED","created":"2021-01-26 20:36:53.000000000","updated":"2021-02-04 16:02:16.000000000","submitted":"2021-02-04 15:59:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"772597-project-update","meta_rev_id":"38be57bf646d9b04b64db6fd178ba148dc33c44c","_number":772597,"virtual_id_number":772597,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-04 15:59:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2021-02-04 15:47:01.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2021-01-27 01:14:44.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2021-01-28 15:40:41.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 20:45:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-27 01:14:44.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2021-01-28 15:40:41.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-02-03 18:09:03.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"}],"messages":[{"id":"0a96f46b2b68a8b9b7d6639e38b001631c561cff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-26 20:36:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"092e836d3d82c31f1bf6369b514f2b794fe5b04e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:45:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c475d2c0148a4d2e98cd0e608aebbc97 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c41e909d4844a1bfe8de8b1738fc51 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e593a27573a3483b9122d559e1ac23dc : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bfdbcbaa5394414b1767c2a1894cc71 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/119fb4a51c61402ab15cf417a8c862f8 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/ea9bd5689dda4e7a94c589b55834df2e : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e893860d4a844044a4b019bf49ddfa0c : FAILURE in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"48d01aef2cd007067bfb97dc43a068955516889f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-01-27 01:05:50.000000000","message":"Topic rbd-iscsi-client removed","accounts_in_message":[],"_revision_number":1},{"id":"48333057a33a2d3af37b804404c984eb718de1ab","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-01-27 01:06:20.000000000","message":"Topic set to project-update","accounts_in_message":[],"_revision_number":1},{"id":"8686390c21c6346ea01a862acd317ea26cd64ff0","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-01-27 01:14:44.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"7aa0b57aa160b65b5c3f088b3d571cd5c8fe92fd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-28 15:40:41.000000000","message":"Patch Set 1: Rollcall-Vote+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba18426a2c11938987a2df01be3e7db114f9d4f7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-29 16:21:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"98753e61b3bc830f4e826441c701a328ff29a01d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-29 16:26:36.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6516b80016ff264fb2b2210a9d71a59ee36a1af2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-01 22:54:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2849e2888bd26b76a31b06f83e75a476425fb8a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 23:05:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59dabb222bff469e965572460bc758ee : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f43c7ae7f848d19a9b388be66c96fe : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c667ca46c114f9dad1d191611189daa : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3046603b4db404cbeed8685ccf6c13b : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61578b016b8e41b89b6e5711dc16c552 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/1bc634aa0acd4bbab9ad0913f138fe91 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b6be7bc9c7994ac98918e672ca51c34b : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"40ca896c814a2792b98e0fc5f8a0e74f093f101e","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-02-04 15:47:01.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2ead479773d1002c0cc8b6d0d9a9f4b4e8104db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 15:47:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23fb1c43c5e7a2351329bec0aa1e030229f8b42c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 15:59:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6555b5c54a144a4ad972d238df85daf : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dad045c223445c1956e5f8b9f27b58a : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd5bc006d337479c94cad5bb8a5f2b8c : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1fbfb8524834fb39b47dade0a5facdb : SUCCESS in 3m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bbfd8295a64f4c77944b8e324c187a2f : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a64d08c1a47cef5dec345d9df4d3917745c9d158","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 15:59:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"38be57bf646d9b04b64db6fd178ba148dc33c44c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 16:02:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/a0d55ccc40d5478e87de3aa62cd0dcf0 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a466d5856b5ab34aeaac1385d24812dda79dde5","revisions":{"3a466d5856b5ab34aeaac1385d24812dda79dde5":{"kind":"REWORK","_number":1,"created":"2021-01-26 20:36:53.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/97/772597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/97/772597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/97/772597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/97/772597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/97/772597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/97/772597/1"}}},"commit":{"parents":[{"commit":"2bdd9cff00fb40b2f95b66cad47ae1cfd14a2f1b","subject":"Merge \"Add link to Xena announcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/2bdd9cff00fb40b2f95b66cad47ae1cfd14a2f1b"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-01-26 20:26:00.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-01-26 20:33:10.000000000","tz":-300},"subject":"Add rbd-iscsi-client to cinder project","message":"Add rbd-iscsi-client to cinder project\n\nAdd the rbd-iscsi-client repo to the cinder project.  It will\nfollow the independent release model.\n\nrbd-iscsi-client was added to global-requirements by commit\n2ce1a5eaab.  The cinder project is assuming maintenance of the\nlibrary.\n\nDepends-on: Icac3f5f74b2fca06c8b25fa78d895bbb710ea308\n\nChange-Id: I8f968adcc827cc7171c8310a585d3c79aba12940\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/3a466d5856b5ab34aeaac1385d24812dda79dde5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/3a466d5856b5ab34aeaac1385d24812dda79dde5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}},{"label":"Rollcall-Vote","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
