)]}'
{"id":"openstack%2Fgovernance~773383","triplet_id":"openstack%2Fgovernance~master~I5b2d06a428738b29a1e2d583950a5e33a2ce4cdb","project":"openstack/governance","branch":"master","topic":"project-update","hashtags":[],"change_id":"I5b2d06a428738b29a1e2d583950a5e33a2ce4cdb","subject":"Create ansible-role-pki repo","status":"MERGED","created":"2021-02-01 12:56:24.000000000","updated":"2021-02-04 16:06:04.000000000","submitted":"2021-02-04 16:06:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773383-project-update","meta_rev_id":"048cf2d637c76269b57bdfcf87b9447dc23659ee","_number":773383,"virtual_id_number":773383,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-04 16:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2021-02-04 15:52:20.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"all":[{"value":1,"date":"2021-02-04 15:52:42.000000000","_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2021-02-04 15:50:37.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2021-02-04 15:51:58.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-04 15:52:15.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-01 13:09:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-04 15:50:37.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2021-02-04 15:51:58.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-02-04 15:52:15.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-02-04 15:52:42.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"}],"messages":[{"id":"9b1d90e763f9a326b5250c3d2aaafb1f2c6acfb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-01 12:56:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9384ccc4ef505b7413e0bf4cf7759c100cd6775","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-01 12:59:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5bd1a85a85f12bb83c67d58b66936de6d1c1bfe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 13:09:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/061956b7f597432d8e163a0b56b7d979 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2b319d3d6b3467bb9286faf21ed3a09 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a7d0e8974f54581821ced4f467b9798 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0be4c7b1e8bf41c294c9d36bd65fb168 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e021d18b93648d698eefd0872193496 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/6befef733c354d97b2a1d9940508ad75 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/12f44e5bca604341adae0beb6060b1b9 : FAILURE in 3m 30s","accounts_in_message":[],"_revision_number":2},{"id":"1131a301bdad6d07334971ff73c66200d67af1c1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-04 09:16:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"21ca0b9e2ad054e426cf7901bb01a02b551e39ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 09:30:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee341e43a05d44d186b2c3f8f2eda9e0 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb55c51b703c46e5ad6b8ed0ebec904e : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56fd70237b4946f79e7fae9d6aa576c7 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13c080a9e63d4941ac36de1de85d56af : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe2bab421c894470bfc9dce95cd4ea25 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/bd4026ffee374f85aaf4167bf548175d : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/05b7e7a23f294338a7dc3722219e5a67 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"ca6f218fdf0e0bdd66e60f4cb48f5262444d3883","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-02-04 15:50:37.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"d7ed3e1583d9456aa8608c0b210b0c3d33656e0f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-02-04 15:50:39.000000000","message":"Topic osa/pki removed","accounts_in_message":[],"_revision_number":2},{"id":"c65cf02039e122e9ee7add3c1dda090e3a228ce6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-02-04 15:50:43.000000000","message":"Topic set to project-update","accounts_in_message":[],"_revision_number":2},{"id":"553e89900d70beb6e76b53db57e3941351dcb5b8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-04 15:51:58.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"db9bf533ec6620089a53ccd110a8e97301149948","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-02-04 15:52:15.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\nMakes sense.","accounts_in_message":[],"_revision_number":2},{"id":"c47598c572d1c85ecc5d4bca084b93c5bcb52e15","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-02-04 15:52:20.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"84de36becc8d4fd4317d5519da2596b471dba426","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-02-04 15:52:42.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"099be2511133871fbefbb7dc1271660902029168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 15:55:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0f91e2f63ed378c34a1e08e8cf39f0ab4e8f34ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 16:05:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9e4ef6f92234fca8e854021aafa631c : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65f4ac88237749ea97a7cc5129f7fe9a : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/700d1cb28043437283051b897eeb46f4 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8176cdc5c4c3465b94106f062360c6d7 : SUCCESS in 3m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1ce2cfe3a0df4ba5aecc47cba0167af2 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"048cf2d637c76269b57bdfcf87b9447dc23659ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 16:06:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"38b78643fe9ade15b8e6ff155b1ee64b93b9fbc5","revisions":{"2c65e0d000a2679a9440d851e8cea6cc411926ae":{"kind":"REWORK","_number":1,"created":"2021-02-01 12:56:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/773383/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/83/773383/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/83/773383/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/83/773383/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/83/773383/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/83/773383/1"}}},"commit":{"parents":[{"commit":"d6a1ad9e24065487ec90bb8d8dae0d1d04aebe0e","subject":"Merge \"Setting Ke Chen as Watcher\u0027s PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/d6a1ad9e24065487ec90bb8d8dae0d1d04aebe0e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-01 12:55:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-01 12:55:56.000000000","tz":120},"subject":"Create ansible-role-pki repo","message":"Create ansible-role-pki repo\n\nThis role aims to manage SSL certificate issue process -\nie generation of self-signed certificates or distribution of user\nprovided ones\n\nChange-Id: I5b2d06a428738b29a1e2d583950a5e33a2ce4cdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/2c65e0d000a2679a9440d851e8cea6cc411926ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/2c65e0d000a2679a9440d851e8cea6cc411926ae"}]},"branch":"refs/heads/master"},"38b78643fe9ade15b8e6ff155b1ee64b93b9fbc5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-01 12:59:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/773383/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/83/773383/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/83/773383/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/83/773383/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/83/773383/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/83/773383/2"}}},"commit":{"parents":[{"commit":"d6a1ad9e24065487ec90bb8d8dae0d1d04aebe0e","subject":"Merge \"Setting Ke Chen as Watcher\u0027s PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/d6a1ad9e24065487ec90bb8d8dae0d1d04aebe0e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-01 12:55:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-01 12:58:50.000000000","tz":120},"subject":"Create ansible-role-pki repo","message":"Create ansible-role-pki repo\n\nThis role aims to manage SSL certificate issue process -\nie generation of self-signed certificates or distribution of user\nprovided ones\n\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/773385\nChange-Id: I5b2d06a428738b29a1e2d583950a5e33a2ce4cdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/38b78643fe9ade15b8e6ff155b1ee64b93b9fbc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/38b78643fe9ade15b8e6ff155b1ee64b93b9fbc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}},{"label":"Rollcall-Vote","status":"MAY","applied_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
