)]}'
{"id":"openstack%2Fgovernance~855276","triplet_id":"openstack%2Fgovernance~master~I536e55e592967795f4435a5d316083481d5bbcfe","project":"openstack/governance","branch":"master","topic":"formal-vote","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-10 14:59:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I536e55e592967795f4435a5d316083481d5bbcfe","subject":"Switch Release Management to DPL","status":"MERGED","created":"2022-08-30 19:36:10.000000000","updated":"2022-09-10 15:01:18.000000000","submitted":"2022-09-10 14:59:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"855276-formal-vote","meta_rev_id":"393d3a64da23106c2820a80d20b280cb6e96caa8","_number":855276,"virtual_id_number":855276,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-10 14:59:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-08-31 14:00:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-09-01 15:58:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2022-08-31 12:19:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2022-09-10 14:49:44.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2022-09-08 15:38:31.000000000","_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2022-09-01 18:25:43.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2022-09-01 15:56:20.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-09-01 15:58:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2022-08-31 12:58:59.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-01 16:04:25.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-30 19:39:20.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-08-30 19:52:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-31 10:20:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-08-31 12:19:48.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-08-31 14:00:26.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-09-01 15:56:20.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-09-01 15:58:32.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2022-09-01 16:04:25.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-09-01 18:25:43.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-09-08 15:38:31.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"692655cea8ee059b7f4c8acf84757d70fac16486","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-30 19:36:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e94fdddfd3008cd93eeec384fea7d5170466a5a0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-08-30 19:39:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"27be19bd9bd3872b19baf1c2ac535cdf56c8e7cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 19:52:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f2418e540a4460e85890b95be0bf57e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c242bdedbc514d3f8e44167bed0c0be4 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/754e58d59c5240a6922f1af0162ca23a : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a10ff0f055443f2a520431c0da1f7d3 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b48ca5d80d04bcaa60e63a8020518b8 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92d829f8cfb84023994b93013769a752 : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/3b2e5d6197af4981bc863776b79f48ee : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3a2784bf027e47baac7ad211057a2288 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9e16090edee752a7e62c9f2714e81d89a0533538","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-31 09:13:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd7fc2da5c046f1b898184fcca25c84ca3d285de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 09:22:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ac054e94d70494ba02f8c6b82434b5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6e18e0417fa403fa9a06df8ecbcdd43 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5998d668ca054ffdb09aa1cee0ba51e3 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/384f02d23e0e406ca7256bce1173c1e7 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df23573440d54665a2d641012698cb5a : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bb78124196e4a028f75e6a8b6130d7b : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/92ffec4c98a343768f1d9b1378e73044 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e4e44c31c5684881943adebc1e007440 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"5a372bbe373fd0134b26d3c5856cfb600f8b94f0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-08-31 10:20:37.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d1df2059d50c42fd5c6e1e1e6b1df6ae97ca4ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-31 12:05:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d1d314b1c68fddd9c34564ebc5c52d3569076d23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 12:12:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eab519e6233d4b4c9e1278dcebe40f96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54c27a8b6dfa43ef98339e0387fe0403 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3fbd75a808f4866a9fa45b4600333a8 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f730d4458e8459892b91a47d8e48da7 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2ac090fd8094f5c9428fa39f4e8ddc3 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29ea80c00bf346d38cf8409221fe996f : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/01f980b6695842f083a0df255ff08ad0 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fe98df1d83c0486f95bc336843e8ac3e : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":3},{"id":"8d594be53d5d4bf4d3e3ad25eedaac50b64d281b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-31 12:19:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"95540266b18d405a3ea838af62725d1f4875794c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-08-31 12:58:51.000000000","message":"Topic set to formal-vote","accounts_in_message":[],"_revision_number":3},{"id":"60430d5e7702293e9dc3edb8330bb818d09a81e3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-08-31 12:58:59.000000000","message":"Patch Set 3: Rollcall-Vote+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d62807685a095f6a3776d28e56d884ce0cac9f80","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-08-31 14:00:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"20a6f99ca1b5b08b75ce4097c429547b3f0ddabc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-01 15:56:20.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"fe262befd8a94bf17813bd382713ff7958c0f78f","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2022-09-01 15:58:32.000000000","message":"Patch Set 3: Rollcall-Vote+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"dec0fc5722ffde97c5da37deb59739b95dd19fae","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-09-01 16:04:25.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"8bb499beb6f26e47407dcc312966147443980e0b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-09-01 18:25:43.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"a78eaca6716ce766230600276feb7386d4fda47e","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-09-08 15:38:31.000000000","message":"Patch Set 3: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"a9bb631eb5da4f7bf6a83a14cbbc30c4e307e4f9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-09-10 14:49:44.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"59adaba5dd875c31b46581442e82bad17a64a9e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-10 14:50:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"37051c76c5b26b74e40f2350345e15827dbf5185","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-10 14:59:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02aa69bcf5554010bb9a2a126e9a20f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c14b82e009a429cb5dc3532a6a9c755 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caf108f317aa4a658017055435953e0e : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcc9f44353e840159440091d1c42195a : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c21bad388b84f06bf2bb30903b4d867 : SUCCESS in 4m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ce32f04f8d3e4f46a6b9cfa1df29bb86 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":3},{"id":"a9df4912a9fa9a92bc0b8c5e58f3988bbb9fe3f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-10 14:59:54.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nThe change was submitted with unreviewed changes in the following files:\n\n```\nThe name of the file: reference/projects.yaml\nInsertions: 8, Deletions: 0.\n\n@@ -2417,6 +2417,14 @@\n       - name: Elod Illes\n         irc: elodilles\n         email: elod.illes@est.tech\n+    tact-sig:\n+      - name: Jeremy Stanley\n+        irc: fungi\n+        email: fungi@yuggoth.org\n+    security:\n+      - name: Thierry Carrez\n+        irc: ttx\n+        email: thierry@openstack.org\n requirements:\n   ptl:\n     name: Matthew Thode\n```\n","accounts_in_message":[],"_revision_number":3},{"id":"393d3a64da23106c2820a80d20b280cb6e96caa8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-10 15:01:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17e2998b619948df898858767e4fd61d\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/d7bde28ac8274e6e863e703d486110f2 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a181a8113d168480f6162d09993a0060a7d804ef","revisions":{"6f542b9d05a2ef6eacc835dda88f750f4c72ac8f":{"kind":"REWORK","_number":1,"created":"2022-08-30 19:36:10.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/76/855276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/76/855276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/76/855276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/76/855276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/76/855276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/76/855276/1"}}},"commit":{"parents":[{"commit":"57e0b3f09ffe3489cccbc2a8462b9b8d28f7821a","subject":"Merge \"Retire openstack-helm-addons\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/57e0b3f09ffe3489cccbc2a8462b9b8d28f7821a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-08-30 19:34:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-08-30 19:34:25.000000000","tz":120},"subject":"Switch Release Management to DPL","message":"Switch Release Management to DPL\n\nThis patch switches Release Management to distributed leadership model.\n\nChange-Id: I536e55e592967795f4435a5d316083481d5bbcfe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/6f542b9d05a2ef6eacc835dda88f750f4c72ac8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/6f542b9d05a2ef6eacc835dda88f750f4c72ac8f"}]},"branch":"refs/heads/master"},"f51acbf085ce18f849a23a860cab6c34116b6f6e":{"kind":"REWORK","_number":2,"created":"2022-08-31 09:13:08.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/76/855276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/76/855276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/76/855276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/76/855276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/76/855276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/76/855276/2"}}},"commit":{"parents":[{"commit":"57e0b3f09ffe3489cccbc2a8462b9b8d28f7821a","subject":"Merge \"Retire openstack-helm-addons\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/57e0b3f09ffe3489cccbc2a8462b9b8d28f7821a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-08-30 19:34:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-08-31 09:12:53.000000000","tz":120},"subject":"Switch Release Management to DPL","message":"Switch Release Management to DPL\n\nThis patch switches Release Management to distributed leadership model.\n\nChange-Id: I536e55e592967795f4435a5d316083481d5bbcfe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f51acbf085ce18f849a23a860cab6c34116b6f6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f51acbf085ce18f849a23a860cab6c34116b6f6e"}]},"branch":"refs/heads/master"},"a181a8113d168480f6162d09993a0060a7d804ef":{"kind":"REWORK","_number":3,"created":"2022-08-31 12:05:33.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/76/855276/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/76/855276/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/76/855276/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/76/855276/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/76/855276/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/76/855276/3"}}},"commit":{"parents":[{"commit":"57e0b3f09ffe3489cccbc2a8462b9b8d28f7821a","subject":"Merge \"Retire openstack-helm-addons\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/57e0b3f09ffe3489cccbc2a8462b9b8d28f7821a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-08-30 19:34:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-08-31 12:05:10.000000000","tz":120},"subject":"Switch Release Management to DPL","message":"Switch Release Management to DPL\n\nThis patch switches Release Management to distributed leadership model.\n\nChange-Id: I536e55e592967795f4435a5d316083481d5bbcfe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/a181a8113d168480f6162d09993a0060a7d804ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/a181a8113d168480f6162d09993a0060a7d804ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Rollcall-Vote","status":"MAY","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[]}
