)]}'
{"id":"openstack%2Fgovernance~862637","triplet_id":"openstack%2Fgovernance~master~I584cd5227723759da5325d8015b4fa2f29912cdc","project":"openstack/governance","branch":"master","topic":"documentation-change","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2022-11-16 19:44:23.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-11-15 13:18:09.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-11-16 19:37:38.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I584cd5227723759da5325d8015b4fa2f29912cdc","subject":"Update CHAIR.rst for few updates","status":"MERGED","created":"2022-10-25 19:56:56.000000000","updated":"2022-11-16 19:45:29.000000000","submitted":"2022-11-16 19:44:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"862637-documentation-change","meta_rev_id":"ba07b4f954f7a8bcfcdb4497ab7313e0ff06dc90","_number":862637,"virtual_id_number":862637,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-16 19:44:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-16 19:37:38.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-11-16 18:16:10.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-11-16 06:57:10.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2022-11-16 19:37:38.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-25 20:06:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-27 08:34:44.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-10-27 15:43:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2022-10-27 15:45:01.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2022-11-15 13:18:09.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-11-16 18:16:10.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"4c7178625b4c0ca03741166c2e3b6eca62aebf4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-25 19:56:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a01d6bb793e185b14c4e07fd34ed6ce73816a4a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 20:06:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f30be8b9a35f4e8a96393107ee63ba51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4bc73acff44e889e04dba5bcd25f95 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8a30bdaca1b4519825b977cf6d49f2c : SUCCESS in 2m 41s\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/5379a78bf9ef4816ba43b738e3d5ea19 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f4c1216a797d4b15a8bb7634bdf9e45d : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1ee46adc47e22f01c57d9001795092fbc61e4119","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-27 08:34:44.000000000","message":"Patch Set 1: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"8bdac920288ea1ef97d7fc75b4cdf14179110747","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-10-27 15:43:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03e7a587f3395617557a6f09ac8acba509015765","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-10-27 15:45:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"245fd84d521f997e2cf342f912794ce360e061b8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-27 15:47:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c526b516c9fd9509ede42b92bf251e3edefe056f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-27 15:57:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"15f8763949cfc76d118c0426b0f9ab7f7733f1dd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-27 15:58:30.000000000","message":"Topic set to documentation-change","accounts_in_message":[],"_revision_number":2},{"id":"0d5e4cc5016ab0f292484090b52f399a1f3cf478","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 16:10:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d01dfd191684425af1e6666c219b8b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f51074bbdd5d4cf9a069fb9d6fff39f6 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7227f2afe56c499e9a6539884ee39cac : SUCCESS in 3m 02s\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/0a166506dc984a65a820d63473f3f8bd : SUCCESS in 2m 30s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7378623547434333a1652917a2cb5543 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":2},{"id":"88ea311f72cc7edd3d4d95d8bc36351b428d3675","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-11-15 13:18:09.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"8a37632a4611dc1725930bbff9ab45852daf9f00","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-16 06:55:30.000000000","message":"Patch Set 2: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":2},{"id":"7a35a83e66e8d450aeacba3c9b3811e71b672419","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-16 06:56:55.000000000","message":"Patch Set 2: -Rollcall-Vote\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f192c756271a7592b97c42cf1be6ddf0868de7b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-16 06:57:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2a3d16c542d465b2a9ea51e1955365cd2871dacc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 07:02:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53fe3404b38347f18618bc7a204080a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb9ec2575ff40498f0138805fa26492 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc5d83106154fd093889497ee2ab8c3 : SUCCESS in 2m 39s\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/52c9a94e8f8d4caaa9cbba415f4e8b6e : SUCCESS in 2m 55s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a8c8734574554eb28e1dc250c3a64485 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"b439154072294c20be23204b4f2082cfd8e4871f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-16 18:16:10.000000000","message":"Patch Set 3: Rollcall-Vote+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"44084f7c6187b24a1da5a4da625ecd03010d0084","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-16 19:37:38.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ebc74b5d95fec3324ca7eb344bc7cff806c9fb7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:37:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4382ca179492a48d15fe24222ed74c283cd226ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:44:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a4ce0bf9d294f3a86ea8a8e21cc5d5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19e5b9956df2430fb710c23030106065 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69a3285eb8a64446b17b347f653e2531 : SUCCESS in 2m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7f3c81a9106b4119b3ed7bfa3495b9cb : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":3},{"id":"d1ca1b492ec67793464c01514931a5bbd22d5aef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:44:23.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nThe change was submitted with unreviewed changes in the following files:\n\n```\nThe name of the file: CHAIR.rst\nInsertions: 3, Deletions: 2.\n\n@@ -44,8 +44,9 @@\n   that 2 TC members are signed up to manage the goal selection process.\n * Ensure PTI and testing runtime update is done.\n * Election planning\n-  * During PTG or at the start of cycle, select a TC member who can be a\n-    election liaison and point of contact for the next election.\n+  * During PTG or at the start of cycle, select a TC member who is not standing\n+    for [re]election to serve as election liaison and point of contact for\n+    the next election.\n   * Encourage the TC members that are not standing for [re]election to help\n     as the election officials.\n   * During R-9 week, check \"PTL and TC elections\".\n@@ -78,11 +79,8 @@\n After the election results are confirmed, the outgoing chair should:\n \n * start the chair selection process by encouraging TC members to\n-  consider serving as chair by proposing a patch to the member list to\n-  add \"chair\" status next to their name\n-* monitor and coordinate the discussion and selection of the new\n-  chair, including establishing a CIVS poll if there are multiple\n-  candidates\n+  consider serving as chair and as per the process defined in\n+  https://governance.openstack.org/tc/reference/tc-chair-elections.html\n * after a chair is selected, ensure they appear in the\n   tech-committee-chair group in gerrit:\n   https://review.opendev.org/#/admin/groups/206,members\n```\n","accounts_in_message":[],"_revision_number":3},{"id":"ba07b4f954f7a8bcfcdb4497ab7313e0ff06dc90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:45:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea538e5c78b642e886145e7bc494e3d9\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/3c8c4fc2fae9483ab086ddef735686d7 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f43a25466a5cddda122e6ea6a786ddb8694fbeaf","revisions":{"cc169743d634b92770ef20edca894393d33ceb74":{"kind":"REWORK","_number":1,"created":"2022-10-25 19:56:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/37/862637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/37/862637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/37/862637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/37/862637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/37/862637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/37/862637/1"}}},"commit":{"parents":[{"commit":"d0f91370b2dc9b5dcefe89f1e34433fb0baee20a","subject":"Merge \"Appoint hongbin as Zun PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/d0f91370b2dc9b5dcefe89f1e34433fb0baee20a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-10-25 19:55:40.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-10-25 19:55:40.000000000","tz":-300},"subject":"Update CHAIR.rst for few updates","message":"Update CHAIR.rst for few updates\n\nIt removes the release naming task which is not handled\nby TC members anymore. Also add the task to select the\nelection liaison at the start of cycle.\n\nChange-Id: I584cd5227723759da5325d8015b4fa2f29912cdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/cc169743d634b92770ef20edca894393d33ceb74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/cc169743d634b92770ef20edca894393d33ceb74"}]},"branch":"refs/heads/master"},"c9d86c4cd9246cf9ebb476d8d221c10b0cbbb08e":{"kind":"REWORK","_number":2,"created":"2022-10-27 15:57:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/37/862637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/37/862637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/37/862637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/37/862637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/37/862637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/37/862637/2"}}},"commit":{"parents":[{"commit":"d0f91370b2dc9b5dcefe89f1e34433fb0baee20a","subject":"Merge \"Appoint hongbin as Zun PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/d0f91370b2dc9b5dcefe89f1e34433fb0baee20a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-10-25 19:55:40.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-10-27 15:57:21.000000000","tz":-300},"subject":"Update CHAIR.rst for few updates","message":"Update CHAIR.rst for few updates\n\nIt removes the release naming task which is not handled\nby TC members anymore. Also add the task to select the\nelection liaison at the start of cycle.\n\nChange-Id: I584cd5227723759da5325d8015b4fa2f29912cdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c9d86c4cd9246cf9ebb476d8d221c10b0cbbb08e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c9d86c4cd9246cf9ebb476d8d221c10b0cbbb08e"}]},"branch":"refs/heads/master"},"f43a25466a5cddda122e6ea6a786ddb8694fbeaf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-11-16 06:57:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/37/862637/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/37/862637/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/37/862637/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/37/862637/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/37/862637/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/37/862637/3"}}},"commit":{"parents":[{"commit":"280637c1fe2a10594a014efbc2cb9c43c95d100a","subject":"Merge \"Adding TC chair election process: nomination in governance repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/280637c1fe2a10594a014efbc2cb9c43c95d100a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-10-25 19:55:40.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-11-16 06:57:10.000000000","tz":0},"subject":"Update CHAIR.rst for few updates","message":"Update CHAIR.rst for few updates\n\nIt removes the release naming task which is not handled\nby TC members anymore. Also add the task to select the\nelection liaison at the start of cycle.\n\nChange-Id: I584cd5227723759da5325d8015b4fa2f29912cdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f43a25466a5cddda122e6ea6a786ddb8694fbeaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f43a25466a5cddda122e6ea6a786ddb8694fbeaf"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Rollcall-Vote","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[]}
