)]}'
{"id":"openstack%2Fgovernance~869752","triplet_id":"openstack%2Fgovernance~master~Ia4dedab2d4d5d7f4090be71115e79aa155906df2","project":"openstack/governance","branch":"master","topic":"project-update","attention_set":{},"removed_from_attention_set":{"33741":{"account":{"_account_id":33741,"name":"Gustavo Sanchez","email":"gustavo.sanchez@canonical.com","username":"gustavosr98"},"last_update":"2024-05-01 18:33:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia4dedab2d4d5d7f4090be71115e79aa155906df2","subject":"Add the woodpecker charm to Openstack charms","status":"ABANDONED","created":"2023-01-10 18:59:54.000000000","updated":"2024-05-01 18:33:14.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32328339cebded2c3623e614d6dc8491a35b58a3","_number":869752,"virtual_id_number":869752,"owner":{"_account_id":33741,"name":"Gustavo Sanchez","email":"gustavo.sanchez@canonical.com","username":"gustavosr98"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-10 19:12:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-02-28 07:26:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 19:12:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-28 07:26:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"ed2b44f7ad8ebe6047e7060c42e2d0b060679593","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33741,"name":"Gustavo Sanchez","email":"gustavo.sanchez@canonical.com","username":"gustavosr98"},"date":"2023-01-10 18:59:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee6140d51e164b9fec60257108ad9ab9ba215a0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33741,"name":"Gustavo Sanchez","email":"gustavo.sanchez@canonical.com","username":"gustavosr98"},"date":"2023-01-10 19:02:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9e4466c79d8ef48304ca3de162e376f2bbd3767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 19:12:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f072242cdac7438faa80949b3ddeae4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31eea58e5a24faa8a5b197de03584e4 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6339ee92af8445dfadd123f97e17f216 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3676e82c1dec42d58cd2f267eb26a4ab : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b14344c8343a41be9f14033eb3d846d1 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/884261588aab443689ce288c565c7330 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/b0bc1888f38f42398eb9abb52070714f : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/eb069e316bf14d8fa9237ba06a992ff4 : FAILURE in 3m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5d6f5c9819e9a19dc0117e7eac44408687151675","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-23 02:23:26.000000000","message":"Topic set to project-update","accounts_in_message":[],"_revision_number":2},{"id":"442be9259af56a1e7729fcefb9c56986f3ef2104","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-28 07:26:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"32328339cebded2c3623e614d6dc8491a35b58a3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-01 18:33:14.000000000","message":"Abandoned\n\nI\u0027m abandoning this since the patch that this one depends on has now been abandoned. Please submit a new patch if you plan to pursue this again: https://review.opendev.org/c/openstack/project-config/+/869751","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"57400816c709668ec7d424b00766ebc2ac38d726","revisions":{"268a499b0b3570a2210eac4a57a36460652f2ab2":{"kind":"REWORK","_number":1,"created":"2023-01-10 18:59:54.000000000","uploader":{"_account_id":33741,"name":"Gustavo Sanchez","email":"gustavo.sanchez@canonical.com","username":"gustavosr98"},"ref":"refs/changes/52/869752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/52/869752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/52/869752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/52/869752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/52/869752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/52/869752/1"}}},"commit":{"parents":[{"commit":"568c4c6ba4bd8492e46ca3f8fcfc006c7dd5836f","subject":"Merge \"Fixing some nits in emerging-inactive project documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/568c4c6ba4bd8492e46ca3f8fcfc006c7dd5836f"}]}],"author":{"name":"Gustavo Sanchez","email":"gustavo.sanchez@canonical.com","date":"2023-01-10 18:59:06.000000000","tz":-240},"committer":{"name":"Gustavo Sanchez","email":"gustavo.sanchez@canonical.com","date":"2023-01-10 18:59:06.000000000","tz":-240},"subject":"Add the woodpecker charm to Openstack charms","message":"Add the woodpecker charm to Openstack charms\n\nThis charm deploy a testing tool that enables to execute\nactions running performance tests to evaluate a ceph cluster.\n\nDepends-On: Ic6b5f5a273d09dcd3f9cf60588d3e7fb90c4c695\nChange-Id: Ia4dedab2d4d5d7f4090be71115e79aa155906df2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/268a499b0b3570a2210eac4a57a36460652f2ab2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/268a499b0b3570a2210eac4a57a36460652f2ab2"}]},"branch":"refs/heads/master"},"57400816c709668ec7d424b00766ebc2ac38d726":{"kind":"REWORK","_number":2,"created":"2023-01-10 19:02:34.000000000","uploader":{"_account_id":33741,"name":"Gustavo Sanchez","email":"gustavo.sanchez@canonical.com","username":"gustavosr98"},"ref":"refs/changes/52/869752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/52/869752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/52/869752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/52/869752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/52/869752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/52/869752/2"}}},"commit":{"parents":[{"commit":"568c4c6ba4bd8492e46ca3f8fcfc006c7dd5836f","subject":"Merge \"Fixing some nits in emerging-inactive project documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/568c4c6ba4bd8492e46ca3f8fcfc006c7dd5836f"}]}],"author":{"name":"Gustavo Sanchez","email":"gustavo.sanchez@canonical.com","date":"2023-01-10 19:02:24.000000000","tz":-240},"committer":{"name":"Gustavo Sanchez","email":"gustavo.sanchez@canonical.com","date":"2023-01-10 19:02:24.000000000","tz":-240},"subject":"Add the woodpecker charm to Openstack charms","message":"Add the woodpecker charm to Openstack charms\n\nThis charm deploy a testing tool that enables to execute\nactions running performance tests to evaluate a ceph cluster.\n\nDepends-On: Ic6b5f5a273d09dcd3f9cf60588d3e7fb90c4c695\nChange-Id: Ia4dedab2d4d5d7f4090be71115e79aa155906df2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/57400816c709668ec7d424b00766ebc2ac38d726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/57400816c709668ec7d424b00766ebc2ac38d726"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"An informational vote on this revision","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Rollcall-Vote","description":"An informational vote from team members","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
