)]}'
{"id":"openstack%2Fgovernance~873845","triplet_id":"openstack%2Fgovernance~master~I8a1c1afdad7f2768e80c1efc5cdbcd914abc8e71","project":"openstack/governance","branch":"master","topic":"project-update","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-04-14 14:58:59.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-04-14 14:58:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8a1c1afdad7f2768e80c1efc5cdbcd914abc8e71","subject":"Add xstatic-angular-fileupload as Horizon team deliverables","status":"MERGED","created":"2023-02-15 07:11:23.000000000","updated":"2023-04-14 14:59:59.000000000","submitted":"2023-04-14 14:58:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873845-project-update","meta_rev_id":"5398edfa081dc63e77aece55038bec5e0733dbf5","_number":873845,"virtual_id_number":873845,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-14 14:58:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2023-04-14 14:48:15.000000000","_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2023-04-14 14:48:15.000000000","_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2023-04-10 21:13:14.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 07:11:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-16 02:01:57.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2023-03-28 22:00:04.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-04-14 14:48:15.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"417903c8b5e34e58178986f8c1fd94625ea4aeea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-15 07:11:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb532832fe54edd53935e940a7d5a34db7c29f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 07:11:47.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"07a95978b0efb21dde9d1fbe96d28076d4bdecc3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-15 07:13:03.000000000","message":"Topic set to project-update","accounts_in_message":[],"_revision_number":1},{"id":"d36b9338d42953776eb43f53d76e397b5c4f3fac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-15 16:50:11.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"899636b2f456f6e518ea9cfb9d36bbf6e51f5a33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 16:59:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03f888a3170944b291e600d6f1a5d4da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc1d4d9f9fbd4487bf6a1420869b1516 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82d1158f3a14f5b8247a0ff018bfdc1 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddda8d8b0a1c4a49a7ef9edfa51ad308 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfee55a7e1724a5097bf3a5fa8d85643 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80b52ec366d1420d9fae9ba45e33dca1 : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/e35ea84087b54d3c9768f1c5b5c3486d : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6be661df1e3144d6934de54f626e9d7f : FAILURE in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"552ddbbbc8a1dd61721a2f41f660536d3e3d99f1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-16 02:01:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c141e404f5b930e755e3f6249a74e34c3750e892","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-03-27 18:51:32.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"497b4b84b3bea0138e3fdcb6af41bb0c29e37a2e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-28 00:44:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3f8847f2a6aa70ea40fbc7eef4cd05996b60a6df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-03-28 17:25:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"53cdfecb2c48c9774f3e8e44533a1950b2fc602b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 17:37:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccd79530a3a349e0a1c1d664b4821fe1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7859d3c52dd84b4ba13558678228b155 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92e29ce568e94ef682e930557f067550 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2577b0f5befe4ccf86724a840a749d37 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56ce8ce7becf43819830194095ee0041 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3f79cc0526746e58b5a19492949d595 : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/a73cbd0cab084ada95da30f12799375a : SUCCESS in 2m 43s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c96b7895d211408380b61d1d04f4392e : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":3},{"id":"479f0c51fec0c6167644f8f4267c85f76363852b","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-03-28 17:38:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2336e861be886e4baefab1e0c29f824f434960fd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-28 22:00:04.000000000","message":"Patch Set 3: Rollcall-Vote+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0384ed08c5465df5233dc333816f5e7272061720","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-04-07 01:16:58.000000000","message":"Patch Set 3: -Rollcall-Vote\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b63e83b8915f7953d5930b836ded6f01bd6d4ea5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-04-10 19:55:34.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"3ae3bb21c57d2fae52fea3e323c3f1af8380b5e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 20:10:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/907efb98ff1443bdabdaf749325f4314\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8abc88df1eef4e769ee384b3a6bc8c80 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cffd58d7c6764d6795fbb1f5c5e8f181 : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7fdf827d169456cb752609c6ce717a1 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78fb859587e5476583164cac5c68824b : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0d5124b544b432797fc5182501292cc : SUCCESS in 2m 59s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/bed2d17c6b7d4755848e0a62823c726c : SUCCESS in 2m 55s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6f7f493e54cb40dc95e26d2d75214478 : FAILURE in 3m 00s","accounts_in_message":[],"_revision_number":4},{"id":"cef90dbc1ee4c3398de8d6b06ca64d6efc636637","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-04-10 20:44:26.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b40b8f9381ae49b4d4caa4e525d1f79a9a056b6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 20:53:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68f485fa3b7148a8a57dd5f78280177d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ec5b0c86dc450793a2c880cac936c8 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8300838a99cc47539cb43c38fb4a4b97 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad6fd1c574834c06b00eb3a6751a86f3 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58048110072344a4b3909a0b36501391 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbb1e558c5a846048fc3292c2038c89b : SUCCESS in 2m 49s (non-voting)\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/216c494b677c458692c1cd90c42c9b72 : SUCCESS in 2m 51s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/56bc178f6bb74f72961cffbf3729dcd7 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":5},{"id":"a25a850e7476dab049daf96ec81cc5e56a4a903e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-04-10 21:13:14.000000000","message":"Patch Set 5: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":5},{"id":"cfecad1c8f47fe7b69022f601a3d2a99b244d7b7","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-04-14 14:48:15.000000000","message":"Patch Set 5: Rollcall-Vote+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ff25ed8131a140a2e428023c1860abdec8e69521","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 14:48:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"62933fe03d58eb0e845eb6fdb07e3e05d821b960","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 14:58:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b832af02a924d7fba1f84018848b233\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/901424390e8b43daa238d46fcfac8ddf : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a9467902f654525bcbd9b00da935cbb : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/feb37722b66f4cdbaa75eeb23afd9687 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b804649811ca4ddc8e36ae09a646c8db : SUCCESS in 2m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1258dafdddf4490ca49130dd0d821b2e : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":5},{"id":"ae59b9e2b54dd272bc9296f3d0e486b00ae3d431","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 14:58:59.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":5},{"id":"5398edfa081dc63e77aece55038bec5e0733dbf5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 14:59:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8fe697bdadd4fb39926830870323758\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/9e582f91fc404d0883d23d48d923e6f3 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3c252a00cb81692360a7ad8d73057497fe7f2b7e","revisions":{"5ba0215617fdb25bb93dbfdb0695bdcadb3df373":{"kind":"REWORK","_number":1,"created":"2023-02-15 07:11:23.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/45/873845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/45/873845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/45/873845/1"}}},"commit":{"parents":[{"commit":"be6eadf1bde9c21f9b53ac61c2e1ee2ecd425c49","subject":"Merge \"Define 2023.2 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/be6eadf1bde9c21f9b53ac61c2e1ee2ecd425c49"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-15 07:07:06.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-15 07:11:19.000000000","tz":330},"subject":"Add xstatic-angular-fileupload as Horizon team deliverables","message":"Add xstatic-angular-fileupload as Horizon team deliverables\n\nxstatic-angular-fileupload pypi is used by Horizon.\nHorizon team discussed this topic in the horizon weekly\nmeeting and decided to move it under the horizon deliverables\nand keep them under the OpenStack namespace [1].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-02-01.log.html#t2023-02-01T16:09:57\n\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/873843\nChange-Id: I8a1c1afdad7f2768e80c1efc5cdbcd914abc8e71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/5ba0215617fdb25bb93dbfdb0695bdcadb3df373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/5ba0215617fdb25bb93dbfdb0695bdcadb3df373"}]},"branch":"refs/heads/master"},"39a55ba442a088a977895c59b0b7f371a55bf4c6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-15 16:50:11.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/45/873845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/45/873845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/45/873845/2"}}},"commit":{"parents":[{"commit":"be6eadf1bde9c21f9b53ac61c2e1ee2ecd425c49","subject":"Merge \"Define 2023.2 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/be6eadf1bde9c21f9b53ac61c2e1ee2ecd425c49"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-15 07:07:06.000000000","tz":330},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2023-02-15 16:50:11.000000000","tz":0},"subject":"Add xstatic-angular-fileupload as Horizon team deliverables","message":"Add xstatic-angular-fileupload as Horizon team deliverables\n\nxstatic-angular-fileupload pypi is used by Horizon.\nHorizon team discussed this topic in the horizon weekly\nmeeting and decided to move it under the horizon deliverables\nand keep them under the OpenStack namespace [1].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-02-01.log.html#t2023-02-01T16:09:57\n\nNeeded-By: https://review.opendev.org/c/openstack/project-config/+/873843\nChange-Id: I8a1c1afdad7f2768e80c1efc5cdbcd914abc8e71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/39a55ba442a088a977895c59b0b7f371a55bf4c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/39a55ba442a088a977895c59b0b7f371a55bf4c6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e4777daae74ed1e3916c19bd5b65ae2e4f7ba81a":{"kind":"REWORK","_number":3,"created":"2023-03-28 17:25:05.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/45/873845/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/45/873845/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/45/873845/3"}}},"commit":{"parents":[{"commit":"be6eadf1bde9c21f9b53ac61c2e1ee2ecd425c49","subject":"Merge \"Define 2023.2 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/be6eadf1bde9c21f9b53ac61c2e1ee2ecd425c49"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-15 07:07:06.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-03-28 17:24:43.000000000","tz":330},"subject":"Add xstatic-angular-fileupload as Horizon team deliverables","message":"Add xstatic-angular-fileupload as Horizon team deliverables\n\nxstatic-angular-fileupload pypi is used by Horizon.\nHorizon team discussed this topic in the horizon weekly\nmeeting and decided to move it under the horizon deliverables\nand keep them under the OpenStack namespace [1]. We also check\nwith moinwiki contributor and they are not using this pypi project [2].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-02-01.log.html#t2023-02-01T16:09:57\n[2] https://github.com/moinwiki/moin/issues/1326#issuecomment-1380310586\n\nNeeded-By: https://review.opendev.org/c/openstack/project-config/+/873843\nChange-Id: I8a1c1afdad7f2768e80c1efc5cdbcd914abc8e71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/e4777daae74ed1e3916c19bd5b65ae2e4f7ba81a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/e4777daae74ed1e3916c19bd5b65ae2e4f7ba81a"}]},"branch":"refs/heads/master"},"a036ad6cb9b945854cf98d8bc2baed8beadc4679":{"kind":"REWORK","_number":4,"created":"2023-04-10 19:55:34.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/45/873845/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/45/873845/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/45/873845/4"}}},"commit":{"parents":[{"commit":"be6eadf1bde9c21f9b53ac61c2e1ee2ecd425c49","subject":"Merge \"Define 2023.2 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/be6eadf1bde9c21f9b53ac61c2e1ee2ecd425c49"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-15 07:07:06.000000000","tz":330},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2023-04-10 19:55:33.000000000","tz":0},"subject":"Add xstatic-angular-fileupload as Horizon team deliverables","message":"Add xstatic-angular-fileupload as Horizon team deliverables\n\nxstatic-angular-fileupload pypi is used by Horizon.\nHorizon team discussed this topic in the horizon weekly\nmeeting and decided to move it under the horizon deliverables\nand keep them under the OpenStack namespace [1]. We also check\nwith moinwiki contributor and they are not using this pypi project [2].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-02-01.log.html#t2023-02-01T16:09:57\n[2] https://github.com/moinwiki/moin/issues/1326#issuecomment-1380310586\n\nNeeded-By: https://review.opendev.org/c/openstack/project-config/+/873843\nChange-Id: I8a1c1afdad7f2768e80c1efc5cdbcd914abc8e71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/a036ad6cb9b945854cf98d8bc2baed8beadc4679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/a036ad6cb9b945854cf98d8bc2baed8beadc4679"}]},"branch":"refs/heads/master"},"3c252a00cb81692360a7ad8d73057497fe7f2b7e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-04-10 20:44:26.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/873845/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/45/873845/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/873845/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/45/873845/5"}}},"commit":{"parents":[{"commit":"0e4eaca37c3697bd8ec0d0bd444488ef07bd0b04","subject":"Fix virtualpdu repo namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/0e4eaca37c3697bd8ec0d0bd444488ef07bd0b04"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-15 07:07:06.000000000","tz":330},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-04-10 20:44:26.000000000","tz":0},"subject":"Add xstatic-angular-fileupload as Horizon team deliverables","message":"Add xstatic-angular-fileupload as Horizon team deliverables\n\nxstatic-angular-fileupload pypi is used by Horizon.\nHorizon team discussed this topic in the horizon weekly\nmeeting and decided to move it under the horizon deliverables\nand keep them under the OpenStack namespace [1]. We also check\nwith moinwiki contributor and they are not using this pypi project [2].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-02-01.log.html#t2023-02-01T16:09:57\n[2] https://github.com/moinwiki/moin/issues/1326#issuecomment-1380310586\n\nNeeded-By: https://review.opendev.org/c/openstack/project-config/+/873843\nChange-Id: I8a1c1afdad7f2768e80c1efc5cdbcd914abc8e71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/3c252a00cb81692360a7ad8d73057497fe7f2b7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/3c252a00cb81692360a7ad8d73057497fe7f2b7e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Rollcall-Vote","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"An informational vote on this revision","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Rollcall-Vote","description":"An informational vote from team members","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
