)]}'
{"id":"openstack%2Fgovernance~874439","triplet_id":"openstack%2Fgovernance~master~I223471c52d65a02333b7f3897dc0a3c484b513f0","project":"openstack/governance","branch":"master","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-02-20 16:55:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I223471c52d65a02333b7f3897dc0a3c484b513f0","subject":"Revert \"Add guidelines about naming versions of the OpenStack projects\"","status":"MERGED","created":"2023-02-20 16:25:06.000000000","updated":"2023-02-20 19:20:23.000000000","submitted":"2023-02-20 16:55:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874439","meta_rev_id":"97d1c9036b37ae0c8c14e2cf85e3fa0e1c452d69","_number":874439,"virtual_id_number":874439,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-20 16:55:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-02-20 16:46:22.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-20 16:35:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-20 16:46:22.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-02-20 18:04:26.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"9f7d561402dd30778f73b38d11081eed7cda0d4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-02-20 16:25:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8f15abfa4c3e3755433c7b1c0dfe41560a53a74","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-02-20 16:26:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c5128810307c75bc408bef17d0dde52b9ccc711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 16:35:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e937b279702f4a49b195268f235d58d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f41e74bb51e4f67a222fdaf80ed0424 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb1809bd934c4acd9c9b8213fc997f03 : SUCCESS in 3m 31s\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/13a86dc73d914707aef23a666db4fed4 : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0064baf2457f4b00a56bae1e53d13901 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6e15e43a66358514325f6ecb1caa1914ae4d43dd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-02-20 16:46:22.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2ac6d509158cf9a45064c7c45d53644172e79ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 16:46:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"384df70b4ef46fcab8aebd6557a976f23d17c089","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 16:55:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52d9692e72404515b2d6b55b23efb88c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20d4eff955304c5994213b9edcc31340 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d38f8398d26d4a519abcf77f2c12aa5f : SUCCESS in 3m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bda0f8c2106b4fc3a97b6e45117f55bc : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f6e3ca29248a7e4dee985e19c21c10064c486213","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 16:55:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"81cffdbf336249c00776532f53d76b6a61974c18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 16:56:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a89b296d75a048db9340a8fa36d8908d\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/58de02294771472bb0bc6bb952344beb : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1},{"id":"a671d516fe249cfb8c5c487637f7e185fee5ca87","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-20 18:04:26.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"97d1c9036b37ae0c8c14e2cf85e3fa0e1c452d69","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-20 19:20:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdc25d3ceb0d34f30327f4995d35b7c5da40f6f1","revisions":{"fdc25d3ceb0d34f30327f4995d35b7c5da40f6f1":{"kind":"REWORK","_number":1,"created":"2023-02-20 16:25:06.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/39/874439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/39/874439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/39/874439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/39/874439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/39/874439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/39/874439/1"}}},"commit":{"parents":[{"commit":"2bcf1ee05e92dec905b75d65cfb3abecae6ad6b6","subject":"Merge \"Step 3: Retire repo from the Governance repo - Retire xstatic-font-awesome\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/2bcf1ee05e92dec905b75d65cfb3abecae6ad6b6"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-02-20 16:24:14.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-02-20 16:24:33.000000000","tz":-480},"subject":"Revert \"Add guidelines about naming versions of the OpenStack projects\"","message":"Revert \"Add guidelines about naming versions of the OpenStack projects\"\n\nThis reverts commit d71d7b59d22071f00790907622b4b5e82361cea3.\n\nChange-Id: I223471c52d65a02333b7f3897dc0a3c484b513f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/fdc25d3ceb0d34f30327f4995d35b7c5da40f6f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/fdc25d3ceb0d34f30327f4995d35b7c5da40f6f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Rollcall-Vote","status":"MAY"}]}],"submit_requirements":[]}
