)]}'
{"id":"openstack%2Fgovernance~874446","triplet_id":"openstack%2Fgovernance~master~Ibe5e4474f8e34c4d000c8e2576e685441beecad0","project":"openstack/governance","branch":"master","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-02-20 18:32:03.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-02-20 18:32:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibe5e4474f8e34c4d000c8e2576e685441beecad0","subject":"Revert \"Revert \"Add guidelines about naming versions of the OpenStack projects\"\"","status":"MERGED","created":"2023-02-20 16:56:51.000000000","updated":"2023-02-20 18:33:04.000000000","submitted":"2023-02-20 18:32:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"874446","meta_rev_id":"07d480b5142e928f5c21640130b887324ea97a0d","_number":874446,"virtual_id_number":874446,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-20 18:32:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-20 18:26:08.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2023-02-20 18:17:01.000000000","_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-20 18:26:08.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-20 17:06:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-20 17:41:16.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2023-02-20 18:26:08.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"568dcc2639ce2a75a96ac38ea9183915fed01835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-02-20 16:56:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c917e9fa50b7ca45c15a3bfe217756ff9a5dcd22","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-02-20 16:59:47.000000000","message":"Patch Set 1: Rollcall-Vote-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01e908024b73c938fcf4a436d6728c862108e1e2","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-02-20 17:03:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb202bf98deedef4953d68e2296e867734f51a6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 17:06:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cca6d0512d84dff8c392d05aca1b964\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/366b1b6420354a1dbd081b04c76e2639 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dda27901e29c473c937cf09524da5492 : SUCCESS in 3m 04s\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/808edb8d36144f29bbd61791882153d0 : SUCCESS in 2m 56s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/94d57272a1a74be28548b8e57f8b6369 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9a0854cbf3e823c7d2221cbc47589fdee31f7986","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-20 17:41:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48ac9a2656a0432426aa5af8f60f3b33b2962074","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-02-20 17:44:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b572205bc09f55eb0ee60e20b8c457622c4e99ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-02-20 18:15:40.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2e4cab3809649e21c711069aef5c6f64dafc4cfb","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-02-20 18:17:01.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ced2837ec3a3528130b59bac4c4aace44fd75852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 18:26:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fafef8bc3f3464f8116657a7f0691f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1f376377cc4f00b5d6429dee8d6e2f : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65b9980e08124f0daa96ee98de258a0d : SUCCESS in 2m 56s\n- openstack-tox-validate_atcs https://zuul.opendev.org/t/openstack/build/4464c38e36244b36ac52a09c9e053419 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/61ff6f53a0774498913f9da8962f2859 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"93738b579734f480dcaa741d5aa281debdb928f2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-20 18:26:08.000000000","message":"Patch Set 2: Rollcall-Vote+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"531386fde5285944803cadbcccd981d09b4a5b11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 18:26:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"791b516f6015452c8d61b9a78ce02712ead91287","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 18:32:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/118acfed6805471fa74e8abb0f69cabd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4313422d5be444abf031e679b053a3d : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7074f23b3792429a8c130b3e4f43fbe1 : SUCCESS in 3m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5762a9fe3a3a40469bbdeeebc72fd54f : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"906a7d25878cccbc148a16a640e58e50f4a70288","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 18:32:03.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2},{"id":"07d480b5142e928f5c21640130b887324ea97a0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 18:33:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e112b21f118442e6b05acdcd53b0f2c0\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/69a3bbdfed0e4c1d8168f1ccb8569564 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"723167429daf5bb3bc886dcecf94f97288204088","revisions":{"7c29a34512386283ae652a163f92c2ad8b262709":{"kind":"REWORK","_number":1,"created":"2023-02-20 16:56:51.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/46/874446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/46/874446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/874446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/874446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/874446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/46/874446/1"}}},"commit":{"parents":[{"commit":"fdc25d3ceb0d34f30327f4995d35b7c5da40f6f1","subject":"Revert \"Add guidelines about naming versions of the OpenStack projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/fdc25d3ceb0d34f30327f4995d35b7c5da40f6f1"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-02-20 16:56:13.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-02-20 16:56:47.000000000","tz":-480},"subject":"Revert \"Revert \"Add guidelines about naming versions of the OpenStack projects\"\"","message":"Revert \"Revert \"Add guidelines about naming versions of the OpenStack projects\"\"\n\nThis was reverted as a fast-track change that had diagreement\npost-merge. It is being resubmitted to allow discussion.\n\nThis reverts commit fdc25d3ceb0d34f30327f4995d35b7c5da40f6f1.\n\nChange-Id: Ibe5e4474f8e34c4d000c8e2576e685441beecad0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/7c29a34512386283ae652a163f92c2ad8b262709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/7c29a34512386283ae652a163f92c2ad8b262709"}]},"branch":"refs/heads/master"},"723167429daf5bb3bc886dcecf94f97288204088":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-20 18:15:40.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/46/874446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/46/874446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/874446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/874446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/874446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/46/874446/2"}}},"commit":{"parents":[{"commit":"fdc25d3ceb0d34f30327f4995d35b7c5da40f6f1","subject":"Revert \"Add guidelines about naming versions of the OpenStack projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/fdc25d3ceb0d34f30327f4995d35b7c5da40f6f1"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-02-20 16:56:13.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-02-20 18:15:40.000000000","tz":0},"subject":"Revert \"Revert \"Add guidelines about naming versions of the OpenStack projects\"\"","message":"Revert \"Revert \"Add guidelines about naming versions of the OpenStack projects\"\"\n\nThis was reverted under the understanding it was a fast-track change; it was not and therefore needs a separate formal-vote to change.\n\nThis reverts commit fdc25d3ceb0d34f30327f4995d35b7c5da40f6f1.\n\nChange-Id: Ibe5e4474f8e34c4d000c8e2576e685441beecad0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/723167429daf5bb3bc886dcecf94f97288204088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/723167429daf5bb3bc886dcecf94f97288204088"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Rollcall-Vote","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[]}
