)]}'
{"id":"openstack%2Fgovernance~903490","triplet_id":"openstack%2Fgovernance~master~Id5e72114e73a765285c99a21e23c4d6eaa3c15ea","project":"openstack/governance","branch":"master","topic":"formal-vote","attention_set":{},"removed_from_attention_set":{"35761":{"account":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"last_update":"2024-09-13 13:22:41.000000000","reason":"\u003cGERRIT_ACCOUNT_35761\u003e replied on the change","reason_account":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-09-11 17:43:41.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-09 05:47:29.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-09-29 18:42:24.000000000","reason":"Change was submitted"},"935":{"account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"last_update":"2024-09-29 18:42:24.000000000","reason":"Change was submitted"},"15993":{"account":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"last_update":"2024-09-11 12:19:17.000000000","reason":"\u003cGERRIT_ACCOUNT_15993\u003e replied on the change","reason_account":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-09-29 18:42:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id5e72114e73a765285c99a21e23c4d6eaa3c15ea","subject":"Retire all single charm repositories","status":"MERGED","created":"2023-12-12 11:23:06.000000000","updated":"2024-09-29 18:43:40.000000000","submitted":"2024-09-29 18:42:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"903490-formal-vote","meta_rev_id":"d0aca47475624ce5081b463faf9889dfda0b3eed","_number":903490,"virtual_id_number":903490,"owner":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-29 18:42:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"all":[{"value":1,"date":"2024-09-13 13:22:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},{"value":1,"date":"2024-09-26 02:31:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2024-09-18 19:11:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-09-11 16:51:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-09-11 16:32:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2024-09-29 16:46:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":0,"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},{"value":1,"date":"2024-09-26 02:31:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2024-09-18 19:11:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2024-09-12 05:36:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-09-11 16:51:29.000000000","_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":1,"date":"2024-09-11 12:19:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-09-11 16:32:59.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"}],"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-12 11:34:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-13 23:53:27.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2023-12-14 00:16:54.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-12-20 09:06:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-03-12 15:49:10.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2024-04-30 05:24:46.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2024-05-03 20:21:02.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2024-08-15 09:12:13.000000000","updated_by":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"reviewer":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"state":"REVIEWER"},{"updated":"2024-09-10 22:36:54.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-09-11 16:51:29.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-09-26 02:31:02.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"85a56f4a6f028229e6ac105c77cfe8c3841b0b07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2023-12-12 11:23:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83c8e90fc8351cab95cfcbac142254c89f673a29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 11:34:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77763010f0724a6ab5cd83fb24279d47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96a443184a58410c8b49991ed691f7ba : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9d40262bb0f470a840073c161fecda3 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19eeec8bd19f49d7b5ba543aa3ba7975 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6752ccb7d9b641918803b92a33e0b8f4 : SUCCESS in 3m 14s\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/674fe1bc6d7347f79f26b36d0bbad980 : SUCCESS in 2m 56s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4b9df7f2d71c45a1a3af26aac852bd93 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d0ada299af4bfc2b3deb9bd557452a700773fb05","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-12-13 23:42:43.000000000","message":"Topic sunbeam/drop-single-charm-repos removed","accounts_in_message":[],"_revision_number":1},{"id":"2f08a8901c6cfa488717128ba78d4a843fefa0ce","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-12-13 23:42:47.000000000","message":"Topic set to project-update","accounts_in_message":[],"_revision_number":1},{"id":"ecc906c16d62f6485d7dbae8307a62b586a4df8a","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-12-13 23:53:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c66d3f4109dd5086be68d763e94c86512cb10912","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-12-14 00:16:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7d75c13ad2b5dd5e127f6a8bee8ed419dfcf0df","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2023-12-14 08:49:21.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1e59fb038891c6c3c6c4e067241ad1c43b323cd5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-20 09:06:43.000000000","message":"Patch Set 1: Rollcall-Vote+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39e3c82771b15b2c7a8f9c4e97818c774d4c86e9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-12-20 19:56:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"664e8106d8546d040ab3ee196fb919f33d55ef74","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-21 09:42:35.000000000","message":"Patch Set 1: -Rollcall-Vote Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f96690d5187852df8b08f3ea571428c64a26a49","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2024-03-12 15:49:10.000000000","message":"Patch Set 1: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"426bd481b6004b689552fb96563daad1f54fdcf8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-30 05:24:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d3ecc6ee8909bff3d5b50218036dd0c8c0b4191","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-04-30 09:47:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e9fc654dfaba5ec012aca93f7416e5c95f07c38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-04-30 09:53:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Rollcall-Vote+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f9cdc2cbe9d81bbecf3ffd599d5179b526d030f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 10:03:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8934febea0454ed281eb96786b25a165\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60205d97a51949abb24c8b9a400bf34a : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32ab43cc04ab47d587eacb42d326ef24 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a27a10e0701b4e55a665c0da6713be0b : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc5f808fbf2d4d36b50f130e02659ef9 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bda81fa4008b4373a49f2abc966e2661 : SUCCESS in 8m 12s (non-voting)\n- governance-validate-legacy https://zuul.opendev.org/t/openstack/build/bc0377a82cb74fafbd5d476b33ae0e41 : FAILURE in 3m 44s\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/a62018c8614f4fe48da22f6d1a3afb40 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e3f8d5629d1f4e0e8785660749e4b298 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"1f9a2fe394dc3f85645c1db78e9fccb786d3940a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-04-30 10:35:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5d0ae626e954bf1785f5202c23018e6f78bb1b6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:00:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63c48297b44347aab605d166874201a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0879078d392249eeb609df1284969730 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85709270444445d7917abd342939f82c : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3218cc1aec0e4922830b6f4bdd44eac6 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb4479bba549432e8eca4a5c82131986 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5766c30254dc444db5baf0f4c9bf3f2f : SUCCESS in 5m 57s (non-voting)\n- governance-validate-legacy https://zuul.opendev.org/t/openstack/build/73bc37479eb24c37ae52cd4e7f43a222 : FAILURE in 4m 34s\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/129aa9b52b9043cb9dc4143ce39ef76a : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bba4f122fb2140969e793b7bd0bf4dd7 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":3},{"id":"a5ecb77d804327add691548945300245ce6b94ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-04-30 13:36:33.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"14e4906446e1dc02ccd87702e464a3ecc80984e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 13:45:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4c45a93e196435aa4a1c6e37d55d7cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e80f47eacb324123a7225f6aa574a0b0 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b34344131f46d6bbcc75ac62b9a5d2 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81fe1ffa017947f2b199d9e53d1714b3 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a2628b62f8284b43bf61f57c12ba9205 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2bf4d3919aca46d4ab830f7ce5582be9 : SUCCESS in 6m 54s (non-voting)\n- governance-validate-legacy https://zuul.opendev.org/t/openstack/build/db4f3e838fd14798b80b804ed2cbbadd : SUCCESS in 4m 06s\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/d6dbdb8e3ec341ce99bd8e7d3d6ed283 : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5e91165d968c407da5375175f7b461e3 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":4},{"id":"c8f90ed153ce3c0fa3c31337a08a00ee21124fba","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-04-30 13:46:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c29ee80c3c99749867daa731980446dcdfcdd125","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-30 17:13:45.000000000","message":"Patch Set 4: Code-Review-1 Rollcall-Vote-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"39cc0a94975754a21929896bcaabd0ebf3dcb79c","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-05-01 06:13:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a734b820255c1ebd08509ed51d8f3ef316cfc34a","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-05-01 06:14:06.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"4fa2bf62384f231e6ccdad9efc3aed7015dbe3de","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-01 14:56:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e24470bd98161f60099af9c9db09cc8b029c5962","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-03 20:21:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"98ff34938940646281146b97ae3d478a54e1e2a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-08-09 07:43:01.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Rollcall-Vote-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"8a84f28e978310e778f13eb017379c3a4cbe4058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 07:54:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09bab924748d4457909ed19a7bead17a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24e72b25425845338d98bf137abacc92 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc097a82ecaa41baa7f543aa9b07614e : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1420610b35c64dfd9dc57392aa2ced74 : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35f3127d8635497badb8cf68c66193e3 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/39c6a99c9b764896967c7d399c4eec33 : SUCCESS in 5m 23s (non-voting)\n- governance-validate-legacy https://zuul.opendev.org/t/openstack/build/85cc6cc4d27f4d1194c64ebf979d427a : FAILURE in 3m 44s\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/ea082baaee264bb18287d212f8e47acb : SUCCESS in 2m 49s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5299af96b3d460dbb22698f7c4bfb5c : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":5},{"id":"34c951c14fa45c8e1105ffa32a1ddff9d5649ae1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-15 06:18:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"224855663458133207ec01f3219692c769217f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 06:27:09.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c566d3f331704a0dac126e6c3ba48556\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b2e9b4c420f4999a6611dd4abaad5df : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d92327961a34cf9ab605ec992c6d2a4 : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ce2f027ab184bd1b46d0f7788dc289c : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d59aad0338e41b19a93bea90df8b5b6 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8a9e11183a34e68a1dc784ed69989a5 : SUCCESS in 4m 06s (non-voting)\n- governance-validate-legacy https://zuul.opendev.org/t/openstack/build/4f11e504c1164a6f87efa6f17c6b1cf5 : FAILURE in 3m 13s\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/c2231e7054b441368ed7768132af497d : SUCCESS in 2m 25s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/01bee79252974c088c52f6f80ea85976 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":5},{"id":"358941c7dcb6a37880a03858193d0e39b3387bed","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-08-15 08:48:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f7f77a3d68ea1d4d11365caa8efa814cc3f78ffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 08:58:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/246b3ac9810941608473388efa2c3796\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ced1567201f047aeaa08af1707dbe6ba : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e164f38363f4a408d9ba4c7c3d1bc08 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/684c2fb546ac46f9b35e7b9ca786fde5 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29337bdd739e4a1ea5218ed49b3f00d1 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d726a6389b24438815905320e25d1a3 : SUCCESS in 3m 52s (non-voting)\n- governance-validate-legacy https://zuul.opendev.org/t/openstack/build/d609537c90ba4232a7512e78d771f42f : SUCCESS in 3m 23s\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/7b26a487fa22477da4468271b846a1fc : SUCCESS in 2m 35s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/561716adf1f14aefb538c9dc9c59ced0 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":5},{"id":"011aa49a3e4c90dff2ad47b88361acda528fce23","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2024-08-15 09:12:13.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f98821cc411e3cf3bf4c687e5ab183462724279a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-15 17:36:16.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d7f861ce41f172a39646785094bac65875e6a25c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-08-21 15:54:11.000000000","message":"Topic sunbeam/drop-single-charm-repos removed","accounts_in_message":[],"_revision_number":5},{"id":"ff305e6579ed189e16635dd4f106316b3f42659e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-08-21 15:54:23.000000000","message":"Topic set to formal-vote","accounts_in_message":[],"_revision_number":5},{"id":"dfc296b4f13f8f580f92f2cfb6c4b5228e6506ad","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-10 22:36:54.000000000","message":"Patch Set 5: Code-Review+1 Rollcall-Vote+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ebe62644d3aa6c75044ed57bfe1b99d733a11b32","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-10 22:42:46.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"040bb7b60021929130fd6875b1ccc30bc379639b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-09-11 11:51:00.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Rollcall-Vote+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e2dc461c7f9e91174b6c176d88ec2949d8e7fe4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 11:59:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2ceadea512d4e7fb837dfedb43be952\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/726ad0c955ee4c18a25de8366d5b4d4b : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/637829959caf47e9a64ac3c797156d09 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29c74d5b7ad84f45aac5f68291fd3f12 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6ae82853fbf46bd89cfa11629edade8 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ae4e51a460247749d4a7ad02509f520 : SUCCESS in 2m 44s (non-voting)\n- governance-validate-legacy https://zuul.opendev.org/t/openstack/build/bd68418a9d9940228d1128214763b526 : SUCCESS in 3m 54s\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/f248b0040c434193a3b8dcf9c589aa56 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/22437b6ecfba4c8b82a57d189c76f24d : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":6},{"id":"2482c9a172f80b31988312d3828352c3f9a2f862","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2024-09-11 12:19:17.000000000","message":"Patch Set 6: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":6},{"id":"93901ba3bad2fc30076f431613e5182a5ccebc97","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-11 16:32:59.000000000","message":"Patch Set 6: Code-Review-1 Rollcall-Vote-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0fd6c9ef42b80743d9f2ec4454e2e385235e08f5","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-09-11 16:51:29.000000000","message":"Patch Set 6: Code-Review+1 Rollcall-Vote+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"689b77f82a9b97da384ce249be46623010bab5a5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-11 17:27:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e9fb95c48b8697b8e6375dad77be3a7fc35f6a14","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-11 17:38:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"64706468aee24fe899e8338a55a29158735b72b2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-11 17:43:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"95be7047d51e83a6bed53fbbb3f7ec94089d0720","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-09-12 05:36:26.000000000","message":"Patch Set 6: Rollcall-Vote+1","accounts_in_message":[],"_revision_number":6},{"id":"4c1c95f239e5be64e74cd3316e60eb6a9a25c27b","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2024-09-13 13:22:41.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"02c05102eaec1494ae3d12a002678ad360f2b374","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-18 19:11:53.000000000","message":"Patch Set 6: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":6},{"id":"060ae7028c43f45bca6aacbff4cb08728b6bdad5","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-09-26 02:31:02.000000000","message":"Patch Set 6: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":6},{"id":"51abb39b2d75be95310b2be25108dad51dc7cbcb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-29 16:46:52.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f91026219d4d0b18b10ed928930302b195dd7de3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-29 16:46:59.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6b57cadc243fd183e97e17ecff1199f764e0fe40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-29 17:25:31.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b5b391b32459298336d20b325eeaddfd68ae1007","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-29 18:42:23.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6a6345569204cc0b1207b269ffba3f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/599ae6e40bc84eea9630b8ac701369b7 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fa33f20c8764770aafe70a6f299e72e : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff3bd240adf242d3a9a451ff7e205b6c : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96482eda93df4663921c23e98522c46f : SUCCESS in 4m 15s\n- governance-validate-legacy https://zuul.opendev.org/t/openstack/build/0d6112893b9c4a81b2637a6b5c0e0789 : SUCCESS in 6m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8ed572a1fa984a84ac7ba0336bd70f39 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":6},{"id":"839221b94174098ab8d2dceb2ce09888e08b1911","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-29 18:42:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d0aca47475624ce5081b463faf9889dfda0b3eed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-29 18:43:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a838fd305664011a9d8da47cfc7bf8d\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/d8447a9bb02d4f13a2145706d21fb4d8 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c2abd8086335fb842d6dcae433b72f6fd6fea043","revisions":{"55ba00ab681dc544082f5eae930cd29356b4db82":{"kind":"REWORK","_number":1,"created":"2023-12-12 11:23:06.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/90/903490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/90/903490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/90/903490/1"}}},"commit":{"parents":[{"commit":"eff5aacd2e8afefb667b5fd19f229f851ed5e7f5","subject":"Create mono repo for sunbeam charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/eff5aacd2e8afefb667b5fd19f229f851ed5e7f5"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2023-12-12 11:21:04.000000000","tz":0},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2023-12-12 11:22:53.000000000","tz":0},"subject":"Retire all single charm repositories","message":"Retire all single charm repositories\n\nThe sunbeam-charms mono-repo now replaces all individual charm\nrepositories - remove from the Sunbeam project.\n\nChange-Id: Id5e72114e73a765285c99a21e23c4d6eaa3c15ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/55ba00ab681dc544082f5eae930cd29356b4db82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/55ba00ab681dc544082f5eae930cd29356b4db82"}]},"branch":"refs/heads/master"},"21e7be86d93d7724db31894fcf5763e7c96da5e4":{"kind":"REWORK","_number":2,"created":"2024-04-30 09:53:56.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/90/903490/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/90/903490/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/90/903490/2"}}},"commit":{"parents":[{"commit":"cbf0607bdce40d40105e7e2ea99b74af33383cb2","subject":"[trivial] fix typos in tc chair responsibilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/cbf0607bdce40d40105e7e2ea99b74af33383cb2"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2023-12-12 11:21:04.000000000","tz":0},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-04-30 09:52:16.000000000","tz":60},"subject":"Retire all single charm repositories","message":"Retire all single charm repositories\n\nThe sunbeam-charms mono-repo now replaces all individual charm\nrepositories - remove from the Sunbeam project.\n\nChange-Id: Id5e72114e73a765285c99a21e23c4d6eaa3c15ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/21e7be86d93d7724db31894fcf5763e7c96da5e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/21e7be86d93d7724db31894fcf5763e7c96da5e4"}]},"branch":"refs/heads/master"},"3d7a3d1f800d3e32cb1c4a12d4911ef6106bdae1":{"kind":"REWORK","_number":3,"created":"2024-04-30 10:35:59.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/90/903490/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/90/903490/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/90/903490/3"}}},"commit":{"parents":[{"commit":"cbf0607bdce40d40105e7e2ea99b74af33383cb2","subject":"[trivial] fix typos in tc chair responsibilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/cbf0607bdce40d40105e7e2ea99b74af33383cb2"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2023-12-12 11:21:04.000000000","tz":0},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-04-30 10:35:29.000000000","tz":60},"subject":"Retire all single charm repositories","message":"Retire all single charm repositories\n\nThe sunbeam-charms mono-repo now replaces all individual charm\nrepositories - remove from the Sunbeam project.\n\nUpdate validate-legacy.py to deal with repositories which don\u0027t\nuse a default branch of \u0027master\u0027.\n\nChange-Id: Id5e72114e73a765285c99a21e23c4d6eaa3c15ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/3d7a3d1f800d3e32cb1c4a12d4911ef6106bdae1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/3d7a3d1f800d3e32cb1c4a12d4911ef6106bdae1"}]},"branch":"refs/heads/master"},"08fe64c98d94173b26a128a03ce66404f1988327":{"kind":"REWORK","_number":4,"created":"2024-04-30 13:36:33.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/90/903490/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/90/903490/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/90/903490/4"}}},"commit":{"parents":[{"commit":"cbf0607bdce40d40105e7e2ea99b74af33383cb2","subject":"[trivial] fix typos in tc chair responsibilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/cbf0607bdce40d40105e7e2ea99b74af33383cb2"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2023-12-12 11:21:04.000000000","tz":0},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-04-30 12:53:50.000000000","tz":60},"subject":"Retire all single charm repositories","message":"Retire all single charm repositories\n\nThe sunbeam-charms mono-repo now replaces all individual charm\nrepositories - remove from the Sunbeam project.\n\nUpdate validate-legacy.py to deal with repositories which don\u0027t\nuse a default branch of \u0027master\u0027 and have in repository zuul\nconfiguration only (required to noop tests).\n\nChange-Id: Id5e72114e73a765285c99a21e23c4d6eaa3c15ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/08fe64c98d94173b26a128a03ce66404f1988327"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/08fe64c98d94173b26a128a03ce66404f1988327"}]},"branch":"refs/heads/master"},"331144012f45f114bba73fe4834baff2ed43ecc3":{"kind":"REWORK","_number":5,"created":"2024-08-09 07:43:01.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/90/903490/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/90/903490/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/90/903490/5"}}},"commit":{"parents":[{"commit":"7d97a0b2cd894e4122809c0ef516b228597e483d","subject":"Merge \"Follow up on the affiliation diversity hanlding\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/7d97a0b2cd894e4122809c0ef516b228597e483d"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2023-12-12 11:21:04.000000000","tz":0},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-08-09 07:42:29.000000000","tz":60},"subject":"Retire all single charm repositories","message":"Retire all single charm repositories\n\nThe sunbeam-charms mono-repo now replaces all individual charm\nrepositories - remove from the Sunbeam project.\n\nUpdate validate-legacy.py to deal with repositories which don\u0027t\nuse a default branch of \u0027master\u0027.\n\nChange-Id: Id5e72114e73a765285c99a21e23c4d6eaa3c15ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/331144012f45f114bba73fe4834baff2ed43ecc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/331144012f45f114bba73fe4834baff2ed43ecc3"}]},"branch":"refs/heads/master"},"c2abd8086335fb842d6dcae433b72f6fd6fea043":{"kind":"REWORK","_number":6,"created":"2024-09-11 11:51:00.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/90/903490/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/90/903490/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/90/903490/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/90/903490/6"}}},"commit":{"parents":[{"commit":"7d97a0b2cd894e4122809c0ef516b228597e483d","subject":"Merge \"Follow up on the affiliation diversity hanlding\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/7d97a0b2cd894e4122809c0ef516b228597e483d"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2023-12-12 11:21:04.000000000","tz":0},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-09-11 11:50:52.000000000","tz":60},"subject":"Retire all single charm repositories","message":"Retire all single charm repositories\n\nThe sunbeam-charms mono-repo now replaces all individual charm\nrepositories - remove from the Sunbeam project.\n\nUpdate validate-legacy.py to deal with repositories which don\u0027t\nuse a default branch of \u0027master\u0027.\n\nChange-Id: Id5e72114e73a765285c99a21e23c4d6eaa3c15ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c2abd8086335fb842d6dcae433b72f6fd6fea043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c2abd8086335fb842d6dcae433b72f6fd6fea043"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Rollcall-Vote","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"An informational vote on this revision","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Rollcall-Vote","description":"An informational vote from team members","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
