)]}'
{"id":"openstack%2Fgovernance~925997","triplet_id":"openstack%2Fgovernance~master~I521eaeef9eecf549fbf376c05ddaef3a15ca34c6","project":"openstack/governance","branch":"master","topic":"project-update","attention_set":{},"removed_from_attention_set":{"36848":{"account":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"last_update":"2024-08-08 19:09:07.000000000","reason":"removed on reply"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-08-09 16:57:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I521eaeef9eecf549fbf376c05ddaef3a15ca34c6","subject":"Re-initiate the DPL model for Freezer project","status":"MERGED","created":"2024-08-08 16:36:15.000000000","updated":"2024-08-09 16:58:31.000000000","submitted":"2024-08-09 16:57:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925997-project-update","meta_rev_id":"5010f090f84cef8308ee80f6302e8229c8769c05","_number":925997,"virtual_id_number":925997,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-09 16:57:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"all":[{"value":1,"date":"2024-08-08 19:09:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},{"value":1,"date":"2024-08-09 16:44:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2024-08-08 16:38:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2024-08-08 16:36:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},{"value":1,"date":"2024-08-09 16:45:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},{"value":1,"date":"2024-08-09 16:44:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2024-08-08 16:36:58.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-08 16:36:58.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-08-08 16:38:20.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"state":"REVIEWER"},{"updated":"2024-08-08 16:49:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-09 16:44:46.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"3a94041bce1c061e44fe911089d496ba378d231c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-08 16:36:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b44b294d950929addc5d4f9812df97af970c3eb6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-08 16:36:58.000000000","message":"Patch Set 1: Code-Review+1 Rollcall-Vote+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf31faccfc6dffed5fbe787367a94cfb097bec2d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-08 16:38:20.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e39d782f2c64e3da2598328bc44a0037658f5b74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 16:49:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23a5ba1d18ac46b1aaf9737804b92689\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8775bac2fb924b83959b2c928190e7d7 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/275104ba13604479b9a4ee8958808b82 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a7442a94b6b4b7baf04f2666b68c106 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53497fad00c14b8590080029cfcb5fa0 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/329951953b7943039dee00a918e985af : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/7074202f9f36437290d1fa5be05b9cba : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4086251160704239acdd52e01d78552c : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"c10543888046a143a1d11e2c622e2c53b2386acb","author":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"date":"2024-08-08 19:09:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3ef0657c3947bf01fd1ab785ec83bcbd33b09763","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-08-09 16:18:23.000000000","message":"Topic project-updates removed","accounts_in_message":[],"_revision_number":1},{"id":"d23b1a88912a21e1cd691cf50fcd78ffe719d3f4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-08-09 16:18:29.000000000","message":"Topic set to project-update","accounts_in_message":[],"_revision_number":1},{"id":"b9510fc0a96403cba8ac357a18180374c925b4ee","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-08-09 16:44:46.000000000","message":"Patch Set 1: Code-Review+1 Rollcall-Vote+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c8f8a69c9e62d97ad376de5cebd5e87c3ebbb1c2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-08-09 16:45:10.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"223abd55136a46933809666ed5b9297fd8487410","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 16:45:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ee1f6fe866496316209b21c161ce84af1f3ecbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 16:57:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b02d1c624e741cf82af7bce3fda4da1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a5567aaa204ade862b7d5c51049c5d : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/114d6419d58545ef96ecb2bdca5b0986 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6518dff00c324f3eb760729037bb92b5 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b2f0ec880b64258920325b40b37914b : SUCCESS in 2m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8c93445ab9ec4f669b08d92478729161 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ed353f0a1c97b8eccdc7400514b1d2d53dd06a48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 16:57:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5010f090f84cef8308ee80f6302e8229c8769c05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 16:58:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2bbea70dc3a4a358f7a9d4dca5433a1\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/52ee02f169cc4cf28cf21d0b8fdc5c82 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"415f54773c7196cc58543ccf33c5177496cff5e0","revisions":{"415f54773c7196cc58543ccf33c5177496cff5e0":{"kind":"REWORK","_number":1,"created":"2024-08-08 16:36:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/97/925997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/97/925997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/97/925997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/97/925997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/97/925997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/97/925997/1"}}},"commit":{"parents":[{"commit":"7c7818054175c7b2436bb6d4d4a53802a8d4bbd7","subject":"Reset the DPL model for Freezer project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/7c7818054175c7b2436bb6d4d4a53802a8d4bbd7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-08-08 16:35:51.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-08-08 16:35:51.000000000","tz":120},"subject":"Re-initiate the DPL model for Freezer project","message":"Re-initiate the DPL model for Freezer project\n\nDue to decision for resetting DPL model each cycle, I\u0027m opting-in for\nthe DPL model for Freezer project for the next cycle as well.\n\nChange-Id: I521eaeef9eecf549fbf376c05ddaef3a15ca34c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/415f54773c7196cc58543ccf33c5177496cff5e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/415f54773c7196cc58543ccf33c5177496cff5e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Rollcall-Vote","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"An informational vote on this revision","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Rollcall-Vote","description":"An informational vote from team members","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
