)]}'
{"id":"openstack%2Fgovernance~950446","triplet_id":"openstack%2Fgovernance~master~I5237376d79bc765c2941a0fcfcd02f766bbba0ae","project":"openstack/governance","branch":"master","topic":"documentation-change","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-05-29 13:44:08.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-29 13:35:09.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I5237376d79bc765c2941a0fcfcd02f766bbba0ae","subject":"Fix outdated info on the tc-guide","status":"MERGED","created":"2025-05-20 16:50:27.000000000","updated":"2025-05-29 13:46:37.000000000","submitted":"2025-05-29 13:44:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950446-documentation-change","meta_rev_id":"a30131b15d3b62781c7a9c52303322d4a19a0113","_number":950446,"virtual_id_number":950446,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-29 13:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-22 06:40:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-05-29 13:46:37.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-05-29 13:35:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-22 06:40:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-05-29 13:35:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-20 17:08:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-20 17:22:27.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2025-05-22 06:40:38.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-05-29 13:35:09.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"4f0b6f611c88b3ced62c3be78d29f71057f0225c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-20 16:50:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"558bb4f3ea1f00dcdb410f93ecfb838168eb091d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 17:08:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b9fa53a74b9416aa06e7e02ccad3064\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/238df87a70eb4487b8dc74b2d8bee6ce : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e30060d0fbca489783c6b2a47de4eba3 : SUCCESS in 3m 10s\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/03ad4fdca2364e1fb2ad1b7ecfc9d420 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/189a2806497e4455931f8d37161fb106 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3faf9ede684de1d438d8be349db3f3010d553a9d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-20 17:22:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4108777eabbfaf675839aff7427023460d94ff0f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-21 21:52:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2cdaff2056579801219a63e40766d3357fd1d6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-22 06:13:35.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"24e393c41b2f82e83e30da11982d0aee6c826343","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-22 06:14:40.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"dbf72975fc17c689d137392da7d7c5738a4e85e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-05-22 06:15:29.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"7f5459b67bd1c77501fe3d06c29c4b54a24885c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 06:25:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bee021748d6e4fe280ab0afe3b2af38a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/425d5cad0ff147d4956fdd27714f0697 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5db0b396666e471fb3db227a68fe9f06 : SUCCESS in 3m 40s\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/9d5d249ef17a438587374487d277adb5 : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7811bc476f784b6497946463e9cdd7b3 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":4},{"id":"3f9f7fa30026ab2766cb864260ee7bd7e3943056","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-22 06:40:38.000000000","message":"Patch Set 4: Code-Review+1 Rollcall-Vote+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ae82ecf527e8bf940221968270def2c0c05fcf7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-29 13:35:09.000000000","message":"Patch Set 4: Rollcall-Vote+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"df61910a0d50f588780cc4c6b3c32191c97d0e5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 13:35:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"78a55bd96260865f6e4015d16072571472e63d7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 13:44:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d72cfdadce214ab8822f240adb931c44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78070c16a9b94d7b82720ba9abdb414d : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db3d3e415d9d46eba25761b95115357f : SUCCESS in 3m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/223aca789df8429c947723dfe014ba46 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":4},{"id":"ff06604fec46b977049e5510aa40ca3cc9c29d29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 13:44:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"699c78ec076d7f8e6d36717bfae151536b93e47f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-29 13:44:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d3376f730ea3ce1fa7e707436ee11172ee6122e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 13:45:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb480b52341044468129e9421e754ae0\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/62343d7e4a53445cba0d7a84116f3ac8 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4},{"id":"a30131b15d3b62781c7a9c52303322d4a19a0113","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-29 13:46:37.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"686fc81f0ffd2f539d6cfa3770209d816d6a0dfa","revisions":{"826ad741ea553fcf7b6c024b375c04f908593c99":{"kind":"REWORK","_number":1,"created":"2025-05-20 16:50:27.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/46/950446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/46/950446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/950446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/950446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/950446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/46/950446/1"}}},"commit":{"parents":[{"commit":"8cdd19ac689ec9a0ffddcd8dfd704df90ad72d2e","subject":"Merge \"Retire openstack-helm-infra repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/8cdd19ac689ec9a0ffddcd8dfd704df90ad72d2e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-05-20 16:49:18.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-05-20 16:50:06.000000000","tz":-420},"subject":"Fix outdated info on the tc-guide","message":"Fix outdated info on the tc-guide\n\nMeeting times can change with each new TC term;\nlet\u0027s redirect to the eavesdrop page rather than\nhave to update it in multiple places.\n\nChange-Id: I5237376d79bc765c2941a0fcfcd02f766bbba0ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/826ad741ea553fcf7b6c024b375c04f908593c99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/826ad741ea553fcf7b6c024b375c04f908593c99"}]},"branch":"refs/heads/master"},"f08a329c03973517876c208fac7a3417c36130a4":{"kind":"REWORK","_number":2,"created":"2025-05-22 06:13:35.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/46/950446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/46/950446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/950446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/950446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/950446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/46/950446/2"}}},"commit":{"parents":[{"commit":"8cdd19ac689ec9a0ffddcd8dfd704df90ad72d2e","subject":"Merge \"Retire openstack-helm-infra repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/8cdd19ac689ec9a0ffddcd8dfd704df90ad72d2e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-05-20 16:49:18.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-05-22 06:13:35.000000000","tz":0},"subject":"Fix outdated info on the tc-guide","message":"Fix outdated info on the tc-guide\n\nMeeting times can change with each new TC term;\nlet\u0027s redirect to the eavesdrop page rather than\nhave to update it in multiple places.\n\nChange-Id: I5237376d79bc765c2941a0fcfcd02f766bbba0ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f08a329c03973517876c208fac7a3417c36130a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f08a329c03973517876c208fac7a3417c36130a4"}]},"branch":"refs/heads/master"},"f60230019d7d81cc67d62f52fee23974f3628071":{"kind":"REWORK","_number":3,"created":"2025-05-22 06:14:40.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/46/950446/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/46/950446/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/950446/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/950446/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/950446/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/46/950446/3"}}},"commit":{"parents":[{"commit":"8cdd19ac689ec9a0ffddcd8dfd704df90ad72d2e","subject":"Merge \"Retire openstack-helm-infra repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/8cdd19ac689ec9a0ffddcd8dfd704df90ad72d2e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-05-20 16:49:18.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-05-22 06:14:40.000000000","tz":0},"subject":"Fix outdated info on the tc-guide","message":"Fix outdated info on the tc-guide\n\nMeeting times can change with each new TC term;\nlet\u0027s redirect to the eavesdrop page rather than\nhave to update it in multiple places.\n\nChange-Id: I5237376d79bc765c2941a0fcfcd02f766bbba0ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f60230019d7d81cc67d62f52fee23974f3628071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/f60230019d7d81cc67d62f52fee23974f3628071"}]},"branch":"refs/heads/master"},"686fc81f0ffd2f539d6cfa3770209d816d6a0dfa":{"kind":"REWORK","_number":4,"created":"2025-05-22 06:15:29.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/46/950446/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/46/950446/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/950446/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/950446/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/46/950446/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/46/950446/4"}}},"commit":{"parents":[{"commit":"8cdd19ac689ec9a0ffddcd8dfd704df90ad72d2e","subject":"Merge \"Retire openstack-helm-infra repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/8cdd19ac689ec9a0ffddcd8dfd704df90ad72d2e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-05-20 16:49:18.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-05-22 06:15:28.000000000","tz":0},"subject":"Fix outdated info on the tc-guide","message":"Fix outdated info on the tc-guide\n\nMeeting times can change with each new TC term;\nlet\u0027s redirect to the eavesdrop page rather than\nhave to update it in multiple places.\n\nChange-Id: I5237376d79bc765c2941a0fcfcd02f766bbba0ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/686fc81f0ffd2f539d6cfa3770209d816d6a0dfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/686fc81f0ffd2f539d6cfa3770209d816d6a0dfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Rollcall-Vote","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"An informational vote on this revision","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Rollcall-Vote","description":"An informational vote from team members","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
