)]}'
{"id":"openstack%2Fgovernance~953549","triplet_id":"openstack%2Fgovernance~master~I2cb6ce5d3297bd97c5097494ae8072ae78662b8f","project":"openstack/governance","branch":"master","topic":"project-update","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-07-10 20:49:29.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-10 20:49:29.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-07-09 17:26:12.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-07-10 20:49:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2cb6ce5d3297bd97c5097494ae8072ae78662b8f","subject":"Deprecate shade, os-client-config","status":"MERGED","created":"2025-06-27 09:30:05.000000000","updated":"2025-07-10 20:50:23.000000000","submitted":"2025-07-10 20:49:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953549-project-update","meta_rev_id":"91400fc750cd0b5fe17603b3ddfb3570d9734fb1","_number":953549,"virtual_id_number":953549,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 20:49:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2025-07-10 20:38:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2025-07-09 17:04:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-07-09 17:04:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-09 17:04:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2025-07-10 20:38:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2025-07-10 20:38:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2025-07-09 17:04:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-07-09 17:04:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-09 17:04:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":1,"date":"2025-07-09 17:26:12.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-27 09:39:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-27 10:46:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-06-27 10:46:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-06-27 10:46:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-06-27 14:04:51.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-06-27 16:35:37.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2025-06-30 06:45:46.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-07-08 17:22:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"cf0ec5c9f5c074e96129540dbfed383e7b6480c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-27 09:30:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd83d3c6a62536c11ecbb4bab0b6d5f398fe9d0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 09:39:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3d074dcbdae480c8e20fa3bf17906ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df249f5e74c412c99e3b4536a4ac2d3 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3b7d25adc344ad89478bfd232f2f372 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ef1199b22564509bde8f7eab37124ee : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0dde05f71d314e2fbcd7f2d5805e4a38 : SUCCESS in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fa03ab1cb134d61aca29ac3679863c9 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/0f42c2d23fdc45f388792c0be7c5442d : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0da73d65585c458184f48903778b0dcf : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7fa722bd73cb6b320dca4356fd39c9534d13d89c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-27 10:46:53.000000000","message":"Patch Set 1: Code-Review+1 Rollcall-Vote+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19b6972dec4c034e741d4d373203ac25e6e50f66","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-06-27 14:04:51.000000000","message":"Patch Set 1: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"60d763630946eb844eabfc53a8394df4b7933232","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-27 16:35:37.000000000","message":"Patch Set 1: Rollcall-Vote-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2e1f7a21fe9f6ef3c5bbb8b367a3608dbcc0769","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-06-30 06:45:35.000000000","message":"Topic sdk removed","accounts_in_message":[],"_revision_number":1},{"id":"dfd605d21dca8f0a8924a893cb965877d9ddb72e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-06-30 06:45:41.000000000","message":"Topic set to project-update","accounts_in_message":[],"_revision_number":1},{"id":"36023bd105c0085ebf228468a20562d3bb3f751b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-06-30 06:45:46.000000000","message":"Patch Set 1: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"ddc2ce8725fb1babc04b32e2fa69e3571bda98f7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-30 15:22:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73d5bd6f8773b3720274454e2851f0131db7ce24","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-08 17:22:56.000000000","message":"Patch Set 1: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":1},{"id":"49cc275bb1ecfb871c2c8606e3c542af20125daf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-09 17:04:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n* Rollcall-Vote+1, Rollcall-Vote-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c109c0ca220684b5573d3bae0aaeceea840d3153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 17:25:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a336eac64b364b09b79be31a2814201e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e15cd7498e0549d3b0286569e58877bd : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56aa8e9acf39494abdbbb63b9d55a12b : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b5f42183b514d6db04992ecdcb5cab0 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21b77c8af3e64322b799ac6dca72ed3b : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c863fa27d1cf47e79743b99c175769ac : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/09819f4a919146b09bc2d7440f75d5f0 : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0cf3defd1bbd4f0b83f4c7e36e7d4dca : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b9ec5f9a1d6848ba3a859c1e68ac2513e3dd5a6f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-07-09 17:26:12.000000000","message":"Patch Set 2: Rollcall-Vote+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8ba6a5065563b574b9e0cbe0852d6f4de76f92f9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-07-10 20:38:57.000000000","message":"Patch Set 2: Code-Review+1 Rollcall-Vote+1 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7b69bbfc83a1685719b7a3d7246778bf1df7769a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:39:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7a4c2241439e5fd32e4a51378a444a55f26ee7df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:49:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcbadaa7becd4114bee503e9b689982a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd37d240ab9d472e9439d1c96a1bc432 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4543ce4a615b44f8bdfd0b310e86bc78 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7556a0f816646c8ac2ef4b99390c816 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bca73a388f7a49b4b65d26e66af7d8df : SUCCESS in 3m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ca55c546b747447690e61d9594c58833 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":2},{"id":"f86b25a2f6a75a99a6aa38f08441a099a4fb3a2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:49:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"91400fc750cd0b5fe17603b3ddfb3570d9734fb1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:50:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c6c78e216784cefba24eefad3b7621d\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/fa7b903d1fe54b28843281de127629be : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ae1e276cea2e54342fba035a28c94b083aa1e2d","revisions":{"92c567ccdd07fe9675826f65d168a5e6b9e4ac32":{"kind":"REWORK","_number":1,"created":"2025-06-27 09:30:05.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/49/953549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/49/953549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/953549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/953549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/953549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/49/953549/1"}}},"commit":{"parents":[{"commit":"050b84dd815895cf5dc2694b808fed5ee70a11f1","subject":"Add service-types-authority to SDK deliverables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/050b84dd815895cf5dc2694b808fed5ee70a11f1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-27 09:29:54.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-27 09:29:54.000000000","tz":60},"subject":"Deprecate shade, os-client-config","message":"Deprecate shade, os-client-config\n\nThese have both been superseded by openstacksdk. Make our reference\ndocs reflect this reality.\n\nChange-Id: I2cb6ce5d3297bd97c5097494ae8072ae78662b8f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/92c567ccdd07fe9675826f65d168a5e6b9e4ac32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/92c567ccdd07fe9675826f65d168a5e6b9e4ac32"}]},"branch":"refs/heads/master"},"8ae1e276cea2e54342fba035a28c94b083aa1e2d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-09 17:04:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/49/953549/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/49/953549/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/953549/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/953549/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/49/953549/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/49/953549/2"}}},"commit":{"parents":[{"commit":"1ca66b9119f897640875d1a10074074b9d5f7b98","subject":"Merge \"Remove mentions to CLA in licensing guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/1ca66b9119f897640875d1a10074074b9d5f7b98"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-27 09:29:54.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-07-09 17:04:40.000000000","tz":0},"subject":"Deprecate shade, os-client-config","message":"Deprecate shade, os-client-config\n\nThese have both been superseded by openstacksdk. Make our reference\ndocs reflect this reality.\n\nChange-Id: I2cb6ce5d3297bd97c5097494ae8072ae78662b8f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/8ae1e276cea2e54342fba035a28c94b083aa1e2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/8ae1e276cea2e54342fba035a28c94b083aa1e2d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Rollcall-Vote","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"An informational vote on this revision","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Rollcall-Vote","description":"An informational vote from team members","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
