)]}'
{"id":"openstack%2Fgovernance~961598","triplet_id":"openstack%2Fgovernance~master~I110dc88b2032971049e4676a5fc3b112cf8e916f","project":"openstack/governance","branch":"master","topic":"election-results","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-09-24 00:46:32.000000000","reason":"Change was submitted"},"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2025-09-23 21:49:21.000000000","reason":"\u003cGERRIT_ACCOUNT_14482\u003e replied on the change","reason_account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2025-09-23 17:13:43.000000000","reason":"\u003cGERRIT_ACCOUNT_5890\u003e replied on the change","reason_account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-09-23 19:07:35.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-09-24 00:46:32.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-09-24 00:46:32.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2025-09-24 00:46:32.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2025-09-24 00:46:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I110dc88b2032971049e4676a5fc3b112cf8e916f","subject":"Add TC/PTL results from 2026.1 election","status":"MERGED","created":"2025-09-18 07:28:34.000000000","updated":"2025-09-24 00:47:23.000000000","submitted":"2025-09-24 00:46:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"961598-election-results","meta_rev_id":"481256b06b82d3df4fca4d0bde1b33517d75a3bd","_number":961598,"virtual_id_number":961598,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-24 00:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-09-23 07:53:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2025-09-23 17:13:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2025-09-23 19:07:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2025-09-23 19:07:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2025-09-23 17:13:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2025-09-23 19:07:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-18 07:40:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-18 13:00:41.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2025-09-19 02:45:02.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-09-19 08:51:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-09-22 07:45:59.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-09-22 16:32:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-09-22 16:44:42.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2025-09-22 16:44:42.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2025-09-22 18:47:01.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2025-09-23 06:49:52.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2025-09-23 07:53:34.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"8fd8b02a17240041c98728a3fdb8ab1c17b6cf8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-18 07:28:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23c6d73133e78c62614785b461cd688b9d627ad7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 07:40:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ad8b3dc4614082b136aa65dd1c611c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d80ce16512f4c50974ffee8f01c4da5 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c7a2ac5d8084e7e92e9adfbbdcbe547 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3500a36e1924548b59cfce07723b86e : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/782d59e18c4d4f209fcac0555ba1d972 : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9532377ab9324e5f9176be27502b06ef : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/f7df05e4ccbf4eea9812a74c347cb66f : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f7cf6c2dd71f47fcb21ed67100fd633f : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9011b9ab17c7db30177ea74acf57989ac139d52c","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-09-18 13:00:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1457d948c8b526696c89dbbd2a29f4f6600c6291","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-19 02:45:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0a017d4d252999060fc10401b1995617aed41965","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-19 07:38:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"76e842a917982843ed54104dbd13e885de49d612","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-19 07:41:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db6b8ed781fd8f7cc8468ad1433e17159fd87e45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 08:04:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e3b77e38b474739af7eaf1d65702eaa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005b4d5ad93b4d71af83a1f2aa08329f : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8726d066e55d4f648c69747210dd0d82 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12a605e5963e4d2786ec749bf4c71767 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7673b204ff7469ba1a4cf5a088cf91a : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da637ac3ef124fcaafb84ba28888e879 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/97792f6f5c6f4298a34485c5e05a4719 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d49248f47bba4d33aca8c4aaf84c0d78 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"19143fb02f9d654766d6e735b20e34c5db988192","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-19 08:51:17.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a6c2c49ffcfce72ba7e940b48614fe70b0ae68a8","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-09-19 22:17:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1359ae1dba6824c16092d16a5f044f8d1f006860","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-22 07:43:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"785ab03ac2d0b8b83105418a543ab6e9397e6c3c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-22 07:43:42.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"332f51bd9b81028c9c14aee10030838e1244d6e9","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-09-22 07:45:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9db648fbd6310c6f99eedd66515d36968fb695ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 07:58:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/305d945af77741bd9d1cece90a911671\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47707fc9346642ac96c8834d67f9652d : FAILURE in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a05079347b49a4be437106db6cc15a : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42f610075c034b10b4447708d63a895d : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b434e339503c4394bfd6f8343eb1a775 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b1b3b9cb73e844ef9048c7ff331a3bf8 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/bbcb9c635d214998bc2747528d06d6c8 : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6f39ca90c0824d04b714f656a583486d : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"6428c4a827794794b96b37f3c845915d7c88d8a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-22 08:24:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5ba18acbd9e35eb791b34ecd67a842d2e59544ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 08:37:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cbd1a0e3dd94d79a5c404a7f9f97ada\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a74ba8f7bdc246f08604729526eff707 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2becd977112041f2b59a650fc5a0daad : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ac45a48d1f541ce86b68f6d948cd9d0 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85f8d3b727f64a1ba5f67c0733f180b7 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98c630864a8946fba4fa49a91abbe771 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/5c5631e1bba54d01808afdb0c66dfb3a : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bac6bb9ed4424d89ad5b09ac1ae6c7f2 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":4},{"id":"374a73387fd68e4c2a764855594dc64ecbac6273","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-22 16:28:55.000000000","message":"Topic I110dc88b2032971049e4676a5fc3b112cf8e916f removed","accounts_in_message":[],"_revision_number":4},{"id":"414141a61e01dcd3f0e9a85c9b36053fc6029be5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-22 16:29:03.000000000","message":"Topic set to election-results","accounts_in_message":[],"_revision_number":4},{"id":"c3e13656792eec7a20885bfd9d783b75e8fa47ee","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-22 16:29:31.000000000","message":"Patch Set 4: Code-Review+1 Rollcall-Vote+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e8cd62d31122c3e21c259ec8daa0fbdb349345ab","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-22 16:32:07.000000000","message":"Patch Set 4: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":4},{"id":"3a36d433c4e7228280d501a98efad826d8e6a4a7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-22 16:42:29.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a5ddc15ff3c92e6e103b5bf85e404278006344a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 16:42:32.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee2acbff4a1c4ef19a4198f3d71200d5\n\n\nWarning:\n  Change 961596,2 in project openstack/election does not share a change queue with 961598 in project openstack/governance","accounts_in_message":[],"_revision_number":4},{"id":"d71bb551f3a99448aa3b1c715d03945b33240798","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-22 16:44:42.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c4a8f8443f7d18623a5fedd34d2f3678b80fa60f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 16:53:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7014c9b2b40d4fc8a88d9d052f69a18f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb3fd1aa27b4c45853d9409c990a0e3 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afbf331f17974683aff9509041fc8c52 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a64c30d753ba461ca4d2c8f4ba00bf66 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf5181f5b2cb41f0902e40703bcd47ee : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f7a8f3cf9e04ac7945f952a4529f243 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/0761a4ca41834aff9da0dc45c3ecdd62 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0a7afaf891724d14b839ca4d0bd66589 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":4},{"id":"e43feb2ff611809e6dca40edf94bd3057908f5d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 16:53:30.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb0ec0e177104d9f8cc3aef43e55ef47\n\n\nWarning:\n  Change 961596,2 in project openstack/election does not share a change queue with 961598 in project openstack/governance","accounts_in_message":[],"_revision_number":4},{"id":"55d052c9a45d5fc5de7e36b4bc91af9f65f81a65","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-09-22 18:47:01.000000000","message":"Patch Set 4: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":4},{"id":"4a609aa0e73c5b229004dd9d34c259ac5462def3","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-09-23 06:49:52.000000000","message":"Patch Set 4: Rollcall-Vote+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7a40e0fb2d8002e61f29fa30dacb8b6f773bfb36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-23 07:52:38.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Rollcall-Vote+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"18f1cdf39aa75d682ea9eb559d672abf10012027","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-23 07:53:34.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bc0e27566bb78029b842b02eb5f2e4dbaa051d69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 08:02:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1afa6e0368b549cbbc19575d7ecec923\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71f947461724eb9be86c8cef9e0b2a4 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1e4108e2b914710a076afdce9f92ab3 : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea6fbae06d24458d92191b91ca2e5bdb : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05c2c98f17484aad9d2195890b4eb4b8 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd86a20f030e46eba64a34fb582cf954 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/bf361fe8dadc409ba083dfa08b2ee88a : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/10132d2fed73420bbd1249918f81dd68 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":5},{"id":"437074c462bef6199d2ead09107627117d65daba","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-09-23 17:13:43.000000000","message":"Patch Set 5: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":5},{"id":"2830a58442c53680cf89b8d0ecc4d5ce88a6f3ae","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-23 19:07:35.000000000","message":"Patch Set 5: Code-Review+1 Rollcall-Vote+1 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"eb7cdcc4efd0e8287744ad8844c7b9dc93b476cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 19:07:37.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/01aaccd867874cdaa8b4b988015b6b70\n\n\nWarning:\n  Change 961596,2 in project openstack/election does not share a change queue with 961598 in project openstack/governance","accounts_in_message":[],"_revision_number":5},{"id":"2a691f3beb60b85ff1ced6eaee59eb6f85b964de","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-09-23 21:49:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fde14a4bec68d4ba88349744aff27ef657078f87","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-09-24 00:25:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"67f4e8b4e619c938dfbae0ae73f4f10040497a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 00:37:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4f011d948064433ac42e3c953a3f638\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2289d8bb3d4b40b2beb64de4fc554da5 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4d9a4cf5eb542c8a15c84871ca35cbd : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d57d05e97ba48e7b091682af90960c3 : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/200e820a79e544e1ab57fa39ffb2a148 : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e29709e33c4347f993b0b24c3bd7ca36 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate_acs https://zuul.opendev.org/t/openstack/build/033c1b474aba40e890d2215c8c1d1ffd : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f17bbe9a7e0b4ce2b7c50f836f362e05 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":5},{"id":"12ad26e7a57b64a4881f22ef1fd8fb7a890680aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 00:37:29.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"69aad2c4455321383332ed4c36e8416a1280873b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 00:46:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47fb172d23114d36a5b8768a3147aa9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf81987bd3f94ffbb1f363d0314f715d : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd7adc6c95c646fa9a16a4f0871baf21 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bdfd338143da44afb5d7d7888bb33837 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f5c33b387114b6aba002dee8215688b : SUCCESS in 3m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a5d1edeb1f0249ff951e34c357c67e18 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":5},{"id":"a01999bfbc5ee0de5efb166fc157b0f805882df8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 00:46:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"481256b06b82d3df4fca4d0bde1b33517d75a3bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 00:47:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b79b9ffdafc74d50aeb1d182360f55bc\n\n- promote-governance-tc https://zuul.opendev.org/t/openstack/build/06f5f8bf22904b3587d73144128449da : SUCCESS in 42s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c8962e0c0fc84c7183da92d3e2479a52623c419f","revisions":{"2b072a725bcd9454dc2d2855327e0bfd0bd9bdd6":{"kind":"REWORK","_number":1,"created":"2025-09-18 07:28:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/98/961598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/98/961598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/98/961598/1"}}},"commit":{"parents":[{"commit":"0f8e416775c54ce85dcd15b6dc18895cc7600214","subject":"Merge \"Show inactive project status in project.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/0f8e416775c54ce85dcd15b6dc18895cc7600214"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-18 07:27:42.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-18 07:28:30.000000000","tz":120},"subject":"Add TC/PTL results from 2026.1 election","message":"Add TC/PTL results from 2026.1 election\n\nAlso clear out existing TC roles and prepare for nominations.\n\nDepends-On: https://review.opendev.org/c/openstack/election/+/961596\n\nChange-Id: I110dc88b2032971049e4676a5fc3b112cf8e916f\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/2b072a725bcd9454dc2d2855327e0bfd0bd9bdd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/2b072a725bcd9454dc2d2855327e0bfd0bd9bdd6"}]},"branch":"refs/heads/master"},"4dd819d3811ff12354001b054cdaf3471d1915fe":{"kind":"REWORK","_number":2,"created":"2025-09-19 07:38:21.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/98/961598/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/98/961598/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/98/961598/2"}}},"commit":{"parents":[{"commit":"0f8e416775c54ce85dcd15b6dc18895cc7600214","subject":"Merge \"Show inactive project status in project.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/0f8e416775c54ce85dcd15b6dc18895cc7600214"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-18 07:27:42.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-19 07:38:11.000000000","tz":120},"subject":"Add TC/PTL results from 2026.1 election","message":"Add TC/PTL results from 2026.1 election\n\nAlso clear out existing TC roles and prepare for nominations.\n\nDepends-On: https://review.opendev.org/c/openstack/election/+/961596\n\nChange-Id: I110dc88b2032971049e4676a5fc3b112cf8e916f\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/4dd819d3811ff12354001b054cdaf3471d1915fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/4dd819d3811ff12354001b054cdaf3471d1915fe"}]},"branch":"refs/heads/master"},"c0df98c709a85cb83ccc794744b05a18a46139f1":{"kind":"REWORK","_number":3,"created":"2025-09-22 07:43:28.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/98/961598/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/98/961598/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/98/961598/3"}}},"commit":{"parents":[{"commit":"0f8e416775c54ce85dcd15b6dc18895cc7600214","subject":"Merge \"Show inactive project status in project.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/0f8e416775c54ce85dcd15b6dc18895cc7600214"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-18 07:27:42.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-22 07:41:35.000000000","tz":120},"subject":"Add TC/PTL results from 2026.1 election","message":"Add TC/PTL results from 2026.1 election\n\nAlso clear out existing TC roles and prepare for nominations.\n\nDepends-On: https://review.opendev.org/c/openstack/election/+/961596\n\nChange-Id: I110dc88b2032971049e4676a5fc3b112cf8e916f\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c0df98c709a85cb83ccc794744b05a18a46139f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c0df98c709a85cb83ccc794744b05a18a46139f1"}]},"branch":"refs/heads/master"},"85992683654a73995b9faad9cec5671582272427":{"kind":"REWORK","_number":4,"created":"2025-09-22 08:24:13.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/98/961598/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/98/961598/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/98/961598/4"}}},"commit":{"parents":[{"commit":"0f8e416775c54ce85dcd15b6dc18895cc7600214","subject":"Merge \"Show inactive project status in project.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/0f8e416775c54ce85dcd15b6dc18895cc7600214"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-18 07:27:42.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-22 08:24:02.000000000","tz":120},"subject":"Add TC/PTL results from 2026.1 election","message":"Add TC/PTL results from 2026.1 election\n\nAlso clear out existing TC roles and prepare for nominations.\n\nDepends-On: https://review.opendev.org/c/openstack/election/+/961596\n\nChange-Id: I110dc88b2032971049e4676a5fc3b112cf8e916f\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/85992683654a73995b9faad9cec5671582272427"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/85992683654a73995b9faad9cec5671582272427"}]},"branch":"refs/heads/master"},"c8962e0c0fc84c7183da92d3e2479a52623c419f":{"kind":"REWORK","_number":5,"created":"2025-09-23 07:52:38.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/98/961598/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/98/961598/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/98/961598/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/98/961598/5"}}},"commit":{"parents":[{"commit":"0f8e416775c54ce85dcd15b6dc18895cc7600214","subject":"Merge \"Show inactive project status in project.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/0f8e416775c54ce85dcd15b6dc18895cc7600214"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-18 07:27:42.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-09-23 07:52:38.000000000","tz":0},"subject":"Add TC/PTL results from 2026.1 election","message":"Add TC/PTL results from 2026.1 election\n\nAlso clear out existing TC roles and prepare for nominations.\n\nDepends-On: https://review.opendev.org/c/openstack/election/+/961596\n\nChange-Id: I110dc88b2032971049e4676a5fc3b112cf8e916f\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c8962e0c0fc84c7183da92d3e2479a52623c419f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/c8962e0c0fc84c7183da92d3e2479a52623c419f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Rollcall-Vote","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"An informational vote on this revision","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Rollcall-Vote","description":"An informational vote from team members","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
