)]}'
{"id":"openstack%2Fgrenade~241018","triplet_id":"openstack%2Fgrenade~master~Ic9fa1190f098bbe375146d960352314f086e4296","project":"openstack/grenade","branch":"master","hashtags":[],"change_id":"Ic9fa1190f098bbe375146d960352314f086e4296","subject":"Support TEMPEST_REGEX environment variable","status":"ABANDONED","created":"2015-11-02 23:09:40.000000000","updated":"2016-06-10 00:24:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"547610a16a4e24870e8a5212641d2607bdb5da7c","_number":241018,"virtual_id_number":241018,"owner":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":1,"date":"2016-02-22 17:24:58.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-11-03 15:36:00.000000000","updated_by":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"reviewer":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"state":"REVIEWER"},{"updated":"2015-11-03 15:36:13.000000000","updated_by":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"reviewer":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"state":"REVIEWER"},{"updated":"2015-11-04 20:15:09.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2016-02-22 17:24:58.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"e1831b0695d611373bf0c7b33ac08e013d505f96","author":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"date":"2015-11-02 23:09:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8f0b0dad7eb6159ba7f326b4aac7bb7f19298ec","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-02 23:59:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-grenade-dsvm http://logs.openstack.org/18/241018/1/check/gate-grenade-dsvm/726bc3b/ : SUCCESS in 49m 55s\n- gate-grenade-dsvm-neutron http://logs.openstack.org/18/241018/1/check/gate-grenade-dsvm-neutron/7ef88ce/ : SUCCESS in 49m 54s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/18/241018/1/check/gate-grenade-dsvm-partial-ncpu/d087636/ : SUCCESS in 44m 04s\n- gate-grenade-dsvm-heat-nv http://logs.openstack.org/18/241018/1/check/gate-grenade-dsvm-heat-nv/b22b4bb/ : SUCCESS in 48m 35s (non-voting)\n- gate-grenade-dsvm-multinode-nv http://logs.openstack.org/18/241018/1/check/gate-grenade-dsvm-multinode-nv/322a2f2/ : SUCCESS in 49m 40s (non-voting)\n- gate-grenade-publish-docs http://docs-draft.openstack.org/18/241018/1/check/gate-grenade-publish-docs/f4a52ab//doc/build/html/ : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1},{"id":"6142ae98171a9a80d5fdabcf1e067ceacf6b6afd","author":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"date":"2015-11-03 00:36:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d721b7de770abdd6f2116363f879209d2c1e0f79","author":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"date":"2015-11-03 00:45:29.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"6fe310bcdec9460797dd9c9c8c52bad47ee9c51e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-03 01:34:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-grenade-dsvm http://logs.openstack.org/18/241018/2/check/gate-grenade-dsvm/126e7f8/ : SUCCESS in 44m 10s\n- gate-grenade-dsvm-neutron http://logs.openstack.org/18/241018/2/check/gate-grenade-dsvm-neutron/16294c8/ : SUCCESS in 44m 56s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/18/241018/2/check/gate-grenade-dsvm-partial-ncpu/0a72679/ : SUCCESS in 43m 34s\n- gate-grenade-dsvm-heat-nv http://logs.openstack.org/18/241018/2/check/gate-grenade-dsvm-heat-nv/50c5665/ : SUCCESS in 58m 16s (non-voting)\n- gate-grenade-dsvm-multinode-nv http://logs.openstack.org/18/241018/2/check/gate-grenade-dsvm-multinode-nv/8139dd5/ : SUCCESS in 51m 38s (non-voting)\n- gate-grenade-publish-docs http://docs-draft.openstack.org/18/241018/2/check/gate-grenade-publish-docs/060b0b1//doc/build/html/ : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2},{"id":"96139353691d33f20cef5c952454426b4aa4446b","author":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"date":"2015-11-03 15:34:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9c21ef6322d949bb405a235b92693e9d16a98a03","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-03 16:46:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-grenade-dsvm http://logs.openstack.org/18/241018/3/check/gate-grenade-dsvm/9075b81/ : SUCCESS in 52m 38s\n- gate-grenade-dsvm-neutron http://logs.openstack.org/18/241018/3/check/gate-grenade-dsvm-neutron/097f64e/ : SUCCESS in 1h 03m 29s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/18/241018/3/check/gate-grenade-dsvm-partial-ncpu/8d2a4a1/ : SUCCESS in 57m 10s\n- gate-grenade-dsvm-heat-nv http://logs.openstack.org/18/241018/3/check/gate-grenade-dsvm-heat-nv/eae45ac/ : SUCCESS in 1h 03m 32s (non-voting)\n- gate-grenade-dsvm-multinode-nv http://logs.openstack.org/18/241018/3/check/gate-grenade-dsvm-multinode-nv/028e27c/ : SUCCESS in 49m 57s (non-voting)\n- gate-grenade-publish-docs http://docs-draft.openstack.org/18/241018/3/check/gate-grenade-publish-docs/98bf320//doc/build/html/ : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3},{"id":"d01dae7f79f210c75036812b229d43be0cc9c522","author":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"date":"2015-11-03 16:54:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c61b1c885faf9d1e04736239a29087f545aa8840","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2015-11-04 20:15:09.000000000","message":"Patch Set 3: Code-Review-1\n\nI don\u0027t really want to bring the REGEX logic into grenade. smoke is very small, and should be tunable via feature flags.","accounts_in_message":[],"_revision_number":3},{"id":"ab6630bb204bc3e61befbc259dcb618994066046","author":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"date":"2015-11-04 23:45:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"2423d876bc742f3ed49cfa2d3cf30e8d50eff878","author":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"date":"2016-02-22 16:09:02.000000000","message":"Restored\n\nDiscussion during the Ironic mid-cycle wants us to go this route for now.","accounts_in_message":[],"_revision_number":3},{"id":"5facfa7a50c27d4e1a37cc7813b3bed8ec7643af","author":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"date":"2016-02-22 16:12:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9db0615e4ed2016051935b2c95dea9faaa8e57a6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-22 17:24:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-grenade-dsvm http://logs.openstack.org/18/241018/4/check/gate-grenade-dsvm/69d040b/ : SUCCESS in 33m 52s\n- gate-grenade-dsvm-neutron http://logs.openstack.org/18/241018/4/check/gate-grenade-dsvm-neutron/b00f92a/ : SUCCESS in 40m 27s\n- gate-grenade-dsvm-heat-nv http://logs.openstack.org/18/241018/4/check/gate-grenade-dsvm-heat-nv/513c4e5/ : SUCCESS in 41m 03s (non-voting)\n- gate-grenade-dsvm-multinode http://logs.openstack.org/18/241018/4/check/gate-grenade-dsvm-multinode/1811b5c/ : SUCCESS in 53m 48s\n- gate-grenade-publish-docs http://docs-draft.openstack.org/18/241018/4/check/gate-grenade-publish-docs/6ece138//doc/build/html/ : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4},{"id":"547610a16a4e24870e8a5212641d2607bdb5da7c","author":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"date":"2016-06-10 00:24:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"602387357ccac3f4d8ac2bc24f978f70da571257","revisions":{"44c6f4fb0a49ad4d9f1a8950dd3f7d0ed373b0ab":{"kind":"REWORK","_number":1,"created":"2015-11-02 23:09:40.000000000","uploader":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"ref":"refs/changes/18/241018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/18/241018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/18/241018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/18/241018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/18/241018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/18/241018/1"}}},"commit":{"parents":[{"commit":"19de9d8b09da7fa034eb4d7aa01c91087ae229b8","subject":"always set upgrade_level to auto","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/19de9d8b09da7fa034eb4d7aa01c91087ae229b8"}]}],"author":{"name":"John L. Villalovos","email":"john.l.villalovos@intel.com","date":"2015-11-02 23:08:21.000000000","tz":-480},"committer":{"name":"John L. Villalovos","email":"john.l.villalovos@intel.com","date":"2015-11-02 23:08:21.000000000","tz":-480},"subject":"Honor DEVSTACK_GATE_TEMPEST_REGEX","message":"Honor DEVSTACK_GATE_TEMPEST_REGEX\n\nHonor the DEVSTACK_GATE_TEMPEST_REGEX environment variable and run\ntempest the same way as devstack-gate does when the environment variable\nexists.\n\nChange-Id: Ic9fa1190f098bbe375146d960352314f086e4296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/44c6f4fb0a49ad4d9f1a8950dd3f7d0ed373b0ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/44c6f4fb0a49ad4d9f1a8950dd3f7d0ed373b0ab"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"19de9d8b09da7fa034eb4d7aa01c91087ae229b8","is_merged_in_target_branch":true,"change_id":"I5c899779866ebdf492d474a625eb396273bb0259","change_number":240045,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e7a9aa995d6dec36eb50a18a187da30f3a6a5888":{"kind":"REWORK","_number":2,"created":"2015-11-03 00:36:23.000000000","uploader":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"ref":"refs/changes/18/241018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/18/241018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/18/241018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/18/241018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/18/241018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/18/241018/2"}}},"commit":{"parents":[{"commit":"19de9d8b09da7fa034eb4d7aa01c91087ae229b8","subject":"always set upgrade_level to auto","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/19de9d8b09da7fa034eb4d7aa01c91087ae229b8"}]}],"author":{"name":"John L. Villalovos","email":"john.l.villalovos@intel.com","date":"2015-11-02 23:08:21.000000000","tz":-480},"committer":{"name":"John L. Villalovos","email":"john.l.villalovos@intel.com","date":"2015-11-03 00:35:17.000000000","tz":-480},"subject":"Support TEMPEST_REGEX environment variable","message":"Support TEMPEST_REGEX environment variable\n\nSupport the TEMPEST_REGEX environment variable and run tempest the same\nway as devstack-gate does when the environment variable\nDEVSTACK_GATE_TEMPEST_REGEX exists.\n\nChange-Id: Ic9fa1190f098bbe375146d960352314f086e4296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/e7a9aa995d6dec36eb50a18a187da30f3a6a5888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/e7a9aa995d6dec36eb50a18a187da30f3a6a5888"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"19de9d8b09da7fa034eb4d7aa01c91087ae229b8","is_merged_in_target_branch":true,"change_id":"I5c899779866ebdf492d474a625eb396273bb0259","change_number":240045,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"13fa7fff67bf9e7a02c7482140c18d5596264ff9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2015-11-03 15:34:47.000000000","uploader":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"ref":"refs/changes/18/241018/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/18/241018/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/18/241018/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/18/241018/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/18/241018/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/18/241018/3"}}},"commit":{"parents":[{"commit":"19de9d8b09da7fa034eb4d7aa01c91087ae229b8","subject":"always set upgrade_level to auto","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/19de9d8b09da7fa034eb4d7aa01c91087ae229b8"}]}],"author":{"name":"John L. Villalovos","email":"john.l.villalovos@intel.com","date":"2015-11-02 23:08:21.000000000","tz":-480},"committer":{"name":"John L. Villalovos","email":"john.l.villalovos@intel.com","date":"2015-11-03 15:34:11.000000000","tz":-480},"subject":"Support TEMPEST_REGEX environment variable","message":"Support TEMPEST_REGEX environment variable\n\nSupport the TEMPEST_REGEX environment variable and run tempest the same\nway as devstack-gate would when the environment variable\nDEVSTACK_GATE_TEMPEST_REGEX exists.\n\nThe Ironic grenade job is failing. The Ironic grenade job does not run\nthe same tests as the normal tempest jobs, as the normal Ironic tempest\njobs set DEVSTACK_GATE_TEMPEST_REGEX\u003dbaremetal.\n\nChange-Id: Ic9fa1190f098bbe375146d960352314f086e4296\nPartial-Bug: #1512544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/13fa7fff67bf9e7a02c7482140c18d5596264ff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/13fa7fff67bf9e7a02c7482140c18d5596264ff9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"19de9d8b09da7fa034eb4d7aa01c91087ae229b8","is_merged_in_target_branch":true,"change_id":"I5c899779866ebdf492d474a625eb396273bb0259","change_number":240045,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"602387357ccac3f4d8ac2bc24f978f70da571257":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2016-02-22 16:12:33.000000000","uploader":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"ref":"refs/changes/18/241018/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/18/241018/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/18/241018/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/18/241018/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/18/241018/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/18/241018/4"}}},"commit":{"parents":[{"commit":"22ee7f82df8e3aa4ea97517d3c0083e41bf529e5","subject":"Merge \"Destroy Nova resource before Neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/22ee7f82df8e3aa4ea97517d3c0083e41bf529e5"}]}],"author":{"name":"John L. Villalovos","email":"john.l.villalovos@intel.com","date":"2015-11-02 23:08:21.000000000","tz":-480},"committer":{"name":"John L. Villalovos","email":"john.l.villalovos@intel.com","date":"2016-02-22 16:10:54.000000000","tz":-480},"subject":"Support TEMPEST_REGEX environment variable","message":"Support TEMPEST_REGEX environment variable\n\nSupport the TEMPEST_REGEX environment variable and run tempest the same\nway as devstack-gate would when the environment variable\nDEVSTACK_GATE_TEMPEST_REGEX exists.\n\nThe Ironic grenade job is failing. The Ironic grenade job does not run\nthe same tests as the normal tempest jobs, as the normal Ironic tempest\njobs set DEVSTACK_GATE_TEMPEST_REGEX\u003dbaremetal.\n\nTrying to make Ironic pass the Tempest smoke tests is a longer term\ngoal. For example Tempest smoke tests assume that there are three VMs\navailable but Ironic only runs with one as each VM consumes 1GB of\nmemory.\n\nChange-Id: Ic9fa1190f098bbe375146d960352314f086e4296\nPartial-Bug: #1512544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/602387357ccac3f4d8ac2bc24f978f70da571257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/602387357ccac3f4d8ac2bc24f978f70da571257"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"22ee7f82df8e3aa4ea97517d3c0083e41bf529e5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
