)]}'
{"id":"openstack%2Fgrenade~619728","triplet_id":"openstack%2Fgrenade~master~If23619885a4766e243b39eae6041efe4a4d0a3af","project":"openstack/grenade","branch":"master","topic":"cd/swift-openstackclient","hashtags":[],"change_id":"If23619885a4766e243b39eae6041efe4a4d0a3af","subject":"Re-install python-openstackclient after swift","status":"MERGED","created":"2018-11-23 12:57:51.000000000","updated":"2018-11-26 23:04:17.000000000","submitted":"2018-11-26 23:04:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"619728-1543273457132-bd3c5299","meta_rev_id":"b859f89daa128df8bad871d8bf17b869a0ad7c9d","_number":619728,"virtual_id_number":619728,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2018-11-26 23:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":2,"date":"2018-11-26 15:17:40.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2018-11-26 15:39:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2018-11-26 15:39:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-26 15:17:40.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-11-26 15:39:10.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2018-11-26 23:04:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67b51ed33f9da017ef8803fa0457ed7c84f2d20c","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-11-23 12:57:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ae7c22d11203b47746b920bb3275eab4fd033c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-23 15:53:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-grenade http://logs.openstack.org/28/619728/1/check/neutron-grenade/4f6bc95/ : SUCCESS in 1h 39m 22s\n- grenade-heat http://logs.openstack.org/28/619728/1/check/grenade-heat/aa0bde3/ : SUCCESS in 1h 19m 51s (non-voting)\n- grenade-py3 http://logs.openstack.org/28/619728/1/check/grenade-py3/c21016d/ : SUCCESS in 1h 03m 58s\n- neutron-grenade-multinode http://logs.openstack.org/28/619728/1/check/neutron-grenade-multinode/8d6081d/ : SUCCESS in 1h 50m 46s\n- tox-docs http://logs.openstack.org/28/619728/1/check/tox-docs/fd43b15/html/ : SUCCESS in 7m 57s\n- ironic-grenade-dsvm http://logs.openstack.org/28/619728/1/check/ironic-grenade-dsvm/30bb5df/ : SUCCESS in 2h 46m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"086a96cf83fffe99fac792049cfd31759a431b8d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-11-26 14:30:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nSoft -1 for the comment in the commit message - I think I\u0027m piecing together why this is a problem from IRC and comments on the patch itself, but that should probably either be in the commit message or a linked grenade bug.","accounts_in_message":[],"_revision_number":1},{"id":"01aff2b01e4e3c6f71e1fef1848c3629fe45286d","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-11-26 14:51:38.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c93573e88f0c3d3eb6b3255d348ae0a468324ace","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-11-26 15:00:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58b3ec0477aa9ff5bfbba0dd4269c83ac53d2cba","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-11-26 15:00:52.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"36339c5f0998949b9d318b95204367c6b21a6a79","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-11-26 15:04:37.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2539a604a6b7ce75b7bb9cabd6638a7950daef8d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-11-26 15:17:40.000000000","message":"Patch Set 3: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":3},{"id":"de6f8487ffebeb210e9d29c97b3fac1a1b992337","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-11-26 15:39:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nYeah, this sucks, but I don\u0027t have any better ideas.","accounts_in_message":[],"_revision_number":3},{"id":"704ede5891f797ae87e5ad4bf96035a5c3b49a45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-26 18:52:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-grenade http://logs.openstack.org/28/619728/3/check/neutron-grenade/ced36bb/ : SUCCESS in 48m 24s\n- grenade-heat http://logs.openstack.org/28/619728/3/check/grenade-heat/ac3dfa0/ : SUCCESS in 1h 16m 49s (non-voting)\n- grenade-py3 http://logs.openstack.org/28/619728/3/check/grenade-py3/47b8617/ : SUCCESS in 53m 11s\n- neutron-grenade-multinode http://logs.openstack.org/28/619728/3/check/neutron-grenade-multinode/052e21a/ : SUCCESS in 57m 21s\n- tox-docs http://logs.openstack.org/28/619728/3/check/tox-docs/bc4dc13/html/ : SUCCESS in 4m 51s\n- ironic-grenade-dsvm http://logs.openstack.org/28/619728/3/check/ironic-grenade-dsvm/61c6527/ : SUCCESS in 2h 21m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f54c895411f61ef4e4cc99443bd1c7d028f38ced","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-26 21:30:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7d3cd89e23703dea4bdc9c52753d3856297d28b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-26 23:04:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/28/619728/3/gate/grenade-py3/58cc120/ : SUCCESS in 55m 50s\n- neutron-grenade http://logs.openstack.org/28/619728/3/gate/neutron-grenade/7fe6b73/ : SUCCESS in 49m 44s\n- neutron-grenade-multinode http://logs.openstack.org/28/619728/3/gate/neutron-grenade-multinode/b64d890/ : SUCCESS in 54m 46s\n- tox-docs http://logs.openstack.org/28/619728/3/gate/tox-docs/4e870df/html/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":3},{"id":"a9b9630011dc68ad7c0168a75aa53fe09f1b49ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-26 23:04:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"674861b787a9861a5585101563a971bf3e571812","revisions":{"2c79a327754533d79570329cdc6c7d7d429c7c1b":{"kind":"REWORK","_number":1,"created":"2018-11-23 12:57:51.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/28/619728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/28/619728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/28/619728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/28/619728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/28/619728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/28/619728/1"}}},"commit":{"parents":[{"commit":"7bae489f38f8f0c82c8eb284d1841ef68d8e9a43","subject":"Add grenade-py3 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/7bae489f38f8f0c82c8eb284d1841ef68d8e9a43"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2018-11-23 12:51:35.000000000","tz":0},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2018-11-23 12:51:35.000000000","tz":0},"subject":"Re-install python-openstackclient after swift","message":"Re-install python-openstackclient after swift\n\nSwift includes openstack client in its requirements so when it installs\nunder Python 2 it clobbers the existing openstack client at\n/usr/local/bin and any previously installed plugins are wiped out\nbecause their entry points aren\u0027t seen.\n\nTo address this, we do a reinstall at the end of swift\u0027s upgrade.\nBecause python-openstack client is in ENABLED_PYTHON3_PACKAGES,\nit will install as Python3 when installed standalone.\n\nChange-Id: If23619885a4766e243b39eae6041efe4a4d0a3af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/2c79a327754533d79570329cdc6c7d7d429c7c1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/2c79a327754533d79570329cdc6c7d7d429c7c1b"}]},"branch":"refs/heads/master"},"e008a306f470c6d64538295f9a6402c4b30a32fd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-11-26 14:51:38.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/28/619728/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/28/619728/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/28/619728/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/28/619728/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/28/619728/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/28/619728/2"}}},"commit":{"parents":[{"commit":"7bae489f38f8f0c82c8eb284d1841ef68d8e9a43","subject":"Add grenade-py3 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/7bae489f38f8f0c82c8eb284d1841ef68d8e9a43"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2018-11-23 12:51:35.000000000","tz":0},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2018-11-26 14:47:42.000000000","tz":0},"subject":"Re-install python-openstackclient after swift","message":"Re-install python-openstackclient after swift\n\nWhen running grenade under python3, swift runs under python3.  Swift\nincludes openstack client in its requirements so when it installs under\nPython 2 it clobbers the existing openstack client at /usr/local/bin and\nany previously installed plugins are wiped out because their entry\npoints aren\u0027t seen. This was discovered because the osc-placement\ncommands were not available to nova inventory verification commands\nduring the nova upgrade, which comes after swift.\n\nTo address this, we do a reinstall at the end of swift\u0027s upgrade.\nBecause python-openstack client is in ENABLED_PYTHON3_PACKAGES,\nit will install as Python3 when installed standalone.\n\nNote that the issue being fixed here is python3 specific, not\nOpenStack release specific.\n\nCloses-Bug: 1805156\nChange-Id: If23619885a4766e243b39eae6041efe4a4d0a3af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/e008a306f470c6d64538295f9a6402c4b30a32fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/e008a306f470c6d64538295f9a6402c4b30a32fd"}]},"branch":"refs/heads/master"},"674861b787a9861a5585101563a971bf3e571812":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2018-11-26 15:04:37.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/28/619728/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/28/619728/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/28/619728/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/28/619728/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/28/619728/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/28/619728/3"}}},"commit":{"parents":[{"commit":"7bae489f38f8f0c82c8eb284d1841ef68d8e9a43","subject":"Add grenade-py3 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/7bae489f38f8f0c82c8eb284d1841ef68d8e9a43"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2018-11-23 12:51:35.000000000","tz":0},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2018-11-26 15:04:30.000000000","tz":0},"subject":"Re-install python-openstackclient after swift","message":"Re-install python-openstackclient after swift\n\nWhen running grenade under python3, swift runs under python2.  Swift\nincludes openstack client in its requirements so when it installs under\nPython 2 it clobbers the existing openstack client at /usr/local/bin and\nany previously installed plugins are wiped out because their entry\npoints aren\u0027t seen. This was discovered because the osc-placement\ncommands were not available to nova inventory verification commands\nduring the nova upgrade, which comes after swift.\n\nTo address this, we do a reinstall at the end of swift\u0027s upgrade.\nBecause python-openstack client is in ENABLED_PYTHON3_PACKAGES,\nit will install as Python3 when installed standalone.\n\nNote that the issue being fixed here is python3 specific, not\nOpenStack release specific.\n\nCloses-Bug: 1805156\nChange-Id: If23619885a4766e243b39eae6041efe4a4d0a3af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/674861b787a9861a5585101563a971bf3e571812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/674861b787a9861a5585101563a971bf3e571812"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
