)]}'
{"id":"openstack%2Fgrenade~639774","triplet_id":"openstack%2Fgrenade~master~I14427b087b224f66f65c379f72f2c0e5b9ba939a","project":"openstack/grenade","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I14427b087b224f66f65c379f72f2c0e5b9ba939a","subject":"DNM - grenade test jobs (heat)","status":"ABANDONED","created":"2019-02-27 18:58:22.000000000","updated":"2020-05-13 07:49:25.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b3d1747bf4c9b061ec34346162e00ce4f83b7953","_number":639774,"virtual_id_number":639774,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-26 10:59:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-28 17:23:49.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-02-12 22:32:48.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-04-24 01:49:00.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-04-25 15:16:58.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-26 10:59:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fea17a4be38ef39e32e94282241e029d76cf4ef9","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-02-27 18:58:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7db711d3ee2b487904471b34ed8139cdfb3b8ae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 23:40:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/1/check/grenade/779884f/ : SUCCESS in 53m 45s\n- grenade-heat-test http://logs.openstack.org/74/639774/1/check/grenade-heat-test/433745d/ : SUCCESS in 1h 14m 55s\n- grenade-py3 http://logs.openstack.org/74/639774/1/check/grenade-py3/69651c8/ : TIMED_OUT in 2h 03m 44s\n- grenade-postgresql http://logs.openstack.org/74/639774/1/check/grenade-postgresql/fe2d2ba/ : SUCCESS in 51m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4ef2ec0f21a59b55f88aaa8d3529426daa4ddbb","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-02-28 11:56:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2b16e7e68dc92cb22c677b59e4e1dfb15a18f666","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-28 13:12:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade http://logs.openstack.org/74/639774/2/check/grenade/afb5646/ : SUCCESS in 50m 05s\n- grenade-heat-test http://logs.openstack.org/74/639774/2/check/grenade-heat-test/9d4f68f/ : SUCCESS in 1h 09m 55s\n- grenade-py3 http://logs.openstack.org/74/639774/2/check/grenade-py3/2817f0e/ : SUCCESS in 51m 01s\n- grenade-postgresql http://logs.openstack.org/74/639774/2/check/grenade-postgresql/5928ed3/ : SUCCESS in 51m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f3df94aa6f747eb2e523ad1241328c3655ae464","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-02-28 13:53:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"08770f8a60f43a09b6e44887782b9f864e805a74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-28 15:06:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- grenade http://logs.openstack.org/74/639774/3/check/grenade/039acf2/ : SUCCESS in 47m 05s\n- grenade-heat-test http://logs.openstack.org/74/639774/3/check/grenade-heat-test/62a8029/ : SUCCESS in 1h 10m 06s\n- grenade-py3 http://logs.openstack.org/74/639774/3/check/grenade-py3/3c2aa4b/ : SUCCESS in 52m 45s\n- grenade-postgresql http://logs.openstack.org/74/639774/3/check/grenade-postgresql/0238560/ : SUCCESS in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"45ffb4d570cabc1b59b6449b089c9321987eb8c6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-04 11:46:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e821a42321b2c3d8791e1ed4e5ffec8636ed968f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-04 13:10:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/4/check/grenade/c622436/ : SUCCESS in 52m 52s\n- grenade-heat-test http://logs.openstack.org/74/639774/4/check/grenade-heat-test/4c6d98c/ : FAILURE in 1h 18m 24s\n- grenade-py3 http://logs.openstack.org/74/639774/4/check/grenade-py3/d1ebd3c/ : SUCCESS in 57m 11s\n- grenade-postgresql http://logs.openstack.org/74/639774/4/check/grenade-postgresql/46f6cb3/ : SUCCESS in 47m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1a26454bf319edd4e3782a685bcef03e02721f92","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-04 13:27:38.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"798242e0a1e9d2a7c4eac1203d2ca2275a1a5601","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-04 14:51:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/5/check/grenade/5552f8a/ : SUCCESS in 51m 31s\n- grenade-heat-test http://logs.openstack.org/74/639774/5/check/grenade-heat-test/00b4915/ : FAILURE in 1h 17m 59s\n- grenade-py3 http://logs.openstack.org/74/639774/5/check/grenade-py3/1e70d14/ : SUCCESS in 45m 54s\n- grenade-postgresql http://logs.openstack.org/74/639774/5/check/grenade-postgresql/675244a/ : SUCCESS in 54m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f2fd1ea27606c475269143d2a1eb862c07cae7da","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-04 19:09:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"57b3afd89ca00ec53c912bc4e67ffb3b55a38c8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-04 23:46:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/6/check/grenade/4d8c1cc/ : SUCCESS in 56m 32s\n- grenade-heat-test http://logs.openstack.org/74/639774/6/check/grenade-heat-test/51ef53a/ : FAILURE in 1h 29m 40s\n- grenade-py3 http://logs.openstack.org/74/639774/6/check/grenade-py3/08b76c5/ : SUCCESS in 49m 51s\n- grenade-postgresql http://logs.openstack.org/74/639774/6/check/grenade-postgresql/fb45758/ : SUCCESS in 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"49a16d4907c0ba0c1054ff1fe3001b5809603e76","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-06 00:50:55.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"70f5086df01c9d287e71ed75ddc6ce7885b0440f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 03:33:03.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/7/check/grenade/a35e4ea/ : SUCCESS in 45m 44s\n- grenade-heat-test http://logs.openstack.org/74/639774/7/check/grenade-heat-test/444fb6f/ : FAILURE in 1h 28m 31s\n- grenade-py3 http://logs.openstack.org/74/639774/7/check/grenade-py3/31c24bb/ : SUCCESS in 47m 02s\n- grenade-postgresql http://logs.openstack.org/74/639774/7/check/grenade-postgresql/4c45848/ : SUCCESS in 46m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"19c7bf03699349919c8bd7c3d41e3d343ea3da47","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-06 11:59:38.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"dac2445389bfc788c8c444966b5d6df0aed41da5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 13:21:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade http://logs.openstack.org/74/639774/8/check/grenade/f480135/ : SUCCESS in 1h 06m 19s\n- grenade-heat-test http://logs.openstack.org/74/639774/8/check/grenade-heat-test/a7320d4/ : SUCCESS in 1h 16m 28s\n- grenade-py3 http://logs.openstack.org/74/639774/8/check/grenade-py3/09c9754/ : SUCCESS in 51m 45s\n- grenade-postgresql http://logs.openstack.org/74/639774/8/check/grenade-postgresql/7ba5e68/ : SUCCESS in 49m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"93ef6fc11991ef374a73d8cb16368f3a6e902035","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-12 21:46:18.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dcd84dc389d444e726f3a6c267d67c944f54c697","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 01:21:57.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/9/check/grenade/57092dd/ : SUCCESS in 45m 38s\n- grenade-heat-test http://logs.openstack.org/74/639774/9/check/grenade-heat-test/f25a4ba/ : FAILURE in 1h 41m 51s\n- grenade-py3 http://logs.openstack.org/74/639774/9/check/grenade-py3/bf81f0d/ : SUCCESS in 56m 14s\n- grenade-postgresql http://logs.openstack.org/74/639774/9/check/grenade-postgresql/f2c002f/ : SUCCESS in 47m 31s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ad2ac3d50652e24ecf8d1392c0150ba2702722dc","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-21 10:54:39.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"fba5d3c82b1b0bd3b11ab815ad7b6884025ac0d3","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-21 12:40:15.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"0c6e3c27a34a694514cd7d9a9e49df78c78a8e9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 14:01:12.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade http://logs.openstack.org/74/639774/11/check/grenade/6d92637/ : SUCCESS in 52m 47s\n- grenade-heat-test http://logs.openstack.org/74/639774/11/check/grenade-heat-test/05da0f8/ : SUCCESS in 1h 15m 20s\n- grenade-py3 http://logs.openstack.org/74/639774/11/check/grenade-py3/3cd39dc/ : SUCCESS in 57m 42s\n- grenade-postgresql http://logs.openstack.org/74/639774/11/check/grenade-postgresql/1b98001/ : SUCCESS in 51m 32s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ce2eeccb5d1ebbf7f9a1f9e98e106beb3b3f44d3","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-21 17:13:35.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"4eb974362e660d1ddd58be2f9ed6f1a8763881d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 22:03:48.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade http://logs.openstack.org/74/639774/12/check/grenade/5c70aff/ : SUCCESS in 51m 41s\n- grenade-heat-test http://logs.openstack.org/74/639774/12/check/grenade-heat-test/ffa86af/ : SUCCESS in 1h 19m 49s\n- grenade-py3 http://logs.openstack.org/74/639774/12/check/grenade-py3/d8c4c04/ : SUCCESS in 59m 47s\n- grenade-postgresql http://logs.openstack.org/74/639774/12/check/grenade-postgresql/b6fbeaa/ : SUCCESS in 51m 43s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"01a20818c1bc593a201cff2b9bc8624cc99acaf8","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-22 10:42:50.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"6172e1ddce3938a84a6f1fc72db5f1e649fe4490","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 11:53:24.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/12/check/grenade/f6379bf/ : POST_FAILURE in 45m 07s\n- grenade-heat-test http://logs.openstack.org/74/639774/12/check/grenade-heat-test/f73b701/ : POST_FAILURE in 1h 08m 59s\n- grenade-py3 http://logs.openstack.org/74/639774/12/check/grenade-py3/8f7ae31/ : POST_FAILURE in 49m 42s\n- grenade-postgresql http://logs.openstack.org/74/639774/12/check/grenade-postgresql/8283cec/ : POST_FAILURE in 46m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"549c3df7a7320981e383b5f5bd203c78d1a47130","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-22 12:19:10.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"701ec6c7077c8773b972e537d82a9f1e28a5b6c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 13:08:29.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"bf758ddf69ecd6843688c80e196abab4031fa5ff","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-22 13:08:52.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"62a4fede79756cfef416bf3ba3fb4f7869eced18","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-22 13:27:06.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"99f8c301cde271f22ac564594ed08eeb222d1ca5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 13:27:08.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"7de9333b81b8037fa504cdd30af0dc34cf3a602e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 14:41:26.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade http://logs.openstack.org/74/639774/12/check/grenade/2c162cb/ : SUCCESS in 47m 39s\n- grenade-heat-test http://logs.openstack.org/74/639774/12/check/grenade-heat-test/0933c03/ : SUCCESS in 1h 10m 51s\n- grenade-py3 http://logs.openstack.org/74/639774/12/check/grenade-py3/3835002/ : SUCCESS in 1h 00m 51s\n- grenade-postgresql http://logs.openstack.org/74/639774/12/check/grenade-postgresql/6870859/ : SUCCESS in 56m 48s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3847c01fc748edcf2f69f9a243dbab00bd433dda","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-26 19:01:07.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"71bce6b6189333058f60c84ba9369fbcf0a632a6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-26 21:03:16.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"e17402cd049d63216725c5018ebac5722186e4a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-27 00:38:38.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade http://logs.openstack.org/74/639774/14/check/grenade/ed7a221/ : SUCCESS in 50m 46s\n- grenade-heat-test http://logs.openstack.org/74/639774/14/check/grenade-heat-test/ba5eba7/ : SUCCESS in 1h 15m 25s\n- grenade-py3 http://logs.openstack.org/74/639774/14/check/grenade-py3/66464bf/ : SUCCESS in 1h 07m 12s\n- grenade-postgresql http://logs.openstack.org/74/639774/14/check/grenade-postgresql/c2657c4/ : SUCCESS in 1h 01m 01s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"7c9e1efdc4b0f2ccfb60ba174d1a176768dff08e","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-27 00:50:44.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"922f6b8f2a2c79aa6ea7bab988a1b3cd3591425f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-27 02:04:43.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade http://logs.openstack.org/74/639774/15/check/grenade/8dd656b/ : SUCCESS in 46m 25s\n- grenade-heat-test http://logs.openstack.org/74/639774/15/check/grenade-heat-test/b1ead60/ : SUCCESS in 1h 07m 46s\n- grenade-py3 http://logs.openstack.org/74/639774/15/check/grenade-py3/0131b73/ : SUCCESS in 59m 40s\n- grenade-postgresql http://logs.openstack.org/74/639774/15/check/grenade-postgresql/45b8bcc/ : SUCCESS in 52m 29s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"7c0a13dd0eae62d61508d9b4c5f6a2dc634fa812","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-27 18:04:09.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"67b18f96f56c7a2ac67dab197fcf2753cac714d9","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-27 19:42:32.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"16dbce419968b2424c504939b19008e36b90d4c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-27 21:31:43.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade http://logs.openstack.org/74/639774/17/check/grenade/020b09d/ : SUCCESS in 55m 30s\n- grenade-heat-test http://logs.openstack.org/74/639774/17/check/grenade-heat-test/f8469fd/ : SUCCESS in 1h 10m 22s\n- grenade-py3 http://logs.openstack.org/74/639774/17/check/grenade-py3/005d58d/ : SUCCESS in 49m 08s\n- grenade-postgresql http://logs.openstack.org/74/639774/17/check/grenade-postgresql/1d60e58/ : SUCCESS in 53m 15s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"5e7dfe748985c148df5cc94c394fa6c27f7bc763","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-28 10:03:38.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"1866385f902e27083f14a9d9451700d958926a32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 11:30:29.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade http://logs.openstack.org/74/639774/18/check/grenade/31b6135/ : SUCCESS in 58m 14s\n- grenade-heat-test http://logs.openstack.org/74/639774/18/check/grenade-heat-test/17c45cc/ : SUCCESS in 1h 19m 20s\n- grenade-py3 http://logs.openstack.org/74/639774/18/check/grenade-py3/d3b3ee9/ : SUCCESS in 1h 05m 50s\n- grenade-postgresql http://logs.openstack.org/74/639774/18/check/grenade-postgresql/42044ad/ : SUCCESS in 1h 03m 24s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"d16b34c4519e0d6c7e1bb08ede764650204fe806","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-28 13:47:39.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"e8c09539677b4d6f7d4fe80db187794ec4f0682b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 15:01:25.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade http://logs.openstack.org/74/639774/19/check/grenade/752a984/ : SUCCESS in 58m 14s\n- grenade-heat-test http://logs.openstack.org/74/639774/19/check/grenade-heat-test/68b52cf/ : SUCCESS in 1h 12m 15s\n- grenade-py3 http://logs.openstack.org/74/639774/19/check/grenade-py3/d3eb4ff/ : SUCCESS in 51m 53s\n- grenade-postgresql http://logs.openstack.org/74/639774/19/check/grenade-postgresql/4a2163a/ : SUCCESS in 49m 25s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"59821389de4552da8a8447c1319e3f2ec26ba810","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-29 10:45:30.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"c0680ca74c13cbce6f30e4d351b2cff5226fd417","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 12:05:07.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/20/check/grenade/e89311a/ : POST_FAILURE in 51m 35s\n- grenade-heat-test http://logs.openstack.org/74/639774/20/check/grenade-heat-test/c5f74c2/ : POST_FAILURE in 1h 16m 50s\n- grenade-py3 http://logs.openstack.org/74/639774/20/check/grenade-py3/d37eb5b/ : POST_FAILURE in 53m 05s\n- grenade-postgresql http://logs.openstack.org/74/639774/20/check/grenade-postgresql/ec88274/ : POST_FAILURE in 50m 35s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"8592e790566fdee9a0d74a1ac2fd6fd6e14b434b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-29 13:57:33.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"07ea403dbe77b34a2e7e8ba00765cb7f98b7a48b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 15:30:43.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade http://logs.openstack.org/74/639774/20/check/grenade/f9c659c/ : SUCCESS in 1h 26m 03s\n- grenade-heat-test http://logs.openstack.org/74/639774/20/check/grenade-heat-test/0781f95/ : SUCCESS in 1h 18m 55s\n- grenade-py3 http://logs.openstack.org/74/639774/20/check/grenade-py3/32bcfdb/ : SUCCESS in 1h 00m 37s\n- grenade-postgresql http://logs.openstack.org/74/639774/20/check/grenade-postgresql/a0ac3bf/ : SUCCESS in 49m 09s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"cb83d8c55fca66a94e20f9dc47eac401563165af","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-29 15:31:42.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"f6ff755bb4a7ca9d89f22ae4cca70956f75dd0f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 17:00:10.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/21/check/grenade/d40ff1c/ : SUCCESS in 1h 14m 11s\n- grenade-heat-test http://logs.openstack.org/74/639774/21/check/grenade-heat-test/e01deb2/ : SUCCESS in 1h 21m 42s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/21/check/grenade-heat-multinode-test/6a6cdcc/ : RETRY_LIMIT in 11m 16s\n- grenade-py3 http://logs.openstack.org/74/639774/21/check/grenade-py3/de5773f/ : SUCCESS in 1h 12m 35s\n- grenade-postgresql http://logs.openstack.org/74/639774/21/check/grenade-postgresql/a2157bd/ : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":21},{"id":"10baa56115ff5a2e2c657e1f5262e5408ea8371f","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-29 20:44:34.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"ffcabb8a21f24f9de683e67f2780434d327c75c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 22:12:46.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/22/check/grenade/d465d86/ : SUCCESS in 1h 04m 14s\n- grenade-heat-test http://logs.openstack.org/74/639774/22/check/grenade-heat-test/5b1c1aa/ : SUCCESS in 1h 17m 50s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/22/check/grenade-heat-multinode-test/6539048/ : RETRY_LIMIT in 11m 47s\n- grenade-py3 http://logs.openstack.org/74/639774/22/check/grenade-py3/e6463dc/ : SUCCESS in 59m 29s\n- grenade-postgresql http://logs.openstack.org/74/639774/22/check/grenade-postgresql/6301bd5/ : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":22},{"id":"96676e9727d075542cad5e75d1e1a44ff8a1d21f","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-29 22:50:08.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"1600cf676bd83c1afe0d6f04c29b4c89fe884389","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-30 00:48:19.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/23/check/grenade/cb895a7/ : SUCCESS in 1h 05m 04s\n- grenade-heat-test http://logs.openstack.org/74/639774/23/check/grenade-heat-test/4f6afdd/ : SUCCESS in 1h 56m 52s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/23/check/grenade-heat-multinode-test/3a05533/ : FAILURE in 1h 17m 58s\n- grenade-py3 http://logs.openstack.org/74/639774/23/check/grenade-py3/dc635bc/ : SUCCESS in 1h 02m 16s\n- grenade-postgresql http://logs.openstack.org/74/639774/23/check/grenade-postgresql/45c6b38/ : SUCCESS in 54m 34s","accounts_in_message":[],"_revision_number":23},{"id":"dd1402449098a494ced9478efa24a14230e8f514","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-03-30 00:48:36.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"fdbf4c4d00eaca1e47f0b50e0421690751a7972c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-30 02:05:15.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/23/check/grenade/17cf990/ : SUCCESS in 47m 51s\n- grenade-heat-test http://logs.openstack.org/74/639774/23/check/grenade-heat-test/85fa22f/ : SUCCESS in 1h 15m 06s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/23/check/grenade-heat-multinode-test/9e05041/ : FAILURE in 1h 14m 54s\n- grenade-py3 http://logs.openstack.org/74/639774/23/check/grenade-py3/39027a1/ : SUCCESS in 52m 00s\n- grenade-postgresql http://logs.openstack.org/74/639774/23/check/grenade-postgresql/f4c5b73/ : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":23},{"id":"32a5e9a3d47b4cc65e6bb4c981e7c937e671d144","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-01 08:15:44.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"6b16256a93c776193ff417c50f7e02f6d8a977eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 09:12:22.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/24/check/grenade/28abf89/ : FAILURE in 43m 05s\n- grenade-heat-test http://logs.openstack.org/74/639774/24/check/grenade-heat-test/da432a5/ : FAILURE in 37m 08s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/24/check/grenade-heat-multinode-test/1ce6401/ : FAILURE in 31m 16s\n- grenade-py3 http://logs.openstack.org/74/639774/24/check/grenade-py3/16ddb4a/ : FAILURE in 42m 28s\n- grenade-postgresql http://logs.openstack.org/74/639774/24/check/grenade-postgresql/6d5f09a/ : FAILURE in 44m 07s","accounts_in_message":[],"_revision_number":24},{"id":"e0ad9b1238075419db145e35426d948066887e2a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-01 09:18:35.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"7f0fced2140b59b738b1a691ffada900bea5632c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 10:10:04.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/25/check/grenade/6c03c72/ : FAILURE in 33m 16s\n- grenade-heat-test http://logs.openstack.org/74/639774/25/check/grenade-heat-test/9b54877/ : FAILURE in 31m 36s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/25/check/grenade-heat-multinode-test/575ccd4/ : FAILURE in 47m 38s\n- grenade-py3 http://logs.openstack.org/74/639774/25/check/grenade-py3/121a278/ : FAILURE in 37m 29s\n- grenade-postgresql http://logs.openstack.org/74/639774/25/check/grenade-postgresql/b5db86e/ : FAILURE in 47m 59s","accounts_in_message":[],"_revision_number":25},{"id":"c4045665f41403848adf2543f911012d2ff23960","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-02 10:35:05.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"4af682dc98ced108aef33d906a31bd8519ed9cbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 12:20:35.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/26/check/grenade/d10d23a/ : FAILURE in 42m 13s\n- grenade-heat-test http://logs.openstack.org/74/639774/26/check/grenade-heat-test/a196a21/ : FAILURE in 40m 42s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/26/check/grenade-heat-multinode-test/69234c3/ : FAILURE in 49m 23s\n- grenade-py3 http://logs.openstack.org/74/639774/26/check/grenade-py3/dc0246c/ : FAILURE in 51m 52s\n- grenade-postgresql http://logs.openstack.org/74/639774/26/check/grenade-postgresql/83885e2/ : FAILURE in 46m 15s","accounts_in_message":[],"_revision_number":26},{"id":"5df09ca65d49e9901f48752b626827a9d7cacc10","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-02 15:47:05.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.","accounts_in_message":[],"_revision_number":27},{"id":"39a237ac9749164806dc92b7b7f22e056be96d2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 19:25:40.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/27/check/grenade/d8e6797/ : SUCCESS in 1h 03m 25s\n- grenade-heat-test http://logs.openstack.org/74/639774/27/check/grenade-heat-test/8d88f28/ : SUCCESS in 1h 19m 41s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/27/check/grenade-heat-multinode-test/ddfd228/ : FAILURE in 1h 26m 40s\n- grenade-py3 http://logs.openstack.org/74/639774/27/check/grenade-py3/6e2e1f1/ : SUCCESS in 1h 02m 17s\n- grenade-postgresql http://logs.openstack.org/74/639774/27/check/grenade-postgresql/1008a0a/ : SUCCESS in 46m 13s","accounts_in_message":[],"_revision_number":27},{"id":"28777cf89a5ea78a959476be6a948672fc9941cc","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-03 07:47:06.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.","accounts_in_message":[],"_revision_number":28},{"id":"1079213ed39b07aeaef8b639b0ba18638ba8198a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 09:01:55.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/28/check/grenade/1ecf744/ : SUCCESS in 56m 31s\n- grenade-heat-test http://logs.openstack.org/74/639774/28/check/grenade-heat-test/e225614/ : SUCCESS in 1h 13m 26s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/28/check/grenade-heat-multinode-test/239d9e1/ : RETRY_LIMIT in 18m 56s\n- grenade-py3 http://logs.openstack.org/74/639774/28/check/grenade-py3/ef41695/ : SUCCESS in 57m 41s\n- grenade-postgresql http://logs.openstack.org/74/639774/28/check/grenade-postgresql/c85d7e4/ : SUCCESS in 51m 28s","accounts_in_message":[],"_revision_number":28},{"id":"17e3a0311a1b72fdd2b52512ec2b39e4ae2cd145","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-03 09:07:09.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.","accounts_in_message":[],"_revision_number":29},{"id":"9a133050e5c3c2ae50d7e5a186da5a918e38f4bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 11:07:14.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/29/check/grenade/f883798/ : SUCCESS in 56m 05s\n- grenade-heat-test http://logs.openstack.org/74/639774/29/check/grenade-heat-test/455f85b/ : SUCCESS in 1h 14m 49s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/29/check/grenade-heat-multinode-test/35f329f/ : FAILURE in 1h 23m 24s\n- grenade-py3 http://logs.openstack.org/74/639774/29/check/grenade-py3/728bd89/ : SUCCESS in 56m 16s\n- grenade-postgresql http://logs.openstack.org/74/639774/29/check/grenade-postgresql/a68163b/ : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":29},{"id":"dd52ec006439c82357f9a0eb123ce7326bde0006","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-04 16:42:41.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"acfc6a6bf78f9172138c60484ac281dfa46c4a5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 19:03:30.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/30/check/grenade/5c2c513/ : SUCCESS in 53m 36s\n- grenade-heat-test http://logs.openstack.org/74/639774/30/check/grenade-heat-test/d1112c7/ : SUCCESS in 1h 17m 48s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/30/check/grenade-heat-multinode-test/685bdf8/ : FAILURE in 1h 28m 53s\n- grenade-py3 http://logs.openstack.org/74/639774/30/check/grenade-py3/02e2414/ : SUCCESS in 56m 41s\n- grenade-postgresql http://logs.openstack.org/74/639774/30/check/grenade-postgresql/53c3fcd/ : SUCCESS in 53m 54s","accounts_in_message":[],"_revision_number":30},{"id":"56eec7e6f04b668025844b3dc9d215fbbff4efc8","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-04 19:32:35.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.","accounts_in_message":[],"_revision_number":31},{"id":"3b559c76286b0692b333fde7ee4ca7d40c800572","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 20:57:58.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/31/check/grenade/30763a0/ : SUCCESS in 58m 14s\n- grenade-heat-test http://logs.openstack.org/74/639774/31/check/grenade-heat-test/53a7770/ : SUCCESS in 1h 22m 03s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/31/check/grenade-heat-multinode-test/9fd5883/ : FAILURE in 1h 02m 59s\n- grenade-py3 http://logs.openstack.org/74/639774/31/check/grenade-py3/844763f/ : SUCCESS in 56m 08s\n- grenade-postgresql http://logs.openstack.org/74/639774/31/check/grenade-postgresql/ae1ce30/ : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":31},{"id":"12ce30e0a81346a5f81c7db667f06fadb1709e98","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-04 21:30:15.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.","accounts_in_message":[],"_revision_number":32},{"id":"617e9c836cbb20e0ae43c4bbb8e187bde63a3673","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 23:59:20.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/32/check/grenade/7b6ca89/ : SUCCESS in 50m 27s\n- grenade-heat-test http://logs.openstack.org/74/639774/32/check/grenade-heat-test/b947f52/ : TIMED_OUT in 2h 05m 03s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/32/check/grenade-heat-multinode-test/4596df5/ : POST_FAILURE in 1h 40m 39s\n- grenade-py3 http://logs.openstack.org/74/639774/32/check/grenade-py3/5f7c1d9/ : SUCCESS in 1h 04m 26s\n- grenade-postgresql http://logs.openstack.org/74/639774/32/check/grenade-postgresql/84de36e/ : SUCCESS in 40m 59s","accounts_in_message":[],"_revision_number":32},{"id":"0d5ce416b2068af0d0d56d54f9bbdc2543ff9c41","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-05 00:03:03.000000000","message":"Patch Set 32:\n\nrecheck","accounts_in_message":[],"_revision_number":32},{"id":"d7a8c4de9b9ca418f2ec37ec20da977c4018bac5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 01:29:29.000000000","message":"Patch Set 32:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/32/check/grenade/a8c449f/ : SUCCESS in 56m 05s\n- grenade-heat-test http://logs.openstack.org/74/639774/32/check/grenade-heat-test/08082e3/ : SUCCESS in 1h 14m 18s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/32/check/grenade-heat-multinode-test/dbaf305/ : FAILURE in 1h 19m 28s\n- grenade-py3 http://logs.openstack.org/74/639774/32/check/grenade-py3/7017b17/ : SUCCESS in 1h 00m 30s\n- grenade-postgresql http://logs.openstack.org/74/639774/32/check/grenade-postgresql/c534a5e/ : SUCCESS in 38m 43s","accounts_in_message":[],"_revision_number":32},{"id":"7144fd16feda0add11abcd0e27241de8b3cca505","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-05 07:45:31.000000000","message":"Patch Set 32:\n\nrecheck","accounts_in_message":[],"_revision_number":32},{"id":"45745ac4a39d298fc7aaeb052e681fed755151b3","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-05 07:55:51.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"17bab3dfac22c5864ad99fca9a25f03ba46daa19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 10:04:53.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/33/check/grenade/18548a9/ : SUCCESS in 44m 57s\n- grenade-heat-test http://logs.openstack.org/74/639774/33/check/grenade-heat-test/59fc8cf/ : TIMED_OUT in 2h 04m 19s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/33/check/grenade-heat-multinode-test/40b9303/ : FAILURE in 1h 10m 27s\n- grenade-neutron-multinode-test http://logs.openstack.org/74/639774/33/check/grenade-neutron-multinode-test/18c36fc/ : SUCCESS in 1h 11m 43s\n- grenade-py3 http://logs.openstack.org/74/639774/33/check/grenade-py3/139803e/ : SUCCESS in 55m 32s\n- grenade-postgresql http://logs.openstack.org/74/639774/33/check/grenade-postgresql/027bd1a/ : SUCCESS in 52m 15s","accounts_in_message":[],"_revision_number":33},{"id":"09b9fd096334ca626f6e5b02d735d389342d6652","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-05 13:45:37.000000000","message":"Patch Set 33:\n\nrecheck","accounts_in_message":[],"_revision_number":33},{"id":"92de73b05b5b5ed29c5764dea3aca1360db4486a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 15:18:43.000000000","message":"Patch Set 33:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/33/check/grenade/961c9ec/ : SUCCESS in 52m 16s\n- grenade-heat-test http://logs.openstack.org/74/639774/33/check/grenade-heat-test/9a65307/ : SUCCESS in 1h 12m 00s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/33/check/grenade-heat-multinode-test/28a4b29/ : FAILURE in 1h 25m 28s\n- grenade-neutron-multinode-test http://logs.openstack.org/74/639774/33/check/grenade-neutron-multinode-test/3e1703a/ : SUCCESS in 1h 01m 16s\n- grenade-py3 http://logs.openstack.org/74/639774/33/check/grenade-py3/0b22c1d/ : SUCCESS in 51m 29s\n- grenade-postgresql http://logs.openstack.org/74/639774/33/check/grenade-postgresql/ff43e60/ : SUCCESS in 46m 48s","accounts_in_message":[],"_revision_number":33},{"id":"95b310f923d702b183e5e59853fb71f11b406fb5","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-20 13:15:47.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.","accounts_in_message":[],"_revision_number":34},{"id":"ded98d7d144cae569dd57e7314410518e249a316","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 13:16:05.000000000","message":"Patch Set 34: Verified-1\n\n(6 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/grenade on branch master.  The error was:\n\n  Unknown project openstack-dev/devstack\n\nThe error appears in the following job stanza:\n\n  job:\n      name: grenade-base\n      parent: multinode\n      abstract: true\n      description: |\n        Base grenade job\n  \n        This is the base grenade job for both single and multi node grenade\n        tests. It does not define any service to be used, so that this can\n        be used to build an upgrade test with a minimal number of services.\n      nodeset: openstack-single-node-bionic\n      required-projects:\n        - openstack-dev/devstack\n        - openstack/tempest\n      roles:\n        - zuul: openstack-infra/openstack-zuul-jobs\n        - zuul: openstack-dev/devstack\n        - zuul: openstack/tempest\n      vars:\n        devstack_bases:\n          shared: /opt/stack\n          old: /opt/stack/old\n          new: /opt/stack/new\n          logs_old: /opt/stack/logs/old\n          logs_new: /opt/stack/logs/new\n        grenade_devstack_localrc: \u0026grenade_base_devstack_localrc\n          old:\n            DEST: \"{{ devstack_bases.old }}\"\n            LOGFILE: \"{{ devstack_bases.old }}/logs/devstacklog.txt\"\n          new:\n            DEST: \"{{ devstack_bases.new }}\"\n            LOGFILE: \"{{ devstack_bases.new }}/logs/devstacklog.txt\"\n          shared:\n            # Common settings for OpenStack CI\n            DATABASE_PASSWORD: secretdatabase\n            # FIXME: temporarily define MYSQL_PASSWORD too,\n            # because some code in devstack and in grenade\n            # still requires it to be defined.\n            MYSQL_PASSWORD: secretdatabase\n            RABBIT_PASSWORD: secretrabbit\n            ADMIN_PASSWORD: secretadmin\n            SERVICE_PASSWORD: secretservice\n            NETWORK_GATEWAY: 10.1.0.1\n            FIXED_RANGE: 10.1.0.0/20\n            IPV4_ADDRS_SAFE_TO_USE: 10.1.0.0/20\n            FLOATING_RANGE: 172.24.5.0/24\n            PUBLIC_NETWORK_GATEWAY: 172.24.5.1\n            FIXED_NETWORK_SIZE: 4096\n            LOG_COLOR: false\n            VERBOSE: true\n            VERBOSE_NO_TIMESTAMP: true\n            NOVNC_FROM_PACKAGE: true\n            ERROR_ON_CLONE: true\n            # Gate jobs can\u0027t deal with nested virt. Disable it.\n            LIBVIRT_TYPE: qemu\n        # grenade_from_branch is branch specific, it\u0027s the change branch \"-1\".\n        # This has to be updated on master when cutting a new stable branch.\n        # grenade_to_branch is always the current patch on pullup jobs.\n        grenade_from_branch: stable/rocky\n        grenade_localrc:\n          SAVE_DIR: \"{{ devstack_bases.shared }}/save\"\n        # from tempest playbooks\n        devstack_local_conf:\n          test-config:\n            $TEMPEST_CONFIG:\n              compute:\n                min_compute_nodes: \"{{ groups[\u0027compute\u0027] | default([\u0027controller\u0027]) | length }}\"\n        test_results_stage_name: test_results\n        zuul_copy_output:\n          \u0027{{ devstack_bases.logs_old }}\u0027: logs\n          \u0027{{ devstack_bases.logs_new }}\u0027: logs\n          \u0027{{ grenade_save_dir }}\u0027: logs\n          \u0027{{ grenade_log_dir }}/grenade.sh.log\u0027: logs\n          \u0027{{ grenade_log_dir }}/grenade.sh.log.summary\u0027: logs\n          \u0027{{ grenade_conf_dir }}/grenade_localrc\u0027: logs_txt\n          \u0027{{ grenade_conf_dir }}/grenade_pluginrc\u0027: logs_txt\n          \u0027{{ devstack_bases.new }}/tempest/etc/tempest.conf\u0027: logs\n          \u0027{{ devstack_bases.new }}/tempest/etc/accounts.yaml\u0027: logs\n          \u0027{{ devstack_bases.new }}/tempest/tempest.log\u0027: logs\n          \u0027{{ stage_dir }}/{{ test_results_stage_name }}.subunit\u0027: logs\n          \u0027{{ stage_dir }}/{{ test_results_stage_name }}.html\u0027: logs\n          \u0027{{ stage_dir }}/stackviz\u0027: logs\n          \u0027{{ stage_dir }}/apache\u0027: logs\n          \u0027{{ stage_dir }}/apache_config\u0027: logs\n          \u0027{{ stage_dir }}/etc\u0027: logs\n          \u0027/var/log/rabbitmq\u0027: logs\n          \u0027/var/log/postgresql\u0027: logs\n          \u0027/var/log/mysql.err\u0027: logs\n          \u0027/var/log/mysql.log\u0027: logs\n          \u0027/var/log/libvirt\u0027: logs\n          \u0027/etc/sudoers\u0027: logs\n          \u0027/etc/sudoers.d\u0027: logs\n          \u0027{{ stage_dir }}/iptables.txt\u0027: logs\n          \u0027{{ stage_dir }}/df.txt\u0027: logs\n          \u0027{{ stage_dir }}/pip2-freeze.txt\u0027: logs\n          \u0027{{ stage_dir }}/pip3-freeze.txt\u0027: logs\n          \u0027{{ stage_dir }}/dpkg-l.txt\u0027: logs\n          \u0027{{ stage_dir }}/rpm-qa.txt\u0027: logs\n          \u0027{{ stage_dir }}/core\u0027: logs\n          \u0027{{ stage_dir }}/listen53.txt\u0027: logs\n          \u0027{{ stage_dir }}/deprecations.log\u0027: logs\n          \u0027/var/log/ceph\u0027: logs\n          \u0027/var/log/openvswitch\u0027: logs\n          \u0027/var/log/glusterfs\u0027: logs\n          \u0027/etc/glusterfs/glusterd.vol\u0027: logs\n          \u0027/etc/resolv.conf\u0027: logs\n          \u0027/var/log/unbound.log\u0027: logs\n        extensions_to_txt:\n          auto: True\n          conf: True\n          log: True\n          localrc: True\n          stackenv: True\n          summary: True\n      group-vars:\n        subnode:\n          grenade_devstack_localrc:\n            \u003c\u003c: *grenade_base_devstack_localrc\n      pre-run: playbooks/pre.yaml\n      run: playbooks/grenade.yaml\n      post-run: playbooks/post.yaml\n      irrelevant-files:\n        # Documentation related\n        - ^.*\\.rst$\n        - ^api-ref/.*$\n        - ^doc/.*$\n        - ^releasenotes/.*$\n        # Translations\n        - ^.*/locale/.*po$\n\n  in \"openstack/grenade/.zuul.yaml@master\", line 42, column 3","accounts_in_message":[],"_revision_number":34},{"id":"5d8747618657ed89a37cdb7be1ccca54455a72c0","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-20 13:25:49.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.","accounts_in_message":[],"_revision_number":35},{"id":"cae99ebc2075c57a3d4a8d402d53fc273fb38a0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 14:38:03.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade http://logs.openstack.org/74/639774/35/check/grenade/c4903d8/ : SUCCESS in 42m 51s\n- grenade-heat-test http://logs.openstack.org/74/639774/35/check/grenade-heat-test/e4257d7/ : SUCCESS in 1h 07m 23s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/35/check/grenade-heat-multinode-test/d252f09/ : SUCCESS in 1h 11m 11s\n- grenade-neutron-multinode-test http://logs.openstack.org/74/639774/35/check/grenade-neutron-multinode-test/1562701/ : SUCCESS in 54m 36s\n- grenade-py3 http://logs.openstack.org/74/639774/35/check/grenade-py3/18bac01/ : SUCCESS in 39m 52s\n- grenade-postgresql http://logs.openstack.org/74/639774/35/check/grenade-postgresql/4d5a1d5/ : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":35},{"id":"5bdde7e0844c591062a30e47c1397133efd3388a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-20 17:58:31.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.","accounts_in_message":[],"_revision_number":36},{"id":"6c7e28275f7e9171408fd1dd77f82a2c4847cf23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 19:18:54.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/36/check/grenade/c080891/ : SUCCESS in 55m 21s\n- grenade-heat-test http://logs.openstack.org/74/639774/36/check/grenade-heat-test/79ca76d/ : FAILURE in 1h 17m 39s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/36/check/grenade-heat-multinode-test/82bc6ce/ : SUCCESS in 1h 14m 14s\n- grenade-neutron-multinode-test http://logs.openstack.org/74/639774/36/check/grenade-neutron-multinode-test/1e54875/ : SUCCESS in 54m 00s\n- grenade-py3 http://logs.openstack.org/74/639774/36/check/grenade-py3/bb49ff3/ : SUCCESS in 46m 44s\n- grenade-postgresql http://logs.openstack.org/74/639774/36/check/grenade-postgresql/12648ee/ : SUCCESS in 50m 30s","accounts_in_message":[],"_revision_number":36},{"id":"94beda9e8f296e3c1f16352c201e1ad21d40729b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-20 23:22:45.000000000","message":"Patch Set 36:\n\nrecheck","accounts_in_message":[],"_revision_number":36},{"id":"d8ac25bda2d699010258c912238e01315e9e7a90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 01:08:27.000000000","message":"Patch Set 36:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/36/check/grenade/ced7ef3/ : SUCCESS in 51m 28s\n- grenade-heat-test http://logs.openstack.org/74/639774/36/check/grenade-heat-test/c0f437c/ : SUCCESS in 1h 12m 30s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/36/check/grenade-heat-multinode-test/a818a7d/ : FAILURE in 1h 17m 46s\n- grenade-neutron-multinode-test http://logs.openstack.org/74/639774/36/check/grenade-neutron-multinode-test/5bc732f/ : SUCCESS in 1h 01m 18s\n- grenade-py3 http://logs.openstack.org/74/639774/36/check/grenade-py3/ae26c7c/ : SUCCESS in 53m 19s\n- grenade-postgresql http://logs.openstack.org/74/639774/36/check/grenade-postgresql/428897a/ : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":36},{"id":"de2cb978936c045c032b0ae8b4a91706c1177a1b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-21 09:32:45.000000000","message":"Patch Set 36:\n\nrecheck","accounts_in_message":[],"_revision_number":36},{"id":"befbb45c58180f7e146004957ca129ba034dd495","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 11:13:39.000000000","message":"Patch Set 36:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/36/check/grenade/cf6f3ac/ : SUCCESS in 49m 28s\n- grenade-heat-test http://logs.openstack.org/74/639774/36/check/grenade-heat-test/97d023a/ : SUCCESS in 1h 14m 41s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/36/check/grenade-heat-multinode-test/0ff9b41/ : FAILURE in 1h 35m 32s\n- grenade-neutron-multinode-test http://logs.openstack.org/74/639774/36/check/grenade-neutron-multinode-test/042cfea/ : SUCCESS in 59m 00s\n- grenade-py3 http://logs.openstack.org/74/639774/36/check/grenade-py3/3e626b5/ : SUCCESS in 49m 59s\n- grenade-postgresql http://logs.openstack.org/74/639774/36/check/grenade-postgresql/2c77e7e/ : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":36},{"id":"17946f54bf2468fe7bf83eb76c0149c6f7f3df37","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-21 21:57:53.000000000","message":"Patch Set 36:\n\nrecheck","accounts_in_message":[],"_revision_number":36},{"id":"b9d89417ef67c6272ff90a6fd4aec9e47215eac5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 23:41:14.000000000","message":"Patch Set 36:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/36/check/grenade/642ca45/ : SUCCESS in 39m 59s\n- grenade-heat-test http://logs.openstack.org/74/639774/36/check/grenade-heat-test/fe42a00/ : SUCCESS in 1h 00m 24s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/36/check/grenade-heat-multinode-test/a42c9b4/ : FAILURE in 1h 27m 48s\n- grenade-neutron-multinode-test http://logs.openstack.org/74/639774/36/check/grenade-neutron-multinode-test/0cd090a/ : SUCCESS in 58m 07s\n- grenade-py3 http://logs.openstack.org/74/639774/36/check/grenade-py3/bde48d6/ : SUCCESS in 52m 13s\n- grenade-postgresql http://logs.openstack.org/74/639774/36/check/grenade-postgresql/92e1cc5/ : SUCCESS in 48m 39s","accounts_in_message":[],"_revision_number":36},{"id":"70504617c8cc9a269dfa2f1295b2f39736f7b6dc","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-05-23 17:05:54.000000000","message":"Uploaded patch set 37: Patch Set 36 was rebased.","accounts_in_message":[],"_revision_number":37},{"id":"264feb937334b3827ad503388b76db327e63d55e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 22:20:18.000000000","message":"Patch Set 37:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/37/check/grenade/6a2473a/ : SUCCESS in 49m 39s\n- grenade-heat-test http://logs.openstack.org/74/639774/37/check/grenade-heat-test/29d3d37/ : SUCCESS in 1h 09m 21s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/37/check/grenade-heat-multinode-test/867518a/ : FAILURE in 34m 28s\n- grenade-neutron-multinode-test http://logs.openstack.org/74/639774/37/check/grenade-neutron-multinode-test/68aab5e/ : SUCCESS in 1h 03m 08s\n- grenade-py3 http://logs.openstack.org/74/639774/37/check/grenade-py3/0ad5923/ : SUCCESS in 59m 08s\n- grenade-postgresql http://logs.openstack.org/74/639774/37/check/grenade-postgresql/0d19450/ : SUCCESS in 45m 58s","accounts_in_message":[],"_revision_number":37},{"id":"022cd7aff8136b392dd017f04497c3c9c8a4ea21","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-07-25 16:21:22.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.","accounts_in_message":[],"_revision_number":38},{"id":"b2f10c4e66faa8c6b20a7d55742455be241e0b40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 19:39:06.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade http://logs.openstack.org/74/639774/38/check/grenade/3d726c0/ : SUCCESS in 54m 15s\n- grenade-heat-test http://logs.openstack.org/74/639774/38/check/grenade-heat-test/66d6eed/ : SUCCESS in 1h 16m 57s\n- grenade-heat-multinode-test http://logs.openstack.org/74/639774/38/check/grenade-heat-multinode-test/9e4d716/ : FAILURE in 1h 20m 48s\n- grenade-neutron-multinode-test http://logs.openstack.org/74/639774/38/check/grenade-neutron-multinode-test/1911d8b/ : SUCCESS in 1h 15m 16s\n- grenade-py3 http://logs.openstack.org/74/639774/38/check/grenade-py3/8f2112c/ : SUCCESS in 57m 45s\n- grenade-postgresql http://logs.openstack.org/74/639774/38/check/grenade-postgresql/370d123/ : FAILURE in 19m 28s","accounts_in_message":[],"_revision_number":38},{"id":"6f5d86d3859d22fb721807e818089f555c064d4a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-17 17:27:02.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"af2c8b0a029c9d259f3fb6d0730b9b8b76defdd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 19:01:28.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4011c859394b4576a4cf0c9d56ad3a03 : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/e9c211829dc146d4a16417f81d4e9164 : FAILURE in 35m 58s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/c432658c9f184a0198a0f1bec6c7d042 : FAILURE in 46m 55s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/c81a8075d3c64cefaea0dac53f4f6f43 : FAILURE in 1h 02m 49s\n- grenade-neutron-multinode-test https://zuul.opendev.org/t/openstack/build/8fcaea1316414e178f17376947a72a98 : FAILURE in 50m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9aa51af25f1c4c4c8560c3f5840c4d2d : FAILURE in 41m 00s","accounts_in_message":[],"_revision_number":39},{"id":"9bc0004c484c6dece605118886f3ae09f9323771","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-17 19:26:05.000000000","message":"Uploaded patch set 40: Patch Set 39 was rebased.","accounts_in_message":[],"_revision_number":40},{"id":"1eb5cb8bee1612711e1f324eeb2383a914de8887","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-17 21:00:36.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"2308c7344a950d590023e69cbef60c790843b217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 22:31:21.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ee34974e0243a0be3cea59fc34e05b : SUCCESS in 14m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/410c2ff2368543f284659a6c7f735681 : FAILURE in 46m 06s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/a69f38450d314cd2849a77f29044b95b : FAILURE in 1h 02m 00s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/9761847130434d4dbafd8f267be6e69c : FAILURE in 1h 19m 38s\n- grenade-neutron-multinode-test https://zuul.opendev.org/t/openstack/build/6f91fa1f10ec4c1cabf8c1983dd9a125 : SUCCESS in 1h 16m 37s","accounts_in_message":[],"_revision_number":41},{"id":"8f3664ca99ad31d1676352afb3a7e87d8f437573","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-23 15:10:38.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"b29803f88982d5c921b1a260523e6613e78c94f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 16:53:01.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e68c588b8be44f7bab6e892ef9f2a6e4 : SUCCESS in 13m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/ab72d89972204f97b13370fb67e60e26 : SUCCESS in 55m 16s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/81b2d668733546e49b1a58991b0e4c77 : FAILURE in 47m 44s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/f1d43fb4bbd34e48a87708108680518e : FAILURE in 58m 28s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/793a56c51bb048dba2bc769c4f96ba44 : SUCCESS in 1h 11m 38s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/24e861d3bfa54a489044687da39e2022 : SUCCESS in 1h 04m 45s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/04ffe64e659848e4956837f9f6590e9a : SUCCESS in 1h 39m 48s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"8e91ee162e526b59b94167c92f9e04e677b12c06","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-01-20 14:57:24.000000000","message":"Uploaded patch set 43: Patch Set 42 was rebased.","accounts_in_message":[],"_revision_number":43},{"id":"b56fdb5d4ce3acc130b2ab742b384ea73f54c08b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 18:13:25.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cccd892c7b094faa8f6a2314accd4f77 : FAILURE in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/04ad9b2aeea940a598919f5d9b28462a : SUCCESS in 1h 00m 43s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/99274b0f76f0486fae9ad84d35a1ce84 : FAILURE in 48m 58s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/ebfb75b7ec1f469aab87e2e6c8b5d704 : FAILURE in 1h 01m 49s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/3f10316df8554421883bd50b4be93709 : FAILURE in 1h 10m 00s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/0ad8bdc62ca14706bd6ec79639cc6d8a : SUCCESS in 1h 11m 19s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d4626f08580a4c169b18741eff3a9a66 : SUCCESS in 1h 41m 44s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"a7eaf9a949b8cb7228c0ca7fade0211af1c81b89","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-01-28 13:49:47.000000000","message":"Uploaded patch set 44: Commit message was updated.","accounts_in_message":[],"_revision_number":44},{"id":"8dd4b5c77a424817174a8ea1d4e47285280d89f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 15:55:12.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0fea4ed8c794ee98ecd04f0e3607eb9 : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/e3cf45acce0346c697dd6b2cd4c7bb6b : SUCCESS in 59m 19s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/ed0fbab5add14ba1abb1eff40cdccffc : FAILURE in 1h 06m 51s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/5aaf0390daca4f25b78ad0ca715004cc : FAILURE in 1h 39m 18s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/c2ef4d98fc294677876aa940bccb430e : FAILURE in 1h 07m 23s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/64b21030d77548f0aaf3cfc632ea70a4 : SUCCESS in 1h 36m 07s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/7c25c03966a049d699f2b92ccad547ef : SUCCESS in 1h 56m 39s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"7cd167400a37066e117db1fe237e3eca1b494616","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-01-29 13:25:14.000000000","message":"Patch Set 44:\n\nUhm, both the old and the new grenade-heat test failed in the same way! That\u0027s a good sign I guess. I\u0027m not sure why they are failing though, the past results are a bit mixed. Let\u0027s retry.","accounts_in_message":[],"_revision_number":44},{"id":"507e5e5127b0fd153689171137e4d5ea0b2209c7","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-01-29 13:25:20.000000000","message":"Patch Set 44:\n\nrecheck","accounts_in_message":[],"_revision_number":44},{"id":"01bbc0676c5d2fac533475b5fef0b270676e411c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 15:18:54.000000000","message":"Patch Set 44:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d73f61d4a5c046059e2cfda0bc2b64b9 : SUCCESS in 11m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/3cf9a7604c7a42a4ab366666340a43ea : SUCCESS in 55m 50s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/ba0bcdde3acc46c995e679b8936d8068 : FAILURE in 1h 17m 44s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/3b75af0752ef4ec1a50880efd6ba7a20 : FAILURE in 1h 39m 02s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/e822f70538c24243b11a944a1c58171b : FAILURE in 1h 12m 59s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/a218e0019f4647539609e69f16fc5533 : SUCCESS in 1h 01m 57s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/78fe18d442114f2c8a56ae79f79abfc6 : SUCCESS in 1h 47m 02s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"d00e7f2f0cc9003c6215b23d5d0fd721091424a3","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-02-07 14:47:37.000000000","message":"Uploaded patch set 45: Commit message was updated.","accounts_in_message":[],"_revision_number":45},{"id":"cc7fd992ace02cccb222edfba061344309d09c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 16:30:14.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b404c3fbf3434b12ae7dfc9be3b89d4f : SUCCESS in 10m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/cb79ed245833460492558dfc78a0beda : SUCCESS in 1h 04m 37s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/db8479a0bdce4dadaa531940f6afb068 : SUCCESS in 1h 26m 40s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/22512b83e21b45d6b1102c77cf49acda : SUCCESS in 1h 27m 58s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/97feaefbebd54dfc867bd5f38f0a5d6e : SUCCESS in 1h 18m 37s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/8988d74bc254416d94dc9991ad053ebc : SUCCESS in 1h 09m 35s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/1164ee48e93f44d6afed30a757b87e22 : SUCCESS in 1h 37m 19s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"44cf5f570b0b8d75661b500595bad2760e9655a8","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-02-12 22:32:48.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"18712048efd636abfa80f63b338cf503a299665e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 00:12:30.000000000","message":"Patch Set 45:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c33281d04047bc8fc6d78fbc6852fa : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/1feda3b6ca1043fa9ab9af5190d78b86 : SUCCESS in 56m 27s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/db33365fa79342809328b782385c9557 : SUCCESS in 1h 24m 37s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/f7754d690bf1460c8b62ed1144ab9f0f : SUCCESS in 1h 34m 05s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/19b1a9cd50a84093a7396a6e368d61ed : FAILURE in 5m 21s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/715b9330064e40b89a52317b98c75d8c : SUCCESS in 1h 04m 10s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f9a65b9105c8479487957b7318702c12 : FAILURE in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"d0bbf12b199832ab1ecbdc1532d7650506fb6a3a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-02-13 08:21:54.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"d44d56011ce08abc30be5808e3964e5087e6a2a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:23:07.000000000","message":"Patch Set 45:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b185747baca46c9a87fd86ed63716c7 : SUCCESS in 14m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/359c7783705d4f529ace5d2e6572caca : SUCCESS in 59m 38s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/55fb9c9282054528a03103924745eefe : SUCCESS in 1h 19m 05s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/2e13bd9a89c6460c832c2323861dd2be : SUCCESS in 1h 36m 14s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/ee4ad84344f34866858b8be371153d74 : SUCCESS in 1h 15m 25s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/3c96f35af5bf4712a94031ceaedddc08 : SUCCESS in 1h 19m 37s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/232a9dcb8f4846f984b61d36027b06b9 : SUCCESS in 1h 53m 28s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"e41367869a1f5c49724bc80ce5fcb3f62ddd5804","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-02 11:51:43.000000000","message":"Uploaded patch set 46: Patch Set 45 was rebased.","accounts_in_message":[],"_revision_number":46},{"id":"0b3c56a0615231d4940436bc01a0597734518fda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 15:24:43.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e3c90449b604c39979338a809493cde : SUCCESS in 10m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/1382323b87fd43f2bfa715b5188acd19 : SUCCESS in 52m 03s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/6d327a5a9dcd40b19838d9842fa3c01d : FAILURE in 1h 51m 42s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/5f8a8efd12c8492a979349605dc966ff : FAILURE in 1h 04m 49s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/ce8b929c29a04ad6b557c9b1c36a5096 : SUCCESS in 1h 13m 25s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/8eea845a9a5c44ccac9b688962be56ce : SUCCESS in 1h 08m 09s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f3c08ccbb73241198cb24958bb1fbc9e : SUCCESS in 1h 44m 46s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"38e18d37c81a918c77e5b30cdc5e8cf28c5f29f1","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-02 15:26:06.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"e66348b5f6398abbcd1b467e1ed7591a5cd1ae1b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-02 17:20:52.000000000","message":"Uploaded patch set 48: Patch Set 47 was rebased.","accounts_in_message":[],"_revision_number":48},{"id":"b7bd19dce5760d877b758066ad8c55db7c1cd4c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 22:45:54.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa2f734e35b4c31bd7e8ce60742e3bb : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/d7abed20d79045c3ad97321373915119 : SUCCESS in 1h 22m 17s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/987defc29caf42b39cc230f53fbd240e : FAILURE in 35m 46s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/e79874514d5b4a6ca2e6b80f0cbe0c45 : FAILURE in 49m 32s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/e4489f9a0c244ff382cc4c6f2d3c7ee8 : SUCCESS in 1h 13m 13s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/e2bc44de21e0484f96a96eecb6392438 : SUCCESS in 1h 25m 47s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/eeb9a31df7af413ab0f37bed2773fe95 : SUCCESS in 1h 50m 15s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"fe9d487c1aa7f9df93538c639d30bda9df196b37","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-18 22:20:35.000000000","message":"Uploaded patch set 49.","accounts_in_message":[],"_revision_number":49},{"id":"067430bcf231fa71dccf18201e5628c03cb7034d","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-18 23:11:34.000000000","message":"Uploaded patch set 50: Patch Set 49 was rebased.","accounts_in_message":[],"_revision_number":50},{"id":"d6c58a5b34dc30b6d98d4aad50703ae3b6926f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 00:52:27.000000000","message":"Patch Set 50: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1c7d0c33464841a5f8984263283a8a : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/9af1e01cf59140818e61c2100d50a609 : FAILURE in 30m 00s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/1845929e671e4cb780dedc96a51da764 : FAILURE in 35m 03s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/658d7d1bb3924099bbf4631852509ac3 : FAILURE in 44m 42s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/8f709b60571c4b4fa133cfa6f69da6f2 : SUCCESS in 1h 12m 17s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/5c81b5185d394a84ab2e4346e0b8ae7e : FAILURE in 39m 39s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/5d2a25e89283450d9a12dbc11ad8ffb6 : SUCCESS in 1h 39m 06s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"387d5e9016dabe7164fc4f2477b903ed42f394ab","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-06 09:53:55.000000000","message":"Uploaded patch set 51: Patch Set 50 was rebased.","accounts_in_message":[],"_revision_number":51},{"id":"f8fc7313ec7948e962bbe2b2a00d2bfecedc2a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 12:41:06.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd8a28023b9d464ca94bf4c7a826933c : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/f5d5a6dfeb704eabb7caf000d8336580 : SUCCESS in 1h 18m 59s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/e9536566d71e40f99b477b4746d35004 : SUCCESS in 1h 26m 41s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/1837e15266eb4bc780273e08eb759bb1 : SUCCESS in 1h 42m 14s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/381cb4daf4a64e549e3b6de35219954e : SUCCESS in 1h 26m 52s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/01ea4d09de2347349947c295ddd1bdb5 : SUCCESS in 1h 37m 21s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d336da2854dd426fae2cd275034ded40 : SUCCESS in 2h 09m 38s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"62c26425369ed7dc0f673fbf08af75af3d770ac0","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-23 23:37:08.000000000","message":"Uploaded patch set 52: Patch Set 51 was rebased.","accounts_in_message":[],"_revision_number":52},{"id":"9b8d02b1edc29a09362357e8fc57eef4717bbf44","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-23 23:47:40.000000000","message":"Uploaded patch set 53: Patch Set 52 was rebased.","accounts_in_message":[],"_revision_number":53},{"id":"bdb5add15a7adfad3ba29fd03f8f7a8ec0921209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 01:31:47.000000000","message":"Patch Set 53: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af6df693fced491d9f7b403d67a32449 : SUCCESS in 8m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/51f8952d526d431db01f7f6e56ec8957 : SUCCESS in 1h 13m 13s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/011f2315702d40bebb4a2fb622df8dab : SUCCESS in 1h 26m 36s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/cc687d4c5f2945818c1b0be3bb275fb2 : FAILURE in 1h 36m 00s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/c95a233a49944ad8b5c38b26af845aef : SUCCESS in 1h 23m 49s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/306cdf7c753c4a0b8a39cf6920693fef : SUCCESS in 1h 20m 58s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/17b26605dd85451a9ef69d2542b976b5 : SUCCESS in 1h 40m 27s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"815f8814133acc02953fa9ebbe9838665c1d2d40","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-24 01:49:00.000000000","message":"Patch Set 53:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":53},{"id":"645aeee4db0f5f76f4279f283f51cd71be0542dc","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-24 07:52:04.000000000","message":"Patch Set 51:\n\n(1 comment)","accounts_in_message":[],"_revision_number":51},{"id":"f8eec8c19ea8ae34fa9f69e546500feeb438089f","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-24 13:44:31.000000000","message":"Uploaded patch set 54: Patch Set 53 was rebased.","accounts_in_message":[],"_revision_number":54},{"id":"8a266d525b9da730c046233eb16c2ff8aa96b91a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 16:42:18.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f553ca461242c5b225a0754fe76c1e : SUCCESS in 10m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/f1022887f3734c3a9429af8350161322 : SUCCESS in 1h 11m 31s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/d4146e1f89134675949bd4d8936746d8 : SUCCESS in 1h 24m 26s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/2e18dee6a74f47149572cd1a9cb6574e : SUCCESS in 1h 46m 10s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/8cf588a978844659bd92f68849a7f067 : SUCCESS in 1h 24m 09s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/53e4442255ef4326840872ca29a3579b : SUCCESS in 1h 31m 24s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/4daf7c960f724b0495966b9fc5c16a4b : SUCCESS in 1h 52m 37s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"22ca792a3feb3f690a984eef4497718295fd9b17","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-25 13:40:17.000000000","message":"Uploaded patch set 55.","accounts_in_message":[],"_revision_number":55},{"id":"ec54f6c25f4d953a3c4014ba6e51c95fa03cd43f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-25 15:16:58.000000000","message":"Patch Set 55:\n\nmagically this makes original heat job pass too","accounts_in_message":[],"_revision_number":55},{"id":"d1016f5bd0ecea49a34e5cb52dabe2d4d51906e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 15:23:11.000000000","message":"Patch Set 55: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70a2c7f6303e424d85ba266e81012bae : SUCCESS in 6m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/ff9319fa96c2486e9dc7fdc1ff817c27 : SUCCESS in 1h 05m 39s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/c71ef1081cd74f2d9d8349203f6dcaaf : SUCCESS in 1h 27m 27s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/b97e1ddd247d4661803eea3b201434fe : FAILURE in 1h 37m 52s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/81d832f91e5f423ba8e268958dfcb208 : SUCCESS in 1h 13m 13s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/91b06417f80946298195ae52dc17c4c1 : SUCCESS in 1h 18m 10s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/00de2785590240948f0f61337d3d8159 : SUCCESS in 1h 37m 34s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"4d35f2b43c17f9ff24215d4b91881710543b8371","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-25 15:29:00.000000000","message":"Uploaded patch set 56.","accounts_in_message":[],"_revision_number":56},{"id":"b330bdfd69af2745c971111d9a5c9e5e3226f648","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-25 15:40:45.000000000","message":"Patch Set 56:\n\nPatchset 55 tests without setting the default value of VERBOSE in grenade.sh, and everything seems to work.\n\nPatchset 56 explores what happens when VERBOSE is not set in the job; nothing should break, but let\u0027s see.\n\nIf both pass, the grenade.sh change will be applied to the base \"greande zuulv3\" review.","accounts_in_message":[],"_revision_number":56},{"id":"89e690516f8f60a50c62299b21b4e9b869313f6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 15:53:35.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/406be2f21c4e4546a7b329ed4fcc24bc : EXCEPTION\n- grenade https://zuul.opendev.org/t/openstack/build/6a9e8b65a080419885f30c859c8824bc : EXCEPTION\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/23ae8aae830942a28adf912aa76bf17e : EXCEPTION\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/ee39f1dac0924b2ab94c49527338f034 : EXCEPTION\n- grenade-heat https://zuul.opendev.org/t/openstack/build/159d8c8277734efca2b30c4dc9864205 : EXCEPTION (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/f1ac8b50800240fda28fd5db167eeaec : EXCEPTION\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/58ef7ffbefe445ac9f5fbae361753b05 : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"c10133ab6cab4a68116fd9a5958fe8355db22814","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-25 16:17:11.000000000","message":"Patch Set 56:\n\nrecheck","accounts_in_message":[],"_revision_number":56},{"id":"15ad264a69d2d1fc3ac73d1fa1e904e2e8a4f3e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 19:16:14.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/867124b037f747aeb689f21ca7d4f6f7 : SUCCESS in 11m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/b413e2172a8c4b91a436ee2eb6a8e6ed : SUCCESS in 1h 10m 40s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/253d4cc456fd454a8515126e8e3df2e1 : SUCCESS in 1h 24m 28s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/c0bac269559c48369ee47aa6d1e8ffc4 : SUCCESS in 1h 37m 58s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/bc60e64ccf154ec1abc9d8d1dcb5d870 : SUCCESS in 1h 17m 31s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/f68ea461f6fd490a8c73ea90da2066b3 : SUCCESS in 1h 24m 05s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d4baad7c9cde46088a061cd5a3411b71 : RETRY_LIMIT in 13s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"c6b953bbe063393e6780682e43241127d132eef1","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-25 19:38:26.000000000","message":"Uploaded patch set 57.","accounts_in_message":[],"_revision_number":57},{"id":"37a4fd8073ff759ef5c7648afee05e1d5a81ef78","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-25 22:56:19.000000000","message":"Uploaded patch set 58: Patch Set 57 was rebased.","accounts_in_message":[],"_revision_number":58},{"id":"673900171dd5391ed138b5d829f876b7c6b00a77","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-25 23:29:24.000000000","message":"Uploaded patch set 59: Patch Set 58 was rebased.","accounts_in_message":[],"_revision_number":59},{"id":"92125766ea740f4a2d98b9ef944be08a8cbae7fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 01:18:15.000000000","message":"Patch Set 59: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc81ed9a3f7b4d1e8762d4f125b387c2 : SUCCESS in 6m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/38acf0c0b4574c21bf1870d47ca165e5 : FAILURE in 22m 59s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/76259afe29434bcba731d22908a1d02c : FAILURE in 31m 08s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/9104348c924745c4a20ad84b9d1e8234 : FAILURE in 38m 39s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/7ccc0a77ccd945a1b7f2c448550c8993 : SUCCESS in 1h 22m 38s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/595303d6dfe94ac09748eb40feddc2f4 : FAILURE in 33m 53s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/88617751e2d24f29aa31dd05b953bef4 : SUCCESS in 1h 47m 23s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"efb2d4b7664c98eb86e43726b0d2fedfe8a5303e","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-26 08:59:44.000000000","message":"Uploaded patch set 60: Patch Set 59 was rebased.","accounts_in_message":[],"_revision_number":60},{"id":"078fb22d7131e085a4df44ee7d08d3a040f3a7c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 10:59:24.000000000","message":"Patch Set 60: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7419ec78c8e8452eaca6f5f18053e496 : SUCCESS in 7m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/cd7dc38b450248aeac88c501a9d3f814 : SUCCESS in 1h 15m 13s\n- grenade-heat-test https://zuul.opendev.org/t/openstack/build/d7487e92026a4247a146f0be1c722fae : SUCCESS in 1h 25m 43s\n- grenade-heat-multinode-test https://zuul.opendev.org/t/openstack/build/6c6c8cb1cea444bc9178a1f3e1a15de9 : FAILURE in 1h 35m 08s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/81c19ae9e1804935a36c73a2d7b879a9 : SUCCESS in 1h 27m 28s (non-voting)\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/5e3f7dfcbc0d40d9b8b77c7be7452649 : SUCCESS in 1h 21m 43s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/df10025d73004984adee31e5f1a8d3d2 : SUCCESS in 1h 57m 58s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"bc4a98174ae5a8e5a5af1a16ce208f38cdb9bbf5","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-13 07:49:25.000000000","message":"Abandoned\n\nNot needed anymore; folded into https://review.opendev.org/#/c/726790","accounts_in_message":[],"_revision_number":60}],"current_revision_number":60,"current_revision":"21e6703dd80ec554e83fc223038749a2fd807349","revisions":{"a35a60f3cc66ccda90bf5db16d604ead286be46d":{"kind":"REWORK","_number":1,"created":"2019-02-27 18:58:22.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/1"}}},"commit":{"parents":[{"commit":"503b1b14722b992995d2d1ac35be33764edadf39","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/503b1b14722b992995d2d1ac35be33764edadf39"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-02-27 18:41:37.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-02-27 18:43:57.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a35a60f3cc66ccda90bf5db16d604ead286be46d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a35a60f3cc66ccda90bf5db16d604ead286be46d"}]},"branch":"refs/heads/master"},"66990337917d24ccbdf3f74dc050be2d25ff225d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-02-28 11:56:16.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/2"}}},"commit":{"parents":[{"commit":"96dd4b2d958da28a0c40b76929cb5a184d45e1b1","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/96dd4b2d958da28a0c40b76929cb5a184d45e1b1"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-02-27 18:41:37.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-02-28 11:56:04.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/66990337917d24ccbdf3f74dc050be2d25ff225d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/66990337917d24ccbdf3f74dc050be2d25ff225d"}]},"branch":"refs/heads/master"},"31a7bafb76b1a373ac12ba5d5513a2ae667bda25":{"kind":"NO_CHANGE","_number":3,"created":"2019-02-28 13:53:41.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/3"}}},"commit":{"parents":[{"commit":"ff5debee24cb948d09b70e127582af7d422fdffd","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/ff5debee24cb948d09b70e127582af7d422fdffd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-02-27 18:41:37.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-02-28 13:36:10.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/31a7bafb76b1a373ac12ba5d5513a2ae667bda25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/31a7bafb76b1a373ac12ba5d5513a2ae667bda25"}]},"branch":"refs/heads/master"},"b37c676f074fb50681769eb6b1a2e4bbe37f4e0e":{"kind":"REWORK","_number":4,"created":"2019-03-04 11:46:45.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/4"}}},"commit":{"parents":[{"commit":"1c8713aac25aae7bb3b380eed0c8949ea5828a38","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/1c8713aac25aae7bb3b380eed0c8949ea5828a38"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-01 15:59:48.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-04 11:46:25.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/b37c676f074fb50681769eb6b1a2e4bbe37f4e0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/b37c676f074fb50681769eb6b1a2e4bbe37f4e0e"}]},"branch":"refs/heads/master"},"c11de343821d576255dfe1873ccf0b5fcf14f1df":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-03-04 13:27:38.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/5"}}},"commit":{"parents":[{"commit":"d28892f8496dd2f7d0f577ffddce143338fd7e50","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d28892f8496dd2f7d0f577ffddce143338fd7e50"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-01 15:59:48.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-04 13:27:24.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/c11de343821d576255dfe1873ccf0b5fcf14f1df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/c11de343821d576255dfe1873ccf0b5fcf14f1df"}]},"branch":"refs/heads/master"},"a5093d14f34a2b001f00636ce9f52bf9a18006ef":{"kind":"REWORK","_number":6,"created":"2019-03-04 19:09:00.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/6"}}},"commit":{"parents":[{"commit":"39dca9b4029fc44a16730271043bc86714eb22ba","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/39dca9b4029fc44a16730271043bc86714eb22ba"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-04 18:23:47.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-04 19:08:41.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a5093d14f34a2b001f00636ce9f52bf9a18006ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a5093d14f34a2b001f00636ce9f52bf9a18006ef"}]},"branch":"refs/heads/master"},"ab6a5cdd3c84388b2cb3cd1327e6919be3aa4c5f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-03-06 00:50:55.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/7"}}},"commit":{"parents":[{"commit":"77174a8564ca570f56879208c40c4895d4263f0c","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/77174a8564ca570f56879208c40c4895d4263f0c"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-04 18:23:47.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-06 00:50:13.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/ab6a5cdd3c84388b2cb3cd1327e6919be3aa4c5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/ab6a5cdd3c84388b2cb3cd1327e6919be3aa4c5f"}]},"branch":"refs/heads/master"},"bb7ac3f3e55f8c6fa24353f1e953ef5d56fdcc8c":{"kind":"REWORK","_number":8,"created":"2019-03-06 11:59:38.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/8"}}},"commit":{"parents":[{"commit":"77174a8564ca570f56879208c40c4895d4263f0c","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/77174a8564ca570f56879208c40c4895d4263f0c"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-04 18:23:47.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-06 11:59:38.000000000","tz":0},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/bb7ac3f3e55f8c6fa24353f1e953ef5d56fdcc8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/bb7ac3f3e55f8c6fa24353f1e953ef5d56fdcc8c"}]},"branch":"refs/heads/master"},"3cfb233acb49975cdcec3e58b998829db188a619":{"kind":"REWORK","_number":9,"created":"2019-03-12 21:46:18.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/9"}}},"commit":{"parents":[{"commit":"9c7031168072656ed9eb5f475fb01384e5b288f3","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/9c7031168072656ed9eb5f475fb01384e5b288f3"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-12 21:29:48.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-12 21:29:48.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/3cfb233acb49975cdcec3e58b998829db188a619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/3cfb233acb49975cdcec3e58b998829db188a619"}]},"branch":"refs/heads/master"},"44b962fe369f910a5793b5ae4baa1c9f0eff09df":{"kind":"REWORK","_number":10,"created":"2019-03-21 10:54:39.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/10"}}},"commit":{"parents":[{"commit":"b2eeadb0f04dca45fb08fe3875d596ae1e9eb37d","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/b2eeadb0f04dca45fb08fe3875d596ae1e9eb37d"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-14 16:38:51.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-21 10:54:30.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/44b962fe369f910a5793b5ae4baa1c9f0eff09df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/44b962fe369f910a5793b5ae4baa1c9f0eff09df"}]},"branch":"refs/heads/master"},"94c1de871b8c7e775dd4d9793617f3d55eb89a77":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-03-21 12:40:15.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/11"}}},"commit":{"parents":[{"commit":"196c3a21311de1ece85740c99062ade16ba00d93","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/196c3a21311de1ece85740c99062ade16ba00d93"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-14 16:38:51.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-21 12:40:04.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/94c1de871b8c7e775dd4d9793617f3d55eb89a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/94c1de871b8c7e775dd4d9793617f3d55eb89a77"}]},"branch":"refs/heads/master"},"f2197a4076b4ec9780139a9e5f52e22a429f2a64":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":12,"created":"2019-03-21 17:13:35.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/12"}}},"commit":{"parents":[{"commit":"a0a5dc9efd091b19ea079e25877850adf32e6e01","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a0a5dc9efd091b19ea079e25877850adf32e6e01"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-21 17:13:08.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-21 17:13:08.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nDepends-On: https://review.openstack.org/645239\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/f2197a4076b4ec9780139a9e5f52e22a429f2a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/f2197a4076b4ec9780139a9e5f52e22a429f2a64"}]},"branch":"refs/heads/master"},"36bb29c7c013f2e4cf1f52b62ead63ac7870a8c9":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-03-26 19:01:07.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/13"}}},"commit":{"parents":[{"commit":"72fce37e493ee2f350a5d66066f50d9238517318","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/72fce37e493ee2f350a5d66066f50d9238517318"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-21 17:13:08.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-26 19:00:37.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nDepends-On: https://review.openstack.org/645239\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/36bb29c7c013f2e4cf1f52b62ead63ac7870a8c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/36bb29c7c013f2e4cf1f52b62ead63ac7870a8c9"}]},"branch":"refs/heads/master"},"54f8ab9545e1136c3ddadff5c4c938f01cd17829":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-03-26 21:03:16.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/14"}}},"commit":{"parents":[{"commit":"daebdcf9c4d56ff40fec305ce03ad7edbe67faae","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/daebdcf9c4d56ff40fec305ce03ad7edbe67faae"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-21 17:13:08.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-26 21:02:52.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nDepends-On: https://review.openstack.org/645239\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/54f8ab9545e1136c3ddadff5c4c938f01cd17829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/54f8ab9545e1136c3ddadff5c4c938f01cd17829"}]},"branch":"refs/heads/master"},"d2a0b6782a7f14f6c414e64dd143130f60a68bd0":{"kind":"REWORK","_number":15,"created":"2019-03-27 00:50:44.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/15"}}},"commit":{"parents":[{"commit":"bc70ac2943464553f598e15dddec3351a97a2a0a","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/bc70ac2943464553f598e15dddec3351a97a2a0a"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-27 00:45:52.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-27 00:50:30.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nDepends-On: https://review.openstack.org/645239\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d2a0b6782a7f14f6c414e64dd143130f60a68bd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d2a0b6782a7f14f6c414e64dd143130f60a68bd0"}]},"branch":"refs/heads/master"},"07104d5cb96288206c455aabb858276867d2b895":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-03-27 18:04:09.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/16"}}},"commit":{"parents":[{"commit":"048d46368247290d23e61ea8e78e2e0b73dda882","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/048d46368247290d23e61ea8e78e2e0b73dda882"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-27 00:45:52.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-27 18:03:29.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nDepends-On: https://review.openstack.org/645239\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/07104d5cb96288206c455aabb858276867d2b895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/07104d5cb96288206c455aabb858276867d2b895"}]},"branch":"refs/heads/master"},"87adf223ba34b3880679c62fce231a2605d6647e":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-03-27 19:42:32.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/17"}}},"commit":{"parents":[{"commit":"a44afd72b2aeacd0ee0ec8e0810b1ce5e2e8dcd1","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a44afd72b2aeacd0ee0ec8e0810b1ce5e2e8dcd1"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-27 00:45:52.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-27 19:42:19.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nDepends-On: https://review.openstack.org/645239\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/87adf223ba34b3880679c62fce231a2605d6647e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/87adf223ba34b3880679c62fce231a2605d6647e"}]},"branch":"refs/heads/master"},"7c19fe46685d48c8291c34f4740789f1187769c0":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2019-03-28 10:03:38.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/18"}}},"commit":{"parents":[{"commit":"1eaea1a089c5fa8f42efa37b42d60dadf21b16b5","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/1eaea1a089c5fa8f42efa37b42d60dadf21b16b5"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-27 00:45:52.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-28 10:02:55.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nDepends-On: https://review.openstack.org/645239\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/7c19fe46685d48c8291c34f4740789f1187769c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/7c19fe46685d48c8291c34f4740789f1187769c0"}]},"branch":"refs/heads/master"},"0567e0489490a730eb3dbb9e78c9dc4ecf068e31":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2019-03-28 13:47:39.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/19"}}},"commit":{"parents":[{"commit":"08e0f3015d1c8810223094bbd608a88d55388d2a","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/08e0f3015d1c8810223094bbd608a88d55388d2a"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-27 00:45:52.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-28 13:47:27.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nDepends-On: https://review.openstack.org/645239\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/0567e0489490a730eb3dbb9e78c9dc4ecf068e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/0567e0489490a730eb3dbb9e78c9dc4ecf068e31"}]},"branch":"refs/heads/master"},"57afe5881b9d39f81e75bf5b3acbe70501ebd897":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":20,"created":"2019-03-29 10:45:30.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/20"}}},"commit":{"parents":[{"commit":"a5a2dabcc80c07105d69d1152e7c653566c0c2a7","subject":"WIP Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a5a2dabcc80c07105d69d1152e7c653566c0c2a7"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 10:33:55.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 10:33:55.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nDepends-On: https://review.openstack.org/648658\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/57afe5881b9d39f81e75bf5b3acbe70501ebd897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/57afe5881b9d39f81e75bf5b3acbe70501ebd897"}]},"branch":"refs/heads/master"},"ab2a2441d82d90e66b71bfbdc3d652a6c5ffd5b9":{"kind":"REWORK","_number":21,"created":"2019-03-29 15:31:42.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/21"}}},"commit":{"parents":[{"commit":"2d2d6acd2959b2bfdd97b77085b2df34f8146213","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/2d2d6acd2959b2bfdd97b77085b2df34f8146213"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 15:18:46.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 15:20:16.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/ab2a2441d82d90e66b71bfbdc3d652a6c5ffd5b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/ab2a2441d82d90e66b71bfbdc3d652a6c5ffd5b9"}]},"branch":"refs/heads/master"},"e7b56933b5013fd30a390f2c044fe2c1bfff9386":{"kind":"REWORK","_number":22,"created":"2019-03-29 20:44:34.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/22"}}},"commit":{"parents":[{"commit":"2d2d6acd2959b2bfdd97b77085b2df34f8146213","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/2d2d6acd2959b2bfdd97b77085b2df34f8146213"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 20:44:23.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 20:44:23.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/e7b56933b5013fd30a390f2c044fe2c1bfff9386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/e7b56933b5013fd30a390f2c044fe2c1bfff9386"}]},"branch":"refs/heads/master"},"0ac26356167a9d9f1d6fade3f3d0f9fb1d2b5b15":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2019-03-29 22:50:08.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/23"}}},"commit":{"parents":[{"commit":"32a699ab56431404509c67d56b347c4a580cfebf","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/32a699ab56431404509c67d56b347c4a580cfebf"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 20:44:23.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 22:49:52.000000000","tz":60},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/0ac26356167a9d9f1d6fade3f3d0f9fb1d2b5b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/0ac26356167a9d9f1d6fade3f3d0f9fb1d2b5b15"}]},"branch":"refs/heads/master"},"d4117993ca99ca14880e80d99d5552bd3ab237d8":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2019-04-01 08:15:44.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/24"}}},"commit":{"parents":[{"commit":"ecc45de32f502a5c365e834755d25a72741ea056","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/ecc45de32f502a5c365e834755d25a72741ea056"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 20:44:23.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-01 08:15:34.000000000","tz":120},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d4117993ca99ca14880e80d99d5552bd3ab237d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d4117993ca99ca14880e80d99d5552bd3ab237d8"}]},"branch":"refs/heads/master"},"47db8c72e529f2dd9dff0df80d7e48807664b703":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2019-04-01 09:18:35.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/25"}}},"commit":{"parents":[{"commit":"7e4508d6614ded5fa2e70e0526423565ad1b933b","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/7e4508d6614ded5fa2e70e0526423565ad1b933b"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 20:44:23.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-01 09:18:09.000000000","tz":120},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/47db8c72e529f2dd9dff0df80d7e48807664b703"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/47db8c72e529f2dd9dff0df80d7e48807664b703"}]},"branch":"refs/heads/master"},"d49b863db6cda68008ef194916b6d2c003df1839":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2019-04-02 10:35:05.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/26"}}},"commit":{"parents":[{"commit":"8eb51e140fd7bc32ae15ded06470c6f8b5a073cd","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/8eb51e140fd7bc32ae15ded06470c6f8b5a073cd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 20:44:23.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-02 10:34:51.000000000","tz":120},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d49b863db6cda68008ef194916b6d2c003df1839"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d49b863db6cda68008ef194916b6d2c003df1839"}]},"branch":"refs/heads/master"},"ddc199508b4e31f2bb32b4fe168f86aca01938fd":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2019-04-02 15:47:05.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/27"}}},"commit":{"parents":[{"commit":"5f3dcf27e1cdf5350374f9d4e2e3b778e22ce88c","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/5f3dcf27e1cdf5350374f9d4e2e3b778e22ce88c"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 20:44:23.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-02 15:46:53.000000000","tz":120},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/ddc199508b4e31f2bb32b4fe168f86aca01938fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/ddc199508b4e31f2bb32b4fe168f86aca01938fd"}]},"branch":"refs/heads/master"},"adf886ca0397ef9ad0bf6c2378304e07e48ba1c5":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2019-04-03 07:47:06.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/28"}}},"commit":{"parents":[{"commit":"fec32ceae5807a28e0022bfb19126230857ea896","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/fec32ceae5807a28e0022bfb19126230857ea896"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 20:44:23.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-03 07:46:55.000000000","tz":120},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/adf886ca0397ef9ad0bf6c2378304e07e48ba1c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/adf886ca0397ef9ad0bf6c2378304e07e48ba1c5"}]},"branch":"refs/heads/master"},"516b1fecdd8859ec4872891c182745a4338fffdc":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2019-04-03 09:07:09.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/29"}}},"commit":{"parents":[{"commit":"93eb5c07402967b3cff44d2ab1b90173030c4f85","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/93eb5c07402967b3cff44d2ab1b90173030c4f85"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-03-29 20:44:23.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-03 09:06:58.000000000","tz":120},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/516b1fecdd8859ec4872891c182745a4338fffdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/516b1fecdd8859ec4872891c182745a4338fffdc"}]},"branch":"refs/heads/master"},"746ed415130201f275a72eba1fa92de59061f510":{"kind":"REWORK","_number":30,"created":"2019-04-04 16:42:41.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/30"}}},"commit":{"parents":[{"commit":"6a23e8b5c686695d4e8135e4bf6cb0c5dc03965d","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/6a23e8b5c686695d4e8135e4bf6cb0c5dc03965d"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-04 16:42:29.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-04 16:42:29.000000000","tz":120},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/746ed415130201f275a72eba1fa92de59061f510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/746ed415130201f275a72eba1fa92de59061f510"}]},"branch":"refs/heads/master"},"d5b00ea531f42c6f45bf0bb124f4f723a32d606e":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2019-04-04 19:32:35.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/31"}}},"commit":{"parents":[{"commit":"6fa67c8e95e581636c913fabbcf28bc3c78d49dd","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/6fa67c8e95e581636c913fabbcf28bc3c78d49dd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-04 16:42:29.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-04 19:32:11.000000000","tz":120},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d5b00ea531f42c6f45bf0bb124f4f723a32d606e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d5b00ea531f42c6f45bf0bb124f4f723a32d606e"}]},"branch":"refs/heads/master"},"c5728a4ffdec7a19744fa4e9844204d0ff5b5c9f":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2019-04-04 21:30:15.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/32"}}},"commit":{"parents":[{"commit":"37dee7964f5efa947812ee3305a4c6c5dce7df79","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/37dee7964f5efa947812ee3305a4c6c5dce7df79"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-04 16:42:29.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-04 21:30:02.000000000","tz":120},"subject":"DNM - grenade-heat-test job","message":"DNM - grenade-heat-test job\n\nShould be moved to the heat repository\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/c5728a4ffdec7a19744fa4e9844204d0ff5b5c9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/c5728a4ffdec7a19744fa4e9844204d0ff5b5c9f"}]},"branch":"refs/heads/master"},"eff41e16ca52c643d1c5e2eaafb4bf4e2c870fc1":{"kind":"REWORK","_number":33,"created":"2019-04-05 07:55:51.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/33"}}},"commit":{"parents":[{"commit":"37dee7964f5efa947812ee3305a4c6c5dce7df79","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/37dee7964f5efa947812ee3305a4c6c5dce7df79"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-05 07:53:41.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-05 07:53:41.000000000","tz":120},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/eff41e16ca52c643d1c5e2eaafb4bf4e2c870fc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/eff41e16ca52c643d1c5e2eaafb4bf4e2c870fc1"}]},"branch":"refs/heads/master"},"85e29dbe0f516cda37d1364125149bd453ca6181":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2019-04-20 13:15:47.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/34"}}},"commit":{"parents":[{"commit":"f752fc75d362d1db4601378a6e04b7131e623fba","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/f752fc75d362d1db4601378a6e04b7131e623fba"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-05 07:53:41.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-20 13:14:58.000000000","tz":120},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/85e29dbe0f516cda37d1364125149bd453ca6181"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/85e29dbe0f516cda37d1364125149bd453ca6181"}]},"branch":"refs/heads/master"},"6e487f729ca8bea0282500a697303070f06c7e73":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2019-04-20 13:25:49.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/35"}}},"commit":{"parents":[{"commit":"a4c3152a860a4012306020c5a9f9db50c361c2c5","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a4c3152a860a4012306020c5a9f9db50c361c2c5"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-05 07:53:41.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-20 13:25:37.000000000","tz":120},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/6e487f729ca8bea0282500a697303070f06c7e73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/6e487f729ca8bea0282500a697303070f06c7e73"}]},"branch":"refs/heads/master"},"4a82793194fb7acb240963e96eb44164f5fbc22e":{"kind":"NO_CHANGE","_number":36,"created":"2019-04-20 17:58:31.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/36"}}},"commit":{"parents":[{"commit":"54d2f4a8a2394ca01df0fa27858a01caaf95a3f1","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/54d2f4a8a2394ca01df0fa27858a01caaf95a3f1"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-05 07:53:41.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-20 17:58:03.000000000","tz":120},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/4a82793194fb7acb240963e96eb44164f5fbc22e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/4a82793194fb7acb240963e96eb44164f5fbc22e"}]},"branch":"refs/heads/master"},"951bf8b756f6ebc52cd1bb6dba6f3980bf32e49f":{"kind":"NO_CHANGE","_number":37,"created":"2019-05-23 17:05:54.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/37"}}},"commit":{"parents":[{"commit":"392d9c6c9c5996441a7fe6391f37bf54a8b4f7d7","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/392d9c6c9c5996441a7fe6391f37bf54a8b4f7d7"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-05 07:53:41.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-05-23 17:05:38.000000000","tz":120},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/951bf8b756f6ebc52cd1bb6dba6f3980bf32e49f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/951bf8b756f6ebc52cd1bb6dba6f3980bf32e49f"}]},"branch":"refs/heads/master"},"5930bf4ab68ef91b47f7158e4efd99015db324b7":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2019-07-25 16:21:22.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/38"}}},"commit":{"parents":[{"commit":"e9f7d252c001317509859c1735c3216d8a578f85","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/e9f7d252c001317509859c1735c3216d8a578f85"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-05 07:53:41.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-07-25 16:18:43.000000000","tz":120},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/5930bf4ab68ef91b47f7158e4efd99015db324b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/5930bf4ab68ef91b47f7158e4efd99015db324b7"}]},"branch":"refs/heads/master"},"b54e9fe72b70fdcb4398b208b732198ba3ee3b07":{"kind":"REWORK","_number":39,"created":"2019-12-17 17:27:02.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/39"}}},"commit":{"parents":[{"commit":"24c01ca65ef85c2222961dd3bf23b190b71c177e","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/24c01ca65ef85c2222961dd3bf23b190b71c177e"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-05 07:53:41.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-12-17 17:26:13.000000000","tz":60},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/b54e9fe72b70fdcb4398b208b732198ba3ee3b07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/b54e9fe72b70fdcb4398b208b732198ba3ee3b07"}]},"branch":"refs/heads/master"},"890ffa0f5f89bc52585c41a617f5f613ffa051cf":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2019-12-17 19:26:05.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/40"}}},"commit":{"parents":[{"commit":"cd2e2389c0aec54b78ceef3d756cabd1f97f971f","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/cd2e2389c0aec54b78ceef3d756cabd1f97f971f"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-05 07:53:41.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-12-17 19:25:43.000000000","tz":60},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/890ffa0f5f89bc52585c41a617f5f613ffa051cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/890ffa0f5f89bc52585c41a617f5f613ffa051cf"}]},"branch":"refs/heads/master"},"f910016d567d600d5120461fa866870804e5bcf3":{"kind":"REWORK","_number":41,"created":"2019-12-17 21:00:36.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/41"}}},"commit":{"parents":[{"commit":"51551d854ef557c3f555e9f415d92d65925b3507","subject":"Setup grenade-base, grenade, grenade-{py3,postgresql}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/51551d854ef557c3f555e9f415d92d65925b3507"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-05 07:53:41.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-12-17 20:59:18.000000000","tz":60},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/f910016d567d600d5120461fa866870804e5bcf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/f910016d567d600d5120461fa866870804e5bcf3"}]},"branch":"refs/heads/master"},"0d834a6539274b4757e4cea22491d3bdb5568439":{"kind":"REWORK","_number":42,"created":"2019-12-23 15:10:38.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/42"}}},"commit":{"parents":[{"commit":"9c54407533a8cb633d8868b877e76730a54aa900","subject":"Setup grenade-base, grenade, grenade-{postgresql,multinode}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/9c54407533a8cb633d8868b877e76730a54aa900"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-05 07:53:41.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-12-23 15:10:22.000000000","tz":60},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/0d834a6539274b4757e4cea22491d3bdb5568439"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/0d834a6539274b4757e4cea22491d3bdb5568439"}]},"branch":"refs/heads/master"},"ec42bf9b63b220d1fef5af673b26b219e6e3d2c7":{"kind":"TRIVIAL_REBASE","_number":43,"created":"2020-01-20 14:57:24.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/43"}}},"commit":{"parents":[{"commit":"f586cd11731cf763a07694e468a983b45b90bb1b","subject":"Setup grenade-base, grenade, grenade-{postgresql,multinode}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/f586cd11731cf763a07694e468a983b45b90bb1b"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-05 07:53:41.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-01-17 16:09:50.000000000","tz":60},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/ec42bf9b63b220d1fef5af673b26b219e6e3d2c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/ec42bf9b63b220d1fef5af673b26b219e6e3d2c7"}]},"branch":"refs/heads/master"},"06f400882517d09bfc04d8a0942f29c845a8fc9f":{"kind":"NO_CODE_CHANGE","_number":44,"created":"2020-01-28 13:49:47.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/44"}}},"commit":{"parents":[{"commit":"f586cd11731cf763a07694e468a983b45b90bb1b","subject":"Setup grenade-base, grenade, grenade-{postgresql,multinode}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/f586cd11731cf763a07694e468a983b45b90bb1b"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-01-28 13:47:43.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-01-28 13:47:43.000000000","tz":60},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nDepends-On: https://review.opendev.org/704543\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/06f400882517d09bfc04d8a0942f29c845a8fc9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/06f400882517d09bfc04d8a0942f29c845a8fc9f"}]},"branch":"refs/heads/master"},"eb2b2442491b6b044ef3f2414ec06bc82955bd29":{"kind":"NO_CODE_CHANGE","_number":45,"created":"2020-02-07 14:47:37.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/45"}}},"commit":{"parents":[{"commit":"f586cd11731cf763a07694e468a983b45b90bb1b","subject":"Setup grenade-base, grenade, grenade-{postgresql,multinode}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/f586cd11731cf763a07694e468a983b45b90bb1b"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-02-07 14:46:50.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-02-07 14:46:50.000000000","tz":60},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nDepends-On: https://review.opendev.org/706492\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/eb2b2442491b6b044ef3f2414ec06bc82955bd29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/eb2b2442491b6b044ef3f2414ec06bc82955bd29"}]},"branch":"refs/heads/master"},"5e583faf0e10df633df0a74b2dcacae35ebdc4d4":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2020-03-02 11:51:43.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/46"}}},"commit":{"parents":[{"commit":"28c004854f7cd2e06b066afdd40ec30fe406e860","subject":"Setup grenade-base, grenade, grenade-{postgresql,multinode}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/28c004854f7cd2e06b066afdd40ec30fe406e860"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-02-07 14:46:50.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-02 11:07:29.000000000","tz":60},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nDepends-On: https://review.opendev.org/706492\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/5e583faf0e10df633df0a74b2dcacae35ebdc4d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/5e583faf0e10df633df0a74b2dcacae35ebdc4d4"}]},"branch":"refs/heads/master"},"868c77cc01066b470b3269472aac33ae80016c3f":{"kind":"REWORK","_number":47,"created":"2020-03-02 15:26:06.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/47"}}},"commit":{"parents":[{"commit":"fb6c243ae63ffda3ccb870971933cc3a8d6ed24f","subject":"Setup grenade-base, grenade, grenade-{postgresql,multinode}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/fb6c243ae63ffda3ccb870971933cc3a8d6ed24f"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-02-07 14:46:50.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-02 11:50:40.000000000","tz":60},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nDepends-On: https://review.opendev.org/706492\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/868c77cc01066b470b3269472aac33ae80016c3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/868c77cc01066b470b3269472aac33ae80016c3f"}]},"branch":"refs/heads/master"},"3ed7131f1319b9d0c3b6e98138f2ce2302e64f81":{"kind":"TRIVIAL_REBASE","_number":48,"created":"2020-03-02 17:20:52.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/48"}}},"commit":{"parents":[{"commit":"e1f56110dc4e908e48fece114ab450718b104acd","subject":"Setup grenade-base, grenade, grenade-{postgresql,multinode}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/e1f56110dc4e908e48fece114ab450718b104acd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-02-07 14:46:50.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-02 17:20:36.000000000","tz":60},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nDepends-On: https://review.opendev.org/706492\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/3ed7131f1319b9d0c3b6e98138f2ce2302e64f81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/3ed7131f1319b9d0c3b6e98138f2ce2302e64f81"}]},"branch":"refs/heads/master"},"62cb8d3552c7b5d171e33c9cb85013fb454b24cc":{"kind":"REWORK","_number":49,"created":"2020-03-18 22:20:35.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/49"}}},"commit":{"parents":[{"commit":"571d2eb83243d6779a1713f422754b1932524bcd","subject":"Setup native grenade Zuul v3 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/571d2eb83243d6779a1713f422754b1932524bcd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-02-07 14:46:50.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-18 22:20:07.000000000","tz":60},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nDepends-On: https://review.opendev.org/706492\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/62cb8d3552c7b5d171e33c9cb85013fb454b24cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/62cb8d3552c7b5d171e33c9cb85013fb454b24cc"}]},"branch":"refs/heads/master"},"3ac3c5b267643a9bf98bad763a405861a8020ebd":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2020-03-18 23:11:34.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/50"}}},"commit":{"parents":[{"commit":"4234681699b91bf26f040f30d0393500b350bf26","subject":"Setup native grenade Zuul v3 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/4234681699b91bf26f040f30d0393500b350bf26"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-02-07 14:46:50.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-18 23:09:22.000000000","tz":60},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nDepends-On: https://review.opendev.org/706492\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/3ac3c5b267643a9bf98bad763a405861a8020ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/3ac3c5b267643a9bf98bad763a405861a8020ebd"}]},"branch":"refs/heads/master"},"6d903502bec5b18b599415f8fb5bcb728cb168d1":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2020-04-06 09:53:55.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/51"}}},"commit":{"parents":[{"commit":"1ede47438c5999717feeb5eb6cf59c60e2994d0c","subject":"Setup native grenade Zuul v3 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/1ede47438c5999717feeb5eb6cf59c60e2994d0c"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-02-07 14:46:50.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-06 09:53:41.000000000","tz":120},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nDepends-On: https://review.opendev.org/706492\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/6d903502bec5b18b599415f8fb5bcb728cb168d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/6d903502bec5b18b599415f8fb5bcb728cb168d1"}]},"branch":"refs/heads/master"},"a1cc604f0a605061970f7eaafd0d2ce851b991f6":{"kind":"TRIVIAL_REBASE","_number":52,"created":"2020-04-23 23:37:08.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/52"}}},"commit":{"parents":[{"commit":"4f1e0f130326b890c90e9299fd4458a7742d8e33","subject":"Setup native grenade Zuul v3 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/4f1e0f130326b890c90e9299fd4458a7742d8e33"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-02-07 14:46:50.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-23 23:36:55.000000000","tz":120},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nDepends-On: https://review.opendev.org/706492\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a1cc604f0a605061970f7eaafd0d2ce851b991f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a1cc604f0a605061970f7eaafd0d2ce851b991f6"}]},"branch":"refs/heads/master"},"8fa2ddf61de3b0437fc52590cbfcab8caa158641":{"kind":"TRIVIAL_REBASE","_number":53,"created":"2020-04-23 23:47:40.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/53"}}},"commit":{"parents":[{"commit":"3152d6ea5b18c766fab7a0bbb2af97c49b60c865","subject":"Setup native grenade Zuul v3 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/3152d6ea5b18c766fab7a0bbb2af97c49b60c865"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-02-07 14:46:50.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-23 23:46:54.000000000","tz":120},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nDepends-On: https://review.opendev.org/706492\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/8fa2ddf61de3b0437fc52590cbfcab8caa158641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/8fa2ddf61de3b0437fc52590cbfcab8caa158641"}]},"branch":"refs/heads/master"},"a36011f85728ecc9d0a54a5acaf76624118e9d49":{"kind":"TRIVIAL_REBASE","_number":54,"created":"2020-04-24 13:44:31.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/54"}}},"commit":{"parents":[{"commit":"73ff905ce1520ff4d2976db6eaf84cf5013695e5","subject":"Setup native grenade Zuul v3 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/73ff905ce1520ff4d2976db6eaf84cf5013695e5"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-02-07 14:46:50.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-24 13:37:46.000000000","tz":120},"subject":"DNM - grenade test jobs (heat, neutron)","message":"DNM - grenade test jobs (heat, neutron)\n\nShould be moved to the heat and neutron repository respectively.\n\nDepends-On: https://review.opendev.org/706492\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a36011f85728ecc9d0a54a5acaf76624118e9d49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a36011f85728ecc9d0a54a5acaf76624118e9d49"}]},"branch":"refs/heads/master"},"8fbef2b3c533c03d287ac96b4b0bdbc15697f233":{"kind":"REWORK","_number":55,"created":"2020-04-25 13:40:17.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/55"}}},"commit":{"parents":[{"commit":"73ff905ce1520ff4d2976db6eaf84cf5013695e5","subject":"Setup native grenade Zuul v3 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/73ff905ce1520ff4d2976db6eaf84cf5013695e5"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-25 13:39:10.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-25 13:39:10.000000000","tz":120},"subject":"DNM - grenade test jobs (heat)","message":"DNM - grenade test jobs (heat)\n\nShould be moved to the heat repository.\n\nAlso check whether the VERBOSE variable is still needed.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/8fbef2b3c533c03d287ac96b4b0bdbc15697f233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/8fbef2b3c533c03d287ac96b4b0bdbc15697f233"}]},"branch":"refs/heads/master"},"fd5f18be5c3791910ec2ac7a513a33adba2afb7b":{"kind":"REWORK","_number":56,"created":"2020-04-25 15:29:00.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/56"}}},"commit":{"parents":[{"commit":"73ff905ce1520ff4d2976db6eaf84cf5013695e5","subject":"Setup native grenade Zuul v3 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/73ff905ce1520ff4d2976db6eaf84cf5013695e5"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-25 15:28:29.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-25 15:28:29.000000000","tz":120},"subject":"DNM - grenade test jobs (heat)","message":"DNM - grenade test jobs (heat)\n\nShould be moved to the heat repository.\n\nAlso check whether the VERBOSE variable is still needed.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/fd5f18be5c3791910ec2ac7a513a33adba2afb7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/fd5f18be5c3791910ec2ac7a513a33adba2afb7b"}]},"branch":"refs/heads/master"},"86b9a63654e237b2d0cc8d79b2a8cb13948778f5":{"kind":"REWORK","_number":57,"created":"2020-04-25 19:38:26.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/57"}}},"commit":{"parents":[{"commit":"9069bf671dffbe9c592a6d8fb233eb40beb4a692","subject":"Setup native grenade Zuul v3 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/9069bf671dffbe9c592a6d8fb233eb40beb4a692"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-25 13:39:10.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-25 15:35:41.000000000","tz":120},"subject":"DNM - grenade test jobs (heat)","message":"DNM - grenade test jobs (heat)\n\nShould be moved to the heat repository.\n\nAlso check whether the VERBOSE variable is still needed.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/86b9a63654e237b2d0cc8d79b2a8cb13948778f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/86b9a63654e237b2d0cc8d79b2a8cb13948778f5"}]},"branch":"refs/heads/master"},"0c8cfc11c6424627b438e5a1d88fa26fa6c71ae1":{"kind":"TRIVIAL_REBASE","_number":58,"created":"2020-04-25 22:56:19.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/58"}}},"commit":{"parents":[{"commit":"d46e3f8fb499a35c79c4ecae7b16807cda3f39f0","subject":"Setup native grenade Zuul v3 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d46e3f8fb499a35c79c4ecae7b16807cda3f39f0"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-25 13:39:10.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-25 20:52:00.000000000","tz":120},"subject":"DNM - grenade test jobs (heat)","message":"DNM - grenade test jobs (heat)\n\nShould be moved to the heat repository.\n\nAlso check whether the VERBOSE variable is still needed.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/0c8cfc11c6424627b438e5a1d88fa26fa6c71ae1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/0c8cfc11c6424627b438e5a1d88fa26fa6c71ae1"}]},"branch":"refs/heads/master"},"a730009b9e411ace289fc197ed217d5d36d29cc0":{"kind":"TRIVIAL_REBASE","_number":59,"created":"2020-04-25 23:29:24.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/59"}}},"commit":{"parents":[{"commit":"6f0bbe9bf99b2c46f957e33b386f9736f40f393b","subject":"Setup native grenade Zuul v3 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/6f0bbe9bf99b2c46f957e33b386f9736f40f393b"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-25 13:39:10.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-25 23:29:14.000000000","tz":120},"subject":"DNM - grenade test jobs (heat)","message":"DNM - grenade test jobs (heat)\n\nShould be moved to the heat repository.\n\nAlso check whether the VERBOSE variable is still needed.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a730009b9e411ace289fc197ed217d5d36d29cc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/a730009b9e411ace289fc197ed217d5d36d29cc0"}]},"branch":"refs/heads/master"},"21e6703dd80ec554e83fc223038749a2fd807349":{"kind":"TRIVIAL_REBASE","_number":60,"created":"2020-04-26 08:59:44.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/74/639774/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/639774/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/639774/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/639774/60"}}},"commit":{"parents":[{"commit":"514ddc8eaa4a7a0c77d360ef449e506cc2250e48","subject":"Setup native grenade Zuul v3 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/514ddc8eaa4a7a0c77d360ef449e506cc2250e48"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-25 13:39:10.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-26 08:59:17.000000000","tz":120},"subject":"DNM - grenade test jobs (heat)","message":"DNM - grenade test jobs (heat)\n\nShould be moved to the heat repository.\n\nAlso check whether the VERBOSE variable is still needed.\n\nChange-Id: I14427b087b224f66f65c379f72f2c0e5b9ba939a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/21e6703dd80ec554e83fc223038749a2fd807349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/21e6703dd80ec554e83fc223038749a2fd807349"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
