)]}'
{"id":"openstack%2Fgrenade~682975","triplet_id":"openstack%2Fgrenade~stable%2Fqueens~I7b57b95e2d4eeb54d13dcb0960db91cb48b60470","project":"openstack/grenade","branch":"stable/queens","topic":"queens/backport/677245","hashtags":[],"change_id":"I7b57b95e2d4eeb54d13dcb0960db91cb48b60470","subject":"Use docs-pti template","status":"MERGED","created":"2019-09-18 19:40:07.000000000","updated":"2019-09-19 17:40:57.000000000","submitted":"2019-09-19 17:37:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682975-1568914634358-fa3c3f25","meta_rev_id":"1205ba824d2a5f1357b24cfa142f05c54bc0d666","_number":682975,"virtual_id_number":682975,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-09-19 17:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-09-19 15:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-09-19 13:42:30.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2019-09-19 15:31:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 13:42:30.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-19 15:31:51.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-09-19 17:37:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a20607fd5428b790b82842538c519f211419b7d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-18 19:40:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47701b7aa4953160f1c4de9479afb5bd8fdc1a2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 03:13:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/415d385ea9f247ae8a677d7543e6ea3a : SUCCESS in 3m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c2877ae476dc4258ab203bc49a368ea7 : SUCCESS in 49m 42s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/6f97c97b95f148e2b5ad59c4865a985c : SUCCESS in 50m 31s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c76caa94907416589aa1f10e0af0c05 : SUCCESS in 1h 01m 50s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/e393d5359c414679854a08f644484e9a : SUCCESS in 2h 06m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9ff9052d4dcb26f0e54871ae0ff211bf8283fd4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-19 13:42:30.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks OK to me. I\u0027m guessing there was a merge conflict but it\u0027s been resolved.","accounts_in_message":[],"_revision_number":1},{"id":"930e7492eb512390349900dd9c3b9bd156e6e223","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-09-19 15:31:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e2ad064f5d80833c63c8a83fba8732e2304b7db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 15:32:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43330f56897b54e0d9b0f4567e809617f3ccf3d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:37:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8112694866f9e0c520509783187a5a377213cab5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:37:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f7c391d9bb43c49eae836ee4ce5d35 : SUCCESS in 5m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8ce6d95215684d708df0b0d3a21d840d : SUCCESS in 59m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/19d3a32a0e354f21b654a12aaf3fb277 : SUCCESS in 1h 07m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3e008b42cb7e5d8585c2cd22659dbaaab52e8eba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:40:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb247ad9a7045d3b5270fb0987334ae : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c48e1c0ad93a235cbbf4d5fd48106c3b9a41b8d","revisions":{"6c48e1c0ad93a235cbbf4d5fd48106c3b9a41b8d":{"kind":"REWORK","_number":1,"created":"2019-09-18 19:40:07.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/75/682975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/75/682975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/75/682975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/75/682975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/75/682975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/75/682975/1"}}},"commit":{"parents":[{"commit":"6c9a3ed30f11230058aaa837984da9e4b2562f57","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/6c9a3ed30f11230058aaa837984da9e4b2562f57"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-19 16:01:43.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-18 19:39:49.000000000","tz":120},"subject":"Use docs-pti template","message":"Use docs-pti template\n\nInstead of using individual jobs, use the docs-pti template for Zuul\njob configuration so that we can change docs publishing in a single\nplace.\n\nChange-Id: I7b57b95e2d4eeb54d13dcb0960db91cb48b60470\n(cherry picked from commit 969756cbd17aa5b8e4b99ccf95a0f0c713a7bd58)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/6c48e1c0ad93a235cbbf4d5fd48106c3b9a41b8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/6c48e1c0ad93a235cbbf4d5fd48106c3b9a41b8d"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
