)]}'
{"id":"openstack%2Fgrenade~706425","triplet_id":"openstack%2Fgrenade~stable%2Frocky~Icfccaa38bcf9a9567152d0b84580683728c929e5","project":"openstack/grenade","branch":"stable/rocky","topic":"fix-stable-gate","hashtags":[],"change_id":"Icfccaa38bcf9a9567152d0b84580683728c929e5","subject":"Use stable constraint for Tempest pinned stable branches","status":"MERGED","created":"2020-02-07 03:30:52.000000000","updated":"2020-02-08 23:57:30.000000000","submitted":"2020-02-08 23:56:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706425-1581206175899-ed07d331","meta_rev_id":"ef61449ab3dac7014b77b64b072108ed0f7af9f0","_number":706425,"virtual_id_number":706425,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-08 12:05:52.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2020-02-08 21:06:44.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-08 23:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-02-07 21:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-02-07 20:28:22.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-02-07 21:17:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-07 20:28:22.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-07 21:17:20.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-02-08 12:05:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-02-08 23:56:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c2758de365b6f43deb280eeba484e4ddbfd91da","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-07 03:30:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c0766b78098eadd1eba1e662ebf6a8f856e2a0e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-07 03:31:12.000000000","message":"Topic set to fix-stable-gate","accounts_in_message":[],"_revision_number":1},{"id":"22a1003489c1386710545b96ea35b6145ff6241b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 03:50:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd4c8fc5be4474492f7a31f62d2bd2a : FAILURE in 3m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3290c9c7d9a6439380f4faff299ace3d : FAILURE in 17m 47s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/16362694a18e40eaa7ad96e2d9874ecd : FAILURE in 17m 11s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd586a1f0f2d42388cb73bb7288702e0 : FAILURE in 17m 37s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/54ecb21433e741efa3d0e1d45a0342e8 : FAILURE in 17m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05690fd24b99ff03678371e92d2e62cb4753759e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-07 04:01:43.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9f4dae40318a69dc2f1d915f3291dfd4b4cc78bb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-07 04:09:34.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 0cf80cd2fa204c6c6daef4ab969de3e12d5c15ff","accounts_in_message":[],"_revision_number":2},{"id":"3854678984a87f11b4263d41d6fba046796c5420","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-07 04:09:53.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/pike as commit bb4f484edf9889ae4c4542c008af313c7a3aaf90","accounts_in_message":[],"_revision_number":2},{"id":"9873be0e26a84507dcc77222f4d679c089936b45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 04:28:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ed7f447c0d4a54aa66deca9fa87b7f : FAILURE in 2m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/15c7e133f603425b92d962d543388337 : FAILURE in 19m 14s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/5ccddbda1d2845e7b102731a4d2043dd : FAILURE in 19m 23s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/99b1b702811b48c99461560eacc7e18c : FAILURE in 21m 13s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/1873a4162a534232a2d9b022b23d5115 : FAILURE in 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b2640f95617ab0fecf9de891cf2f67c450875416","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-07 12:11:11.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7cafdc9d91d8459680acc7f59f469a15a29b5980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 12:41:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8abd9600dbca4d8e8fdef6aa2b5ea6cc : FAILURE in 3m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c068b40d3e8e4fac8d94fd62574a9726 : FAILURE in 20m 13s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/a7136025e0ae4ea9a89d3d1114c89bb8 : FAILURE in 21m 48s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0fee87838c994374972394cdd9f0fb83 : FAILURE in 24m 51s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/e781db86b684483ab99f593cdf059690 : FAILURE in 22m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ebe1a7aed4e22f05fc5eded4b1ac97e642747f8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-07 19:34:14.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ocata as commit 3b1bd85e8c0813463294d72221783bb514d0cef5","accounts_in_message":[],"_revision_number":3},{"id":"e0fb8f81a3cec04f62c18440811250a71765101f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-07 20:20:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"138dd3d6a396bea99ae4d0c9af4d751bda75336f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-07 20:21:59.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"9aa154bb0fa777bfb53be062d18c16267a28e95d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 20:28:22.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7770565ef1622fcbfec8cf7a5a2c74c9838e0315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 21:03:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ad4aa3d0ae46a0a91f7a2f42941f51 : SUCCESS in 5m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b0acdab457e54229ba6d9ea5d0f29d0a : FAILURE in 29m 29s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/ce940e4b4f484dcbb1d8fc0d6d4826a2 : FAILURE in 22m 47s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40592cf7945145729e0ebf36b65e1e2d : FAILURE in 23m 39s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/69aefe64834648f884523ddc121071e8 : FAILURE in 26m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ecc00908e1facbc34d6dbac1a285f14863536e83","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-02-07 21:16:34.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a18aa4b93328b681b5529f4f68de3e15f125e350","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-02-07 21:17:20.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"91ed528101c26a57f8b19972d7d0e7afbd664c80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 22:07:27.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbb49f4380404d37bbface90cc499783 : SUCCESS in 4m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a7269a9551424984be305a6b2df6c42c : FAILURE in 23m 34s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/83878037171f4056986636dbfd858377 : FAILURE in 26m 48s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/511a1b8f52cb41bb828eac23afba486e : FAILURE in 44m 00s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/66a5a6ce1e3f4acdbc269cc771cc4e61 : FAILURE in 31m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"51c10f0acd069c27bbe9994d3c4896b9016d9dea","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-02-08 12:05:52.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1ac6c1a824f257beb88d67ca0dc0a8237a253df0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 12:37:25.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01ae46690a7e4d1f90e7669e6bc9f292 : SUCCESS in 4m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a75cb39e653b4a29b5df5c0edf0fa46b : FAILURE in 20m 03s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/d1be73dd9cff49b68f2b8b38914067f4 : FAILURE in 21m 57s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c39af014f514cccbab092f754bd913f : FAILURE in 25m 38s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/3267943ebe9749d68ee640d1616d1971 : FAILURE in 22m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"009a883847b4101eb259d18b0697331019730ad1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-08 21:06:44.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e09e717daf2bcb53da1bc8b873ca246edf9d4ef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 22:34:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2be61ce404f54ea4b78b9fe384c5e904 : SUCCESS in 3m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cd12e465812e40778ced3ba42e0fea7e : SUCCESS in 52m 21s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/8e8edc4023e946bd88ade85a278c1e15 : FAILURE in 36m 38s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dadd5f482e9a4978a606e7fc33009d0e : SUCCESS in 1h 02m 44s\n- ironic-grenade-dsvm https://zuul.opendev.org/t/openstack/build/1bad0917ff444a4ba82b6013f03f2f5c : FAILURE in 1h 22m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9565e8ff543e153feb5f86199b8be8f6d7a7bfa9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 22:34:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6767d69c5019a72cdd0a7bdad230d4650f100982","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 23:56:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"fcc95fa40dde5530e6bdc651f85d0da1d83ac3f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 23:56:15.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/524e92a94f8f44b0a601fdad86e45ed1 : SUCCESS in 5m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4bcc6fb49a20484cae411d7060b3a534 : SUCCESS in 1h 01m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed08800a475c4596b7ed90804760d7cf : SUCCESS in 1h 18m 31s","accounts_in_message":[],"_revision_number":5},{"id":"ef61449ab3dac7014b77b64b072108ed0f7af9f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 23:57:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091ad423249842d7a57b761b3ea135c5 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"067357c06c61de23157c66e8e4ba560ca5325937","revisions":{"42178bd5867fb9a9c069c22b997cafdd82f18f85":{"kind":"REWORK","_number":1,"created":"2020-02-07 03:30:52.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/25/706425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/25/706425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/25/706425/1"}}},"commit":{"parents":[{"commit":"920e26b991401b33c0c47bb709cc815da3a3ddc3","subject":"Use docs-pti template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/920e26b991401b33c0c47bb709cc815da3a3ddc3"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 03:28:49.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 03:30:50.000000000","tz":-360},"subject":"Use stable constraint for Tempest pinned stable branches","message":"Use stable constraint for Tempest pinned stable branches\n\nStable branches till stable/rocky is using python\nversion \u003cpy3.6. Tempest test those branch in venv\nbut Tempest tox use the master upper-constraint[1]\nwhich block installation due to dependencies\nrequire \u003e\u003dpy3.6. For exmaple, oslo.concurrency 4.0.0\nis not compatible for \u003cpy3.6.\n\nAs we pin Tempest for EM stable brach, we should be\nable to use stable constraint for Tempest installation\nas well as while running during run-tempest playbook.\n\ntox.ini is hard coded to use master constraint[1] which force\nrun-tempest to recreate the tox env and use the master constraint.\nFix for that- https://review.opendev.org/#/c/705870/\n\ngrenade needs to use the stable u-c so that\nTempest installation in venv will use stable\nbranch constraint.\n\nDepends-On: https://review.opendev.org/#/c/705870/\n\n[1] https://opendev.org/openstack/tempest/src/commit/bc9fe8eca801f54915ff3eafa418e6e18ac2df63/tox.ini#L14\n\nChange-Id: Icfccaa38bcf9a9567152d0b84580683728c929e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/42178bd5867fb9a9c069c22b997cafdd82f18f85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/42178bd5867fb9a9c069c22b997cafdd82f18f85"}]},"branch":"refs/heads/stable/rocky"},"3bee99fc6f7b1c7a491cb70057653012b1547323":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-07 04:01:43.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/25/706425/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/25/706425/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/25/706425/2"}}},"commit":{"parents":[{"commit":"920e26b991401b33c0c47bb709cc815da3a3ddc3","subject":"Use docs-pti template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/920e26b991401b33c0c47bb709cc815da3a3ddc3"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 03:28:49.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-07 04:01:40.000000000","tz":0},"subject":"Use stable constraint for Tempest pinned stable branches","message":"Use stable constraint for Tempest pinned stable branches\n\nStable branches till stable/rocky is using python\nversion \u003cpy3.6. Tempest test those branch in venv\nbut Tempest tox use the master upper-constraint[1]\nwhich block installation due to dependencies\nrequire \u003e\u003dpy3.6. For exmaple, oslo.concurrency 4.0.0\nis not compatible for \u003cpy3.6.\n\nAs we pin Tempest for EM stable brach, we should be\nable to use stable constraint for Tempest installation\nas well as while running during run-tempest playbook.\n\ntox.ini is hard coded to use master constraint[1] which force\nrun-tempest to recreate the tox env and use the master constraint.\nFix for that- https://review.opendev.org/#/c/705870/\n\ngrenade needs to use the stable u-c so that\nTempest installation in venv will use stable\nbranch constraint.\n\n[1] https://opendev.org/openstack/tempest/src/commit/bc9fe8eca801f54915ff3eafa418e6e18ac2df63/tox.ini#L14\n\nChange-Id: Icfccaa38bcf9a9567152d0b84580683728c929e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/3bee99fc6f7b1c7a491cb70057653012b1547323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/3bee99fc6f7b1c7a491cb70057653012b1547323"}]},"branch":"refs/heads/stable/rocky"},"b1d169a7a7858f2361d210bad8911bb18cd59d7f":{"kind":"REWORK","_number":3,"created":"2020-02-07 12:11:11.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/25/706425/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/25/706425/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/25/706425/3"}}},"commit":{"parents":[{"commit":"920e26b991401b33c0c47bb709cc815da3a3ddc3","subject":"Use docs-pti template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/920e26b991401b33c0c47bb709cc815da3a3ddc3"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 03:28:49.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-07 12:10:46.000000000","tz":0},"subject":"Use stable constraint for Tempest pinned stable branches","message":"Use stable constraint for Tempest pinned stable branches\n\nStable branches till stable/rocky is using python\nversion \u003cpy3.6. Tempest test those branch in venv\nbut Tempest tox use the master upper-constraint[1]\nwhich block installation due to dependencies\nrequire \u003e\u003dpy3.6. For exmaple, oslo.concurrency 4.0.0\nis not compatible for \u003cpy3.6.\n\nAs we pin Tempest for EM stable brach, we should be\nable to use stable constraint for Tempest installation\nas well as while running during run-tempest playbook.\n\ntox.ini is hard coded to use master constraint[1] which force\nrun-tempest to recreate the tox env and use the master constraint.\nFix for that- https://review.opendev.org/#/c/705870/\n\ngrenade needs to use the stable u-c so that\nTempest installation in venv will use stable\nbranch constraint.\n\n[1] https://opendev.org/openstack/tempest/src/commit/bc9fe8eca801f54915ff3eafa418e6e18ac2df63/tox.ini#L14\n\nChange-Id: Icfccaa38bcf9a9567152d0b84580683728c929e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/b1d169a7a7858f2361d210bad8911bb18cd59d7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/b1d169a7a7858f2361d210bad8911bb18cd59d7f"}]},"branch":"refs/heads/stable/rocky"},"7c59426778fb8edc2c183e6658c8b347e0f0159b":{"kind":"REWORK","_number":4,"created":"2020-02-07 20:20:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/25/706425/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/25/706425/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/25/706425/4"}}},"commit":{"parents":[{"commit":"920e26b991401b33c0c47bb709cc815da3a3ddc3","subject":"Use docs-pti template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/920e26b991401b33c0c47bb709cc815da3a3ddc3"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 03:28:49.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 20:19:47.000000000","tz":-360},"subject":"Use stable constraint for Tempest pinned stable branches","message":"Use stable constraint for Tempest pinned stable branches\n\nStable branches till stable/rocky is using python\nversion \u003cpy3.6. Tempest test those branch in venv\nbut Tempest tox use the master upper-constraint[1]\nwhich block installation due to dependencies\nrequire \u003e\u003dpy3.6. For exmaple, oslo.concurrency 4.0.0\nis not compatible for \u003cpy3.6.\n\nAs we pin Tempest for EM stable brach, we should be\nable to use stable constraint for Tempest installation\nas well as while running during run-tempest playbook.\n\ntox.ini is hard coded to use master constraint[1] which force\nrun-tempest to recreate the tox env and use the master constraint.\nFix for that- https://review.opendev.org/#/c/705870/\n\ngrenade needs to use the stable u-c so that\nTempest installation in venv will use stable\nbranch constraint.\n\nAlso fixing broken doc job.\n\n[1] https://opendev.org/openstack/tempest/src/commit/bc9fe8eca801f54915ff3eafa418e6e18ac2df63/tox.ini#L14\n\nChange-Id: Icfccaa38bcf9a9567152d0b84580683728c929e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/7c59426778fb8edc2c183e6658c8b347e0f0159b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/7c59426778fb8edc2c183e6658c8b347e0f0159b"}]},"branch":"refs/heads/stable/rocky"},"067357c06c61de23157c66e8e4ba560ca5325937":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-02-07 20:21:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/25/706425/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/25/706425/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/25/706425/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/25/706425/5"}}},"commit":{"parents":[{"commit":"920e26b991401b33c0c47bb709cc815da3a3ddc3","subject":"Use docs-pti template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/920e26b991401b33c0c47bb709cc815da3a3ddc3"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 03:28:49.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-07 20:21:56.000000000","tz":0},"subject":"Use stable constraint for Tempest pinned stable branches","message":"Use stable constraint for Tempest pinned stable branches\n\nStable branches till stable/rocky is using python\nversion \u003cpy3.6. Tempest test those branch in venv\nbut Tempest tox use the master upper-constraint[1]\nwhich block installation due to dependencies\nrequire \u003e\u003dpy3.6. For exmaple, oslo.concurrency 4.0.0\nis not compatible for \u003cpy3.6.\n\nAs we pin Tempest for EM stable brach, we should be\nable to use stable constraint for Tempest installation\nas well as while running during run-tempest playbook.\n\ntox.ini is hard coded to use master constraint[1] which force\nrun-tempest to recreate the tox env and use the master constraint.\nFix for that- https://review.opendev.org/#/c/705870/\n\ngrenade needs to use the stable u-c so that\nTempest installation in venv will use stable\nbranch constraint.\n\nAlso fixing broken doc job.\n\nDepends-On: https://review.opendev.org/#/c/706178/\n\n[1] https://opendev.org/openstack/tempest/src/commit/bc9fe8eca801f54915ff3eafa418e6e18ac2df63/tox.ini#L14\n\nChange-Id: Icfccaa38bcf9a9567152d0b84580683728c929e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/067357c06c61de23157c66e8e4ba560ca5325937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/067357c06c61de23157c66e8e4ba560ca5325937"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
