)]}'
{"id":"openstack%2Fgrenade~71731","triplet_id":"openstack%2Fgrenade~stable%2Fhavana~I6a6e4da06806cb840d7af673673e06f34eb297e0","project":"openstack/grenade","branch":"stable/havana","hashtags":[],"change_id":"I6a6e4da06806cb840d7af673673e06f34eb297e0","subject":"Add upgrade-horizon","status":"MERGED","created":"2014-02-07 03:48:28.000000000","updated":"2014-03-03 17:14:26.000000000","submitted":"2014-03-03 17:14:26.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"14cf4aa011280abafb994e0eff19045f87f47ef8","_number":71731,"virtual_id_number":71731,"owner":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":2,"date":"2014-03-03 17:14:26.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"recommended":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2014-02-11 13:57:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2014-02-11 10:37:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":2,"date":"2014-02-14 02:40:53.000000000","_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"date":"2014-02-14 03:18:08.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"value":2,"date":"2014-03-03 13:39:01.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"date":"2014-02-14 02:40:53.000000000","_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"date":"2014-02-11 03:17:26.000000000","_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"value":1,"date":"2014-03-03 13:39:01.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-02-07 03:48:28.000000000","updated_by":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"reviewer":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"state":"REVIEWER"},{"updated":"2014-02-11 10:37:39.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2014-02-11 13:57:23.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2014-02-14 02:40:53.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2014-03-03 13:39:01.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2014-03-03 17:14:26.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"4df2bf6409bea62b8a4da55945fa07fb54629909","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-07 03:48:29.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"744b0a68756413760852953626859c9997d52833","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-02-07 03:52:27.000000000","message":"Patch Set 1: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"feea69f2ee299d53a9e765182b77569f1d22f291","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-07 04:09:22.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/31/71731/1/check/check-grenade-dsvm/51d9e5b : FAILURE in 19m 09s\n- check-grenade-dsvm-forward http://logs.openstack.org/31/71731/1/check/check-grenade-dsvm-forward/b72c1ec : FAILURE in 15m 10s (non-voting)\n- check-grenade-dsvm-neutron http://logs.openstack.org/31/71731/1/check/check-grenade-dsvm-neutron/6837cd9 : FAILURE in 17m 01s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"51ec34681fcd7614c0aee8ee9cb2ca0c2bd9b3ba","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2014-02-07 18:21:16.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"f38a724534578a8078c0acdc19a4243e195afbc8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-09 16:18:37.000000000","message":"Patch Set 1: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"66abe8412e2e69e78890d8fbd79dd15c0cd047a0","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2014-02-09 16:18:37.000000000","message":"Patch Set 1:\n\nActually, from-* do not exist in stable/havana - is this branch used in gate jobs at all?","accounts_in_message":[],"_revision_number":1},{"id":"94c6292bff6fbef23b4c503b815ec11bbac5a844","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2014-02-09 16:21:50.000000000","message":"Patch Set 1: No score\n\n","accounts_in_message":[],"_revision_number":1},{"id":"3bf90d85de5afc3743d05dcafc3c34233db78c86","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-09 16:35:01.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/31/71731/1/check/check-grenade-dsvm/691a7ec : FAILURE in 15m 19s\n- check-grenade-dsvm-forward http://logs.openstack.org/31/71731/1/check/check-grenade-dsvm-forward/6116953 : FAILURE in 16m 21s (non-voting)\n- check-grenade-dsvm-neutron http://logs.openstack.org/31/71731/1/check/check-grenade-dsvm-neutron/2f50b3d : FAILURE in 15m 46s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"db54cae2ef0a6eb30e550e88482dc5f94de4bc0d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2014-02-10 10:14:23.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"cc80953a5da8d91497d07ee13a4ed4bd48abca26","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2014-02-11 03:17:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"13d97e8358e8b929de7bc0c39438c22b7d4ac0d6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-11 03:17:27.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"e751ff3a90d363dab4e1c787b17262e5cec44d89","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2014-02-11 03:21:13.000000000","message":"Patch Set 1: (2 inline comments)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"8895a89ddff3caab3a3712a25bd1d9298a19b54b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-11 03:53:05.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- check-grenade-dsvm http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm/805c8a4 : SUCCESS in 30m 55s\n- check-grenade-dsvm-forward http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-forward/2240563 : SUCCESS in 29m 06s (non-voting)\n- check-grenade-dsvm-neutron http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-neutron/1e98cb1 : FAILURE in 20m 27s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"bc3c54c68fce2966f8615c04617a959432f1dc14","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2014-02-11 10:37:39.000000000","message":"Patch Set 2: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":2},{"id":"85db0e83e5f463dca79be3a1149e6f8285f36ece","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2014-02-11 13:57:23.000000000","message":"Patch Set 2: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":2},{"id":"544e664257bd148bb4dd1d7afc2f41f6a8d898c1","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2014-02-14 02:40:53.000000000","message":"Patch Set 2: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"4f6fd03a10759d256336856e58b0a3103e156fda","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-14 02:40:57.000000000","message":"Patch Set 2: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"0162c086186532d111eb836b3f5f88d8402c8fcb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-14 03:03:24.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm/618a774 : FAILURE in 14m 50s\n- check-grenade-dsvm-forward http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-forward/acd4ab1 : FAILURE in 14m 23s (non-voting)\n- check-grenade-dsvm-neutron http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-neutron/fe2ed97 : FAILURE in 19m 02s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"9acbe47658bedc9f11796adde22d2ee2403296bb","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2014-02-14 03:18:08.000000000","message":"Patch Set 2:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":2},{"id":"684e176513f778ca61ae1ece5eeea5393b37f724","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-14 03:18:23.000000000","message":"Patch Set 2: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"14b742690152fe5d26c3940540da5a041176953b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-14 03:39:24.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm/5d6a1fa : FAILURE in 14m 19s\n- check-grenade-dsvm-forward http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-forward/b491730 : FAILURE in 14m 56s (non-voting)\n- check-grenade-dsvm-neutron http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-neutron/89deffd : FAILURE in 17m 11s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"2c269ecd314f2b4ce237912c6ac8bfa4b4f5a52d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-15 03:21:04.000000000","message":"Patch Set 2: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"c860e33f450a0ca5739a9e98c9f7624ad071d6d1","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2014-02-15 03:21:04.000000000","message":"Patch Set 2:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":2},{"id":"ebefcee59ed048a0e90e68cdfb6ab80b52451f48","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-15 03:45:09.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm/3bb494c : FAILURE in 13m 41s\n- check-grenade-dsvm-forward http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-forward/2cbe5ce : FAILURE in 13m 30s (non-voting)\n- check-grenade-dsvm-neutron http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-neutron/dc8258b : FAILURE in 13m 23s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"c92c519dd2e8668c60a39b4cc5b0b1275907cb33","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-17 06:22:48.000000000","message":"Patch Set 2: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"c9013c1e96cfe2e576614dd007dd34500dc95095","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2014-02-17 06:22:48.000000000","message":"Patch Set 2:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":2},{"id":"175e4f77358665e2fc7c394e52d1f0fd196f7121","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-17 06:44:25.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm/fe0b04b : FAILURE in 12m 28s\n- check-grenade-dsvm-forward http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-forward/19e0cef : FAILURE in 13m 40s (non-voting)\n- check-grenade-dsvm-neutron http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-neutron/c93bac1 : FAILURE in 16m 53s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"c6a95a13d62c74bc18152dc965f62a43d14da6e8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-21 01:15:25.000000000","message":"Patch Set 2: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"38201b4bb331e811d56146f12016c23005ddf0d3","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2014-02-21 01:15:25.000000000","message":"Patch Set 2:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":2},{"id":"829ea930eee9d74ed5e06cf83ed6b58e6b11271f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-21 03:06:29.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm/969446f : FAILURE in 17m 02s\n- check-grenade-dsvm-forward http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-forward/7f679b5 : FAILURE in 17m 57s (non-voting)\n- check-grenade-dsvm-neutron http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-neutron/355621e : FAILURE in 17m 33s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"b2dd3812964216fb6257143d809c547d88c02cab","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2014-02-26 02:54:30.000000000","message":"Patch Set 2:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":2},{"id":"fb3bafa72e5ead2d746d69e6085399a6db10bb70","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-26 02:54:31.000000000","message":"Patch Set 2: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"81a5e1632ca67a62934bae1c21fa64642e0d476c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-26 03:11:37.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm/fa8113d : FAILURE in 17m 05s\n- check-grenade-dsvm-forward http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-forward/e2574bf : FAILURE in 14m 47s (non-voting)\n- check-grenade-dsvm-neutron http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-neutron/88f311e : FAILURE in 16m 23s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"589b03377458c0f37d2b7d3feaee678b61e43b99","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2014-02-26 22:52:18.000000000","message":"Patch Set 2:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":2},{"id":"d0357b5cc3afc2bce92bf3e377fc221b2671e389","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-26 22:52:19.000000000","message":"Patch Set 2: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"39a2d9f4d8ad4d28c3527c7339b1d444daa08c3e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-26 23:36:23.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm/2986709 : FAILURE in 17m 41s\n- check-grenade-dsvm-forward http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-forward/a46173b : FAILURE in 27m 54s (non-voting)\n- check-grenade-dsvm-neutron http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-neutron/577a7be : FAILURE in 25m 37s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"eb0a3ddd7342f14aab73d5bdccdc1104f1605975","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2014-02-27 03:30:38.000000000","message":"Patch Set 2:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":2},{"id":"d7337efc63191b410a1715b27f532b9eb214c099","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-27 03:30:39.000000000","message":"Patch Set 2: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"60d6b900dfba93598896810f67169e089c25cb0f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-27 03:44:55.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm/ba4c37f : FAILURE in 12m 50s\n- check-grenade-dsvm-forward http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-forward/c6f3e0f : FAILURE in 12m 39s (non-voting)\n- check-grenade-dsvm-neutron http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-neutron/c933f85 : FAILURE in 14m 15s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"1796d83cdda811503ca5fe4bd0251f9b66a45ab8","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2014-03-01 16:38:30.000000000","message":"Patch Set 2:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":2},{"id":"2f9c56dfd067477f29f05a9e7d423855857ad4b2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-01 16:38:31.000000000","message":"Patch Set 2: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"dc7580968984fcfe7dfeeb877d48ea9bde093b95","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-01 17:07:59.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- check-grenade-dsvm http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm/95273d1 : SUCCESS in 28m 05s\n- check-grenade-dsvm-forward http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-forward/842ba9e : SUCCESS in 29m 27s (non-voting)\n- check-grenade-dsvm-neutron http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-neutron/0fdbaf4 : FAILURE in 16m 40s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"01039f55910468bf948e2c36d6e660609c0302bf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-03 13:39:01.000000000","message":"Patch Set 2: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"433f69f48b414460d55e32ad8630ba40ab4f8bb6","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2014-03-03 13:39:01.000000000","message":"Patch Set 2: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":2},{"id":"4b1e4ec6e5323502a8483a77310be675ca9f0a24","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-03 14:20:51.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"1735da41aad7fd6210114d5ca32137bb013fb6cd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-03 14:20:51.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- check-grenade-dsvm http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm/b732770 : SUCCESS in 27m 42s\n- check-grenade-dsvm-forward http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-forward/25848b6 : FAILURE in 41m 48s (non-voting)\n- check-grenade-dsvm-neutron http://logs.openstack.org/31/71731/2/check/check-grenade-dsvm-neutron/fd437b5 : FAILURE in 22m 09s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"2048d7d27ba4acd0debd479c9545cf693eb43a47","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-03 17:14:26.000000000","message":"Patch Set 2: Verified\n\nBuild succeeded.\n\n- gate-grenade-dsvm http://logs.openstack.org/31/71731/2/gate/gate-grenade-dsvm/3dbf2cb : SUCCESS in 42m 35s\n- gate-grenade-dsvm-forward http://logs.openstack.org/31/71731/2/gate/gate-grenade-dsvm-forward/0e5b2ae : SUCCESS in 30m 33s (non-voting)\n- gate-grenade-dsvm-neutron http://logs.openstack.org/31/71731/2/gate/gate-grenade-dsvm-neutron/4a2f34a : FAILURE in 18m 05s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"14cf4aa011280abafb994e0eff19045f87f47ef8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-03 17:14:26.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"97fea9b26799b6b4a6f951af9d3cf8f75f7c00e9","revisions":{"862673a98750971277ae5133ea2b2fc027da8fc5":{"kind":"REWORK","_number":1,"created":"2014-02-07 03:48:28.000000000","uploader":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"ref":"refs/changes/31/71731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/31/71731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/31/71731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/31/71731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/31/71731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/31/71731/1"}}},"commit":{"parents":[{"commit":"510b79c3a67111f7a70d9a21dfdf65b3bde3ab6d","subject":"open up stable/havana branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/510b79c3a67111f7a70d9a21dfdf65b3bde3ab6d"}]}],"author":{"name":"Joe Gordon","email":"joe.gordon0@gmail.com","date":"2013-12-31 02:26:06.000000000","tz":-480},"committer":{"name":"Joshua Hesketh","email":"josh@nitrotech.org","date":"2014-02-07 03:48:24.000000000","tz":660},"subject":"Add upgrade-horizon","message":"Add upgrade-horizon\n\nUpgrade horizon along with the rest of devstack before running new\ndevstack tests and tempest.\n\nupgrade-horizon is based off of upgrade-keystone.\n\nChange-Id: I6a6e4da06806cb840d7af673673e06f34eb297e0\nRelated-Bug: #1265057\n(cherry picked from commit e9546e66e6236a1e2c79734ee0c9dcb0cc50bbb3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/862673a98750971277ae5133ea2b2fc027da8fc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/862673a98750971277ae5133ea2b2fc027da8fc5"}]},"branch":"refs/heads/stable/havana"},"97fea9b26799b6b4a6f951af9d3cf8f75f7c00e9":{"kind":"REWORK","_number":2,"created":"2014-02-11 03:17:26.000000000","uploader":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"ref":"refs/changes/31/71731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/31/71731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/31/71731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/31/71731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/31/71731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/31/71731/2"}}},"commit":{"parents":[{"commit":"510b79c3a67111f7a70d9a21dfdf65b3bde3ab6d","subject":"open up stable/havana branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/510b79c3a67111f7a70d9a21dfdf65b3bde3ab6d"}]}],"author":{"name":"Joe Gordon","email":"joe.gordon0@gmail.com","date":"2013-12-31 02:26:06.000000000","tz":-480},"committer":{"name":"Joshua Hesketh","email":"josh@nitrotech.org","date":"2014-02-11 03:17:05.000000000","tz":660},"subject":"Add upgrade-horizon","message":"Add upgrade-horizon\n\nUpgrade horizon along with the rest of devstack before running new\ndevstack tests and tempest.\n\nupgrade-horizon is based off of upgrade-keystone.\n\nChange-Id: I6a6e4da06806cb840d7af673673e06f34eb297e0\nRelated-Bug: #1265057\n(cherry picked from commit e9546e66e6236a1e2c79734ee0c9dcb0cc50bbb3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/97fea9b26799b6b4a6f951af9d3cf8f75f7c00e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/97fea9b26799b6b4a6f951af9d3cf8f75f7c00e9"}]},"branch":"refs/heads/stable/havana"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
