)]}'
{"id":"openstack%2Fgrenade~872174","triplet_id":"openstack%2Fgrenade~master~I36a3bfe31c77451b3c34838c05d13d00187cc0cd","project":"openstack/grenade","branch":"master","topic":"ensure-rust-cryptography","attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-01-30 14:28:10.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I36a3bfe31c77451b3c34838c05d13d00187cc0cd","subject":"WIP: use ensure-rust","status":"NEW","created":"2023-01-30 14:27:40.000000000","updated":"2023-01-30 16:16:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fc031a3de8689c0415334d75e538e8e280037518","_number":872174,"virtual_id_number":872174,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-30 16:16:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-30 14:28:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5905066e66f24736f282a53a108051396fd7601e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-30 14:27:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c765274a55080a26fbd0afb9f9073fa5bb4a7fd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 14:28:10.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/grenade on branch master.  The error was:\n\n  expected a dictionary @ data[\u0027roles\u0027][0]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: grenade-multinode\n      description: |\n        Basic multinode grenade job\n      parent: grenade\n      roles:\n        - ensure-rust\n      branches:\n        - ussuri\n      nodeset: openstack-two-node-focal\n      pre-run: playbooks/multinode-pre.yaml\n\n  in \"openstack/grenade/.zuul.yaml@master\", line 375, column 3","accounts_in_message":[],"_revision_number":1},{"id":"2492af801af1da7c6df7161926392bf925e6099f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-30 14:35:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc031a3de8689c0415334d75e538e8e280037518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 16:16:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26c35cd75b384dc1b8f9851613b37931\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9e7c680850d453187b6810fed9f48fb : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/d8ec5af40c57451fb57a56210e287bfc : SUCCESS in 59m 34s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/81f50dfa13c04557a5ea7dc114e600fa : SUCCESS in 1h 15m 52s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/c6ff8368d8b14b3aa920f414361a84bc : SUCCESS in 1h 26m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e67a4b805fa94d7d9e9a21048737b8cb : SUCCESS in 59m 33s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/5d304f32c6f04ea19c61725ed775d65d : SUCCESS in 1h 13m 26s (non-voting)\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0b7e6f908e014fa09731b8b82dddaf78 : SUCCESS in 1h 02m 51s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c00c987311e5583edb92a8c62eca8ea6bfe8ad84","revisions":{"d65f9f4bbce94e4b03c8ddcd5ecdf1d6c068c1ce":{"kind":"REWORK","_number":1,"created":"2023-01-30 14:27:40.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/872174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/872174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/872174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/872174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/872174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/872174/1"}}},"commit":{"parents":[{"commit":"c7a6dff51d8e047564b380de1a81cfa3d7798fcd","subject":"Merge \"Remove lib/neutron-legacy leftovers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/c7a6dff51d8e047564b380de1a81cfa3d7798fcd"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-01-30 14:26:20.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-01-30 14:26:20.000000000","tz":60},"subject":"WIP: use ensure-rust","message":"WIP: use ensure-rust\n\nChange-Id: I36a3bfe31c77451b3c34838c05d13d00187cc0cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d65f9f4bbce94e4b03c8ddcd5ecdf1d6c068c1ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d65f9f4bbce94e4b03c8ddcd5ecdf1d6c068c1ce"}]},"branch":"refs/heads/master"},"c00c987311e5583edb92a8c62eca8ea6bfe8ad84":{"kind":"REWORK","_number":2,"created":"2023-01-30 14:35:24.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/872174/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/74/872174/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/872174/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/872174/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/74/872174/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/74/872174/2"}}},"commit":{"parents":[{"commit":"c7a6dff51d8e047564b380de1a81cfa3d7798fcd","subject":"Merge \"Remove lib/neutron-legacy leftovers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/c7a6dff51d8e047564b380de1a81cfa3d7798fcd"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-01-30 14:26:20.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-01-30 14:35:09.000000000","tz":60},"subject":"WIP: use ensure-rust","message":"WIP: use ensure-rust\n\nChange-Id: I36a3bfe31c77451b3c34838c05d13d00187cc0cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/c00c987311e5583edb92a8c62eca8ea6bfe8ad84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/c00c987311e5583edb92a8c62eca8ea6bfe8ad84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
