)]}'
{"id":"openstack%2Fgrenade~895882","triplet_id":"openstack%2Fgrenade~stable%2Fwallaby~I43c696ef2029cbf351f902b293df05f296dbb2a2","project":"openstack/grenade","branch":"stable/wallaby","topic":"update-regexes","attention_set":{},"removed_from_attention_set":{"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2023-09-20 21:59:11.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}},"hashtags":[],"change_id":"I43c696ef2029cbf351f902b293df05f296dbb2a2","subject":"Use re2 compatible regexes in branch matchers","status":"MERGED","created":"2023-09-20 07:33:47.000000000","updated":"2023-09-21 01:26:36.000000000","submitted":"2023-09-21 01:24:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895882-update-regexes","meta_rev_id":"85e427ee4df6b1324ed7f413fb0fbbbe89ffc823","_number":895882,"virtual_id_number":895882,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-21 01:24:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-09-20 15:58:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-20 17:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-20 17:30:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-20 08:59:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-20 15:58:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-09-20 17:30:42.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"5e498d828b271f730587648c8ecf8a195c346e73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-09-20 07:33:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1c712a9c0b590bd271d99062c35de4d8e393d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 08:59:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e34c15f1d23c43e2bf92565910199f57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e61773e8b6346f291f9399c5c2f38d4 : SUCCESS in 17m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/398976a8c14c49ea909afbccb0b3b886 : SUCCESS in 48m 17s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/0df0f3646c814828821bf3f0b618b1c1 : SUCCESS in 1h 06m 31s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/dc32105fc6464c9c8d6c927222856db9 : FAILURE in 1h 24m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/08048569e5cd48c5bb8a8b4849f40d03 : SUCCESS in 44m 02s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/ad9696f75e064027a0c3e4b0323a267f : SUCCESS in 1h 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0edfeb85e58295610468c4656f5e9f28fe0f29ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-09-20 13:43:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d83694b273ae3718aa3f8068f9151a1c0033ed5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 15:31:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/456fd3066770433a9f7d1b2db93a6436\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17bd01e37ea74f4abd2fd471fbce1588 : SUCCESS in 12m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/2ff79d88cd2840d3afc5e266794c6d0f : SUCCESS in 1h 11m 48s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/b013db0a886a4824aa3c46b21612e833 : SUCCESS in 1h 06m 35s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/6be0fea9b0b144c3824a3a106782bd28 : SUCCESS in 1h 19m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/558293bfaad64c6dbf80f6c7eb473722 : SUCCESS in 1h 02m 47s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/b2333fb43917475d9939a6be0784b58e : SUCCESS in 1h 27m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4456d283b5e713c078acd17f835fd79312f9d1ff","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-20 15:57:57.000000000","message":"Topic set to update-regexes","accounts_in_message":[],"_revision_number":2},{"id":"8590e26681f0b3cb915959434e3a325c4809c123","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-20 15:58:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5bde646536a57b7e961fd131faee4f65d9a2c50c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-09-20 17:30:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e40ae07a80c813d4210c2c9b89ec61f35c554c4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 17:31:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"718b7478a5c1480a6b30d6073061a503253eb4cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 19:17:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c62908168814a039866cb96ab24b1c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf1a3aed6ab34b71b7221c363b235041 : SUCCESS in 11m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/e39aac1667a64563ac1a7d05f00a1db4 : SUCCESS in 1h 15m 26s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/61b33273843546f7830858c62c4f6ccf : SUCCESS in 1h 13m 31s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/993c72c965f340b49e1215d483a26ab9 : FAILURE in 54m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c3ab5f0987fb46d7917c8a501443d214 : SUCCESS in 44m 30s","accounts_in_message":[],"_revision_number":2},{"id":"32738d87c327c5a5bded0a2bf98c4ffbcbb5f943","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-09-20 21:59:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"24dfa54f31d1abcb7bf382c67d355fb39511f166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 23:36:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d7fb4f92bc047e7beae17343b1f52d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8ddfa3ff89488ab68aa1ca227f34cd : SUCCESS in 17m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/a5cb14798b7c401f833d826a55c0c11c : SUCCESS in 50m 14s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/4cb1cf218a664926a4ac744ca218729c : SUCCESS in 1h 11m 30s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/28023294d0c54f2285586463b912cc46 : SUCCESS in 1h 32m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cf46923e1b9a4f25bd89b3036c2b3063 : SUCCESS in 56m 38s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/bb3bdf41e03547a4afec4414c33e3d3b : SUCCESS in 1h 18m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3499064adc2dfc14fc1300b0fe185197c14b1918","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 23:37:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"74966cac7c9e3af17aa0379740db75cd3e05ecf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 01:24:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32a4e01f7973454d93b7868f9c934a0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/909b7d97f93b4701bf9ec1cce9636d00 : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/a09d05db4383454a943878abfbf2cb35 : SUCCESS in 1h 10m 05s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/f711d94ec4154c4fa5a146af1a140aa7 : SUCCESS in 1h 15m 21s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/b69df9bf9b3e4caf833985a1d331966b : SUCCESS in 1h 40m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b791e4ba67724c8fa7d104c3e2179dc2 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":2},{"id":"6078d60ee9d570515cbb55ac219d90940a7a9946","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 01:24:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"85e427ee4df6b1324ed7f413fb0fbbbe89ffc823","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 01:26:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28553e1233ba4d7d97857f2c305556d3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df3f9dfbb3c046b6876ac013e0ce1734 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e280e4d4f1b5ba019478f36fd2a6f7c5bb276a76","revisions":{"d2372791514f0feb1856e4cfde3a7c79413c20cd":{"kind":"REWORK","_number":1,"created":"2023-09-20 07:33:47.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/82/895882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/82/895882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/82/895882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/82/895882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/82/895882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/82/895882/1"}}},"commit":{"parents":[{"commit":"c5751f437767db4ef893293084a29a559e8ed9a0","subject":"Don\u0027t enable the dstat service in CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/c5751f437767db4ef893293084a29a559e8ed9a0"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-09-20 07:33:38.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-09-20 07:33:38.000000000","tz":120},"subject":"Use re2 compatible regexes in branch matchers","message":"Use re2 compatible regexes in branch matchers\n\nPython regexes are deprecated in Zuul, so update the use of regexes\nin branch matchers with re2-compatible syntax.\n\nChange-Id: I43c696ef2029cbf351f902b293df05f296dbb2a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d2372791514f0feb1856e4cfde3a7c79413c20cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/d2372791514f0feb1856e4cfde3a7c79413c20cd"}]},"branch":"refs/heads/stable/wallaby"},"e280e4d4f1b5ba019478f36fd2a6f7c5bb276a76":{"kind":"REWORK","_number":2,"created":"2023-09-20 13:43:37.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/82/895882/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/82/895882/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/82/895882/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/82/895882/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/82/895882/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/82/895882/2"}}},"commit":{"parents":[{"commit":"c5751f437767db4ef893293084a29a559e8ed9a0","subject":"Don\u0027t enable the dstat service in CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/c5751f437767db4ef893293084a29a559e8ed9a0"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-09-20 07:33:38.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-09-20 13:43:20.000000000","tz":120},"subject":"Use re2 compatible regexes in branch matchers","message":"Use re2 compatible regexes in branch matchers\n\nPython regexes are deprecated in Zuul, so update the use of regexes\nin branch matchers with re2-compatible syntax.\nThe commit also removes matches for branches that are EOL - all\nbranches until stable/rocky (including).\n\nChange-Id: I43c696ef2029cbf351f902b293df05f296dbb2a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/e280e4d4f1b5ba019478f36fd2a6f7c5bb276a76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/e280e4d4f1b5ba019478f36fd2a6f7c5bb276a76"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
