)]}'
{"id":"openstack%2Fgrenade~914687","triplet_id":"openstack%2Fgrenade~stable%2F2023.1~If5df9bae3c62042493ec7c39ff3da7c09e830619","project":"openstack/grenade","branch":"stable/2023.1","topic":"qa-2024-1-release","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-03-28 19:02:40.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-03-28 23:06:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If5df9bae3c62042493ec7c39ff3da7c09e830619","subject":"Fix grenade-skip-level job to run on respective stable branch","status":"MERGED","created":"2024-03-28 17:45:24.000000000","updated":"2024-03-28 23:10:24.000000000","submitted":"2024-03-28 23:06:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914687-qa-2024-1-release","meta_rev_id":"8a459235256b318d8d3b94fc4a3b3f5029150695","_number":914687,"virtual_id_number":914687,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-28 23:06:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-03-28 19:02:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2024-03-28 21:24:13.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2024-03-28 21:24:13.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2024-03-28 19:01:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-28 18:33:01.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2024-03-28 20:46:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-28 21:24:13.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"6bb33e967117628683948b454cceca96ce5a6705","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-28 17:45:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d1ee9ceaf19444631cbf9978c884e0bde1cde70","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-28 17:54:52.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"50fea1f327498d53f4fb46c7d64db16c9475cfda","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-03-28 18:33:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9926832693cf6ee9dccf601cbfc6f0eef3cea11e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-28 19:01:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af700b790de6c76bb31dedc6348935744fcdb9fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-28 19:01:57.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"07b885f2e0f8aefea9c6c7d38a3541c29fcff7cc","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-03-28 19:02:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1a2b2d95776351642c599196ccd37b27cc0ce636","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 20:46:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44d8dc9b4bec432ea01eb3b344133f9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13a79bff1cee4e09bd6b12242e50729c : SUCCESS in 15m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/ea90f56a706d482ea11560c2600818b1 : SUCCESS in 1h 09m 25s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/3ec1f000acb44c65b143ff4a7e63200c : SUCCESS in 51m 41s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/787ec1aa1b2842b0b61eb370fd0808ee : SUCCESS in 1h 37m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dfd8cfb86dc84244ba66e14f56e8d7d2 : SUCCESS in 1h 03m 41s\n- ironic-grenade https://zuul.opendev.org/t/openstack/build/6e38270f710f4b478838c4c6968e2f11 : SUCCESS in 48m 14s (non-voting)\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/74d52ef184e34d76a62518aaef2fa596 : FAILURE in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47099c20904b3e9df5c6ef17fc916ae6aaaee519","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-03-28 21:24:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"657ccddf88086afbe212a79159063cfdb78effed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 21:24:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3ee63cf6d28b5bf052557f6c714a0146df1e230e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 23:06:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8ef957c9da642659f58b792ee58c842\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2e1957a1d9443fcb3375c3eee509a31 : SUCCESS in 27m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/ca24727fa4d34c71b5fd492450b9945b : SUCCESS in 57m 33s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/48dc76468f20404a8c69440b622e5e7b : SUCCESS in 56m 39s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/e376fc8d7cda4ea89ecf6977c08fd5b0 : SUCCESS in 1h 25m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d5585d2822fe44dea26e6ecdde449bb4 : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c4bc6295401c795a7f376a74a936a6cb9d5f27a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 23:06:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8a459235256b318d8d3b94fc4a3b3f5029150695","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 23:10:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c23dcbc30324e548cbd8fbbf284c76d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec6003e83e984e8a880799462c0b8543 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"57a1265a1e34d0722e9cfd0b24da8b3989143566","revisions":{"71e1a4c92f943b2acc7151a565ab53a81d24d1a2":{"kind":"REWORK","_number":1,"created":"2024-03-28 17:45:24.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/87/914687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/87/914687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/87/914687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/87/914687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/87/914687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/87/914687/1"}}},"commit":{"parents":[{"commit":"4d002199f5e893c6f57fffdcbfbf1e7dc836f6d9","subject":"Use re2 compatible regexes in branch matchers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/4d002199f5e893c6f57fffdcbfbf1e7dc836f6d9"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-03-28 17:39:57.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-03-28 17:39:57.000000000","tz":-420},"subject":"Fix grenade-skip-level job to run on respective stable branch","message":"Fix grenade-skip-level job to run on respective stable branch\n\ngrenade-skip-level job has specific setting of releases it upgrade\nand depends on the setting of grenade_from_branch and grenade_to_branch\nthat job version is supposed to run on specific stable branch.\n\nTo control where this job should run, we used the negative regex which\nrequired to be updated in every release otherwise zuul will pick wrong\nversion of this job. Better is use the branch to run explicitly and\nother stable branches will have their own version and branch specific\nvariable.\n\nChange-Id: If5df9bae3c62042493ec7c39ff3da7c09e830619\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/71e1a4c92f943b2acc7151a565ab53a81d24d1a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/71e1a4c92f943b2acc7151a565ab53a81d24d1a2"}]},"branch":"refs/heads/stable/2023.1"},"57a1265a1e34d0722e9cfd0b24da8b3989143566":{"kind":"REWORK","_number":2,"created":"2024-03-28 19:01:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/87/914687/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grenade","ref":"refs/changes/87/914687/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grenade refs/changes/87/914687/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grenade refs/changes/87/914687/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grenade refs/changes/87/914687/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grenade refs/changes/87/914687/2"}}},"commit":{"parents":[{"commit":"4d002199f5e893c6f57fffdcbfbf1e7dc836f6d9","subject":"Use re2 compatible regexes in branch matchers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/4d002199f5e893c6f57fffdcbfbf1e7dc836f6d9"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-03-28 17:39:57.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-03-28 18:59:04.000000000","tz":-420},"subject":"Fix grenade-skip-level job to run on respective stable branch","message":"Fix grenade-skip-level job to run on respective stable branch\n\ngrenade-skip-level job has specific setting of releases it upgrade\nand depends on the setting of grenade_from_branch and grenade_to_branch\nthat job version is supposed to run on specific stable branch.\n\nTo control where this job should run, we used the negative regex which\nrequired to be updated in every release otherwise zuul will pick wrong\nversion of this job. Better is not to use the branches rule and let zuul\npick the right version of job as per stable branches and job definition.\n\nChange-Id: If5df9bae3c62042493ec7c39ff3da7c09e830619\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/57a1265a1e34d0722e9cfd0b24da8b3989143566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grenade/commit/57a1265a1e34d0722e9cfd0b24da8b3989143566"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
