)]}'
{"id":"openstack%2Fgrian-ui~950343","triplet_id":"openstack%2Fgrian-ui~master~Ib39d20861928a3052eaa9e9adc3f984ae20aad03","project":"openstack/grian-ui","branch":"master","topic":"bootstraping","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-05-20 15:30:10.000000000","reason":"Change was submitted"},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2025-05-20 15:17:31.000000000","reason":"\u003cGERRIT_ACCOUNT_6413\u003e replied on the change","reason_account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}},"hashtags":[],"change_id":"Ib39d20861928a3052eaa9e9adc3f984ae20aad03","subject":"set up basic unittesting","status":"MERGED","created":"2025-05-19 17:36:01.000000000","updated":"2025-05-20 15:31:34.000000000","submitted":"2025-05-20 15:30:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950343-bootstraping","meta_rev_id":"555b254663456b5e9bdf561b8009f2b2a624d770","_number":950343,"virtual_id_number":950343,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-20 15:30:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-20 15:17:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-20 15:17:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-19 17:47:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-20 11:28:32.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"}],"messages":[{"id":"09e80bc6e0713f23bf77511f30affc8e3cb0a8b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-19 17:36:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae0b80c921d4200ffa19ab4b7f0b2187d80a49f1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-19 17:37:16.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28af9cc625a3352e3124b115d2cce5724f26e6f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 17:47:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca2eaa41b34943d5b95efe1fd4531855\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6763cc31f4504e07877bdad3d59b7164 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb6af693070e4ef7bfd7cab30a0ad1bd : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7600f560f8b4c3090c36e59d53dfefd : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"cbfca9a04bba0ddd7a80da6e4834c5b9f21d4792","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-19 23:53:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ab36c4970691275c3ef89c1f0df0c0886780b9ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 00:02:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adb6c844ea3445e3a99ae67036422682\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8b401acee484f98b92ce996e13d3ecd : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82b7d4a5b34e49609c1d189f88983139 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ee1d3a87a5147f9ac6a386ebf07577e : SUCCESS in 6m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66cca8dda5174001ac38a87c790eab42 : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1113950ea87945e28861666fab027fb7 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9cf19df7f16f40b89987b9edc61507d8 : SUCCESS in 7m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7bfeece56a39549f5ec45e75287a13d024a8b488","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-05-20 15:17:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25b142df11b8312b5415a0e1919ad3f01012765a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 15:18:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87421593fc889132ba5b78d7338b8ab8ee45098c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 15:30:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59e7372a484540ae8287e79003c3da42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/038be59e31a642ecb6731a19f00cd2de : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5ec1ca869a54f57b3c4b11339bbb172 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f0f0f174b334edebad586b739d909f1 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a294fba740945e0876d7b3f62bf39ea : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/607a5925e20f421c8cba6f4a50743f50 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"28d987d41a56bd43328a6c3f85910df2d8915a55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 15:30:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"555b254663456b5e9bdf561b8009f2b2a624d770","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 15:31:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bef4d54b7f94980864efbbeeeb0d290\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3137d2d1c7944b709463ee411cf84fac : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7dcbbd4592a45b88dc41e965e9131d6 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8346769c9b5fd5e33de9d49fd07d25db0caad321","revisions":{"06691d5369f7ec3d30469463f54f864063460cca":{"kind":"REWORK","_number":1,"created":"2025-05-19 17:36:01.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/43/950343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/43/950343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/43/950343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/43/950343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/43/950343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/43/950343/1"}}},"commit":{"parents":[{"commit":"93aa237b55955c7f78fb67085f22124c0bccb79c","subject":"add releasenote and docs config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/93aa237b55955c7f78fb67085f22124c0bccb79c"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-05-19 17:33:49.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-05-19 17:33:49.000000000","tz":0},"subject":"set up basic unittesting","message":"set up basic unittesting\n\nthis chnage setup up unit testing with\nstestr as the test runner and\ndjango.test.SimpleTestCase as the base of our tests.\n\nThis avoid the depency on manage.py or run_tests.sh\n\nChange-Id: Ib39d20861928a3052eaa9e9adc3f984ae20aad03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/06691d5369f7ec3d30469463f54f864063460cca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/06691d5369f7ec3d30469463f54f864063460cca"}]},"branch":"refs/heads/master"},"8346769c9b5fd5e33de9d49fd07d25db0caad321":{"kind":"REWORK","_number":2,"created":"2025-05-19 23:53:45.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/43/950343/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/43/950343/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/43/950343/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/43/950343/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/43/950343/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/43/950343/2"}}},"commit":{"parents":[{"commit":"93aa237b55955c7f78fb67085f22124c0bccb79c","subject":"add releasenote and docs config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/93aa237b55955c7f78fb67085f22124c0bccb79c"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-05-19 17:33:49.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-05-19 23:51:57.000000000","tz":0},"subject":"set up basic unittesting","message":"set up basic unittesting\n\nthis chnage setup up unit testing with\nstestr as the test runner and\ndjango.test.SimpleTestCase as the base of our tests.\n\nThis avoid the depency on manage.py or run_tests.sh\n\nChange-Id: Ib39d20861928a3052eaa9e9adc3f984ae20aad03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/8346769c9b5fd5e33de9d49fd07d25db0caad321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/8346769c9b5fd5e33de9d49fd07d25db0caad321"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
