)]}'
{"id":"openstack%2Fgrian-ui~952750","triplet_id":"openstack%2Fgrian-ui~master~Ia4fb4cf1530d712c331389370cda8041ded972e0","project":"openstack/grian-ui","branch":"master","topic":"volume-metrics-basic","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-07-18 11:38:50.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2026-03-26 15:16:54.000000000","reason":"Change was abandoned"},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2026-03-26 15:16:54.000000000","reason":"Change was abandoned"},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2026-03-26 15:16:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia4fb4cf1530d712c331389370cda8041ded972e0","subject":"[WIP] Volume metrics admin dashboard","status":"ABANDONED","created":"2025-06-17 10:32:52.000000000","updated":"2026-03-26 15:16:54.000000000","total_comment_count":43,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"0910f974d2d95c4cc47a375e55b66aa8513bfcc7","_number":952750,"virtual_id_number":952750,"owner":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-09 14:21:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2025-07-15 21:21:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-17 10:33:26.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-06-17 10:43:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-17 15:14:42.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2025-07-09 21:51:10.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"CC"}],"messages":[{"id":"33fffb1fbbe846815470bde0c77f4f1db37d5a5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-06-17 10:32:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b81ce0409516d14117cd43f9288c1317741a517","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-06-17 10:34:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc37d9f9ac70654a00fbf7bdb5a72685b862984f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 10:43:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/502bdb360f8e4794b15c358412ad0406\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e77d33f8d61e402d82fd5c3cc82c2484 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7db779bc6e1440a5af0872a5ca707ded : FAILURE in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bdbeb6829631412fa1e62dd32f4596f8 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1310ada807854193a30ec9270612588b : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17c4fc1f57014f938760fc2a5ee66434 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/366c3579cf0d46fb838c63a4a7a4c588 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"8933542385cba1e122256ad66cc9f1d3adfc3dbd","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-17 11:37:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"424d48fdf5d90c57807af3d99e31c1dd702890ce","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-06-17 15:14:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"125595ffa9a3626c02f84d24d3aabf32ecbe1686","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-17 16:10:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"afbc649fb88a9a569335461c54744c983c56ce48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-06-24 10:39:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d35d17a8e2ee578037fce3b44ecd680db3302053","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 10:45:58.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f3b07d34f0840839a9742e1ed051b5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/012dc26ad1b042a0a906c1ca678485bf : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b4826c3f10d471fa7a8763fcd1d17ab : FAILURE in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee7a7261cc4b496f968fa8d98cb71ad1 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ad0f18541ec4086983a5931bdd6428c : SUCCESS in 2m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/400c85e06e4e4841b3b547a57a93a443 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/757c8c6dd2a045379263ccb71d8a57e0 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"e8a24a17dc6361005172bf94c09502937acdc558","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-06-24 16:15:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"92a25c6066962838f419dbfdd6a0a3b24b06a9a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-06-24 16:16:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6f49ed3ef4d24f5abffe816dcc08c1c7142281c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-06-24 16:22:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0da239565ddcda21555a23079b876efeed0ae5f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 16:33:03.000000000","message":"Patch Set 5: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ff183e55e934d478831a38b4ee48d0e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab9b13585d7e4482b5fee79f9c6e0aa1 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43002e1e771e435bb42ca08266bbe4a5 : FAILURE in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eaf0547d9f064427a06d620f1457ed3e : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd281e62e8704de6b383a2d2d83be756 : SUCCESS in 2m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60e28532ff924c80a34fcf7e45086c84 : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/80740601204741e3bd79d90cdecc3135 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":5},{"id":"bcdbd589b0377f1b266d522b64cb33a9fc749345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-06-24 20:06:23.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3d87cc29634418a09e6c3038340f1c3080d5081b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-06-24 20:12:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3da62c086a33e718e030ccdefcf3db9c21bd21af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 20:19:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe1fc947cd604614b62fd2cbfa3e87fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8991197ab3454b0a941d90d55d22a722 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6acdacb060484e869968929791180ea1 : FAILURE in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5aec440165094b4dafcab9acf7bfd070 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/196af65272fa4436b358de77f78c6370 : SUCCESS in 2m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f50925ebc2ec440e9df5fc44dd3c9323 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e93409d27298481393559965f684bb0b : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":7},{"id":"d675e4a5b4251dabbc8ed39fdc38b789a8a29c22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-06-24 20:37:00.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"ee9264a4c5a708111e864a1c171088741183d0d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 20:44:53.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dbcc883f6fb4f32bc82998093725715\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088d422ae006447899dece7d2ae168af : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab0821ac853b4efaba512f93ecaa0dfe : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d71a3a2bd214f7497e2cb32cffd9140 : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4512ba9f26e450488d0748a63709e57 : SUCCESS in 6m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d6f69393296e40f3a3e7077f48aef3b8 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/87d1db94d89e429f9027a5550a6bd21d : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":8},{"id":"d445232df8734b90a50361ff1a033175e95d15ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-06-30 13:50:54.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"45b10cb2ee7f7aff4f239880f4714ee667ffce8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-06-30 13:53:24.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1fd5e69944d63bbec5e2c9d072c4a52d4b48f3d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 14:06:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f453250cecc14ce095d4c237c881ff7c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be234e67a96a45e4ae3dceda502cf57a : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b34041d5c3449ae9a88cf749f663e64 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2f15649129c494282e3c455b9801b9e : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6659a3dd27a0497d9636db09284a6a30 : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1f2728e85754627ac6eb0004bb478c4 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9bcd2a30af7a4a108d88983ae51e532a : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":10},{"id":"6af90f7ee537a81d99568440daafdce7fd49526d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-07 11:22:37.000000000","message":"Patch Set 10:\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b886dc7e8ca29ee4a8d60888641a5d13c5d30710","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-07-07 16:03:17.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"16f467c85b23fe4dbcc26f0b9cae640233448d95","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-07-07 16:03:39.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4bb23b9a8903db9ee6cd0b266f135ab5dfda9c76","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-07-08 09:47:42.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"618f6c0afb217a565697306031a8cc650baf6991","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-07-08 10:07:36.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"f5a08e8bd61b7a320844f80ddf677dda88a32b00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 10:19:49.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fcda89d3cb24b1cb18b0c0d7f06c598\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e98cd06f394f09a322097105242707 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a28f967e4b4467cad4b63a29410a0a8 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e13e5872a684850937f8e849aacdab2 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7646018f601b4d9fb9a65fb1f9ebe3c2 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d01edc73e57a4b199ec19566aa996807 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0d5efd3a28d545b7b3589a3214312a73 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":11},{"id":"992b7e3119cbc5bdc3d547998e342ef40510b42d","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-07-08 16:23:37.000000000","message":"Patch Set 11: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5b821102cdf1139b1acaf7a27171a5be699afd35","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-08 16:30:36.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a12c396e8ba0418798cad6e4098bc3b5b407c06d","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-07-09 13:39:48.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8053835436345fc66bdce89c189677fc4ddabcc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-07-09 14:09:46.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":12},{"id":"e02e843f8ea08858c10aefd12eb498f8bdad3f01","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-09 14:11:20.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"0a6d201c5eba81ac924e8a8a66dd110b667a2419","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 14:21:43.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6916585e11e4bad86ed3bfdda233445\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee2cb4eb775744b88dc03bd0693077e9 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9878f807c5c040b29bb4b0f50a8bc5ce : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc179548ce7a43d59cbee688545a33c5 : SUCCESS in 7m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ef8b31f9ec34980b1d70a127597a10f : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5162cc2900b1483ea3f52193cb5792a0 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ab31a4adad434a4d97cb6c33ae530b65 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":12},{"id":"ebc71dd5cb377eee55009c9c2f55954cd741bec0","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-07-09 21:51:10.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f3e3bc84f3c7d379ef9ac30f2be0a04cfa636ad6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-10 01:52:05.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"af3585f5264a4016f290d31c6ed6ee24cc59569d","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-07-10 07:12:25.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0b24c387689252b5f225803cf8b336ac3be835c9","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-07-15 16:07:33.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"669a835b1ccc851c8429a8cf3c622bbaace871f5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-15 19:43:41.000000000","message":"Patch Set 12:\n\n(7 comments)","accounts_in_message":[],"_revision_number":12},{"id":"acfa801b9809d5a2341e7e9339e861605542d5cf","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-15 19:43:49.000000000","message":"Patch Set 12: Code-Review-1","accounts_in_message":[],"_revision_number":12},{"id":"742548d15582265fc12f9c57d4d8c829e3ee5b37","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-15 21:21:11.000000000","message":"Patch Set 12: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"95b0f059f44b51c0d35f0d99a4e5cc6fc198a5e7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-15 21:21:20.000000000","message":"Patch Set 12: Code-Review-1","accounts_in_message":[],"_revision_number":12},{"id":"be18f9b0dca89d554e2e34457ffb900f3796b213","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2025-07-18 08:54:55.000000000","message":"Patch Set 12:\n\n(7 comments)","accounts_in_message":[],"_revision_number":12},{"id":"d7ffe33c7a9ccad4165baee283dd918aabb8c42a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-18 11:38:50.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"0910f974d2d95c4cc47a375e55b66aa8513bfcc7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2026-03-26 15:16:54.000000000","message":"Abandoned\n\nAbandoning this change in favor of https://review.opendev.org/c/openstack/grian-ui/+/982253","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"dceca39701953b9d6bfbdf0f255f68c27b386e53","revisions":{"94c17f79993a3b2898b11ce95235386cbf024c92":{"kind":"REWORK","_number":1,"created":"2025-06-17 10:32:52.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/50/952750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/50/952750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/1"}}},"commit":{"parents":[{"commit":"6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50","subject":"Add CLAUDE.md to .gitignore for local development use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-17 10:30:59.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-17 10:32:06.000000000","tz":0},"subject":"[WIP] Volume metrics admin dashboard","message":"[WIP] Volume metrics admin dashboard\n\nChange-Id: Ia4fb4cf1530d712c331389370cda8041ded972e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/94c17f79993a3b2898b11ce95235386cbf024c92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/94c17f79993a3b2898b11ce95235386cbf024c92"}]},"branch":"refs/heads/master"},"cb89a2cbee0ff797aaad32b5cb9037f2802a89b3":{"kind":"REWORK","_number":2,"created":"2025-06-24 10:39:21.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/50/952750/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/50/952750/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/2"}}},"commit":{"parents":[{"commit":"6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50","subject":"Add CLAUDE.md to .gitignore for local development use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-17 10:30:59.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-24 10:37:42.000000000","tz":0},"subject":"[WIP] Volume metrics admin dashboard","message":"[WIP] Volume metrics admin dashboard\n\nDepends on https://review.opendev.org/c/openstack/grian-ui/+/952608\n(using as reference implementation)\n\nChange-Id: Ia4fb4cf1530d712c331389370cda8041ded972e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/cb89a2cbee0ff797aaad32b5cb9037f2802a89b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/cb89a2cbee0ff797aaad32b5cb9037f2802a89b3"}]},"branch":"refs/heads/master"},"a676e780b5ab223c5c36614955a613070fdc1ad5":{"kind":"REWORK","_number":3,"created":"2025-06-24 16:15:15.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/50/952750/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/50/952750/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/3"}}},"commit":{"parents":[{"commit":"6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50","subject":"Add CLAUDE.md to .gitignore for local development use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-17 10:30:59.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-24 16:15:07.000000000","tz":0},"subject":"[WIP] Volume metrics admin dashboard","message":"[WIP] Volume metrics admin dashboard\n\nDepends on https://review.opendev.org/c/openstack/grian-ui/+/952608\n(using as reference implementation)\n\nChange-Id: Ia4fb4cf1530d712c331389370cda8041ded972e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/a676e780b5ab223c5c36614955a613070fdc1ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/a676e780b5ab223c5c36614955a613070fdc1ad5"}]},"branch":"refs/heads/master"},"4956f8c0345e231441709f3d309de15d398878a4":{"kind":"REWORK","_number":4,"created":"2025-06-24 16:16:47.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/50/952750/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/50/952750/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/4"}}},"commit":{"parents":[{"commit":"6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50","subject":"Add CLAUDE.md to .gitignore for local development use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-17 10:30:59.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-24 16:16:37.000000000","tz":0},"subject":"[WIP] Volume metrics admin dashboard","message":"[WIP] Volume metrics admin dashboard\n\nDepends on https://review.opendev.org/c/openstack/grian-ui/+/952608\n(using as reference implementation)\n\nChange-Id: Ia4fb4cf1530d712c331389370cda8041ded972e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/4956f8c0345e231441709f3d309de15d398878a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/4956f8c0345e231441709f3d309de15d398878a4"}]},"branch":"refs/heads/master"},"a6255fd2da79865e16c801aba0a7d45bbf07fec6":{"kind":"REWORK","_number":5,"created":"2025-06-24 16:22:16.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/50/952750/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/50/952750/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/5"}}},"commit":{"parents":[{"commit":"6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50","subject":"Add CLAUDE.md to .gitignore for local development use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-17 10:30:59.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-24 16:22:08.000000000","tz":0},"subject":"[WIP] Volume metrics admin dashboard","message":"[WIP] Volume metrics admin dashboard\n\nDepends on https://review.opendev.org/c/openstack/grian-ui/+/952608\n(using as reference implementation)\n\nChange-Id: Ia4fb4cf1530d712c331389370cda8041ded972e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/a6255fd2da79865e16c801aba0a7d45bbf07fec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/a6255fd2da79865e16c801aba0a7d45bbf07fec6"}]},"branch":"refs/heads/master"},"660587418899d0b7fe6e476cc669fe0fa653f5ca":{"kind":"REWORK","_number":6,"created":"2025-06-24 20:06:23.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/50/952750/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/50/952750/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/6"}}},"commit":{"parents":[{"commit":"6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50","subject":"Add CLAUDE.md to .gitignore for local development use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-17 10:30:59.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-24 20:06:15.000000000","tz":0},"subject":"[WIP] Volume metrics admin dashboard","message":"[WIP] Volume metrics admin dashboard\n\nDepends on https://review.opendev.org/c/openstack/grian-ui/+/952608\n(using as reference implementation)\n\nChange-Id: Ia4fb4cf1530d712c331389370cda8041ded972e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/660587418899d0b7fe6e476cc669fe0fa653f5ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/660587418899d0b7fe6e476cc669fe0fa653f5ca"}]},"branch":"refs/heads/master"},"61994e4dfc8bc69a0bc1472d5e6221e55ec26d7e":{"kind":"REWORK","_number":7,"created":"2025-06-24 20:12:47.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/50/952750/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/50/952750/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/7"}}},"commit":{"parents":[{"commit":"6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50","subject":"Add CLAUDE.md to .gitignore for local development use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-17 10:30:59.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-24 20:12:26.000000000","tz":0},"subject":"[WIP] Volume metrics admin dashboard","message":"[WIP] Volume metrics admin dashboard\n\nDepends on https://review.opendev.org/c/openstack/grian-ui/+/952608\n(using as reference implementation)\n\nChange-Id: Ia4fb4cf1530d712c331389370cda8041ded972e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/61994e4dfc8bc69a0bc1472d5e6221e55ec26d7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/61994e4dfc8bc69a0bc1472d5e6221e55ec26d7e"}]},"branch":"refs/heads/master"},"a3a9632b33e9846bb180167b5666103450866c0c":{"kind":"REWORK","_number":8,"created":"2025-06-24 20:37:00.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/50/952750/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/50/952750/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/8"}}},"commit":{"parents":[{"commit":"6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50","subject":"Add CLAUDE.md to .gitignore for local development use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-17 10:30:59.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-24 20:36:53.000000000","tz":0},"subject":"[WIP] Volume metrics admin dashboard","message":"[WIP] Volume metrics admin dashboard\n\nDepends on https://review.opendev.org/c/openstack/grian-ui/+/952608\n(using as reference implementation)\n\nChange-Id: Ia4fb4cf1530d712c331389370cda8041ded972e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/a3a9632b33e9846bb180167b5666103450866c0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/a3a9632b33e9846bb180167b5666103450866c0c"}]},"branch":"refs/heads/master"},"512de44e04c50be2e7eb77b7dc9b1d9754174f5a":{"kind":"REWORK","_number":9,"created":"2025-06-30 13:50:54.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/50/952750/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/50/952750/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/9"}}},"commit":{"parents":[{"commit":"6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50","subject":"Add CLAUDE.md to .gitignore for local development use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-17 10:30:59.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-30 13:50:37.000000000","tz":0},"subject":"[WIP] Volume metrics admin dashboard","message":"[WIP] Volume metrics admin dashboard\n\nDepends on https://review.opendev.org/c/openstack/grian-ui/+/952608\n(using as reference implementation)\n\nChange-Id: Ia4fb4cf1530d712c331389370cda8041ded972e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/512de44e04c50be2e7eb77b7dc9b1d9754174f5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/512de44e04c50be2e7eb77b7dc9b1d9754174f5a"}]},"branch":"refs/heads/master"},"4e1ccfbfd90a6b0f05293d0ed67a380aa2d89159":{"kind":"REWORK","_number":10,"created":"2025-06-30 13:53:24.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/50/952750/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/50/952750/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/10"}}},"commit":{"parents":[{"commit":"6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50","subject":"Add CLAUDE.md to .gitignore for local development use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-17 10:30:59.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-30 13:52:54.000000000","tz":0},"subject":"[WIP] Volume metrics admin dashboard","message":"[WIP] Volume metrics admin dashboard\n\nDepends on https://review.opendev.org/c/openstack/grian-ui/+/952608\n(using as reference implementation)\n\nChange-Id: Ia4fb4cf1530d712c331389370cda8041ded972e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/4e1ccfbfd90a6b0f05293d0ed67a380aa2d89159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/4e1ccfbfd90a6b0f05293d0ed67a380aa2d89159"}]},"branch":"refs/heads/master"},"f01eceb82ea6e79fd77f299e2e661afb82deab08":{"kind":"REWORK","_number":11,"created":"2025-07-08 10:07:36.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/50/952750/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/50/952750/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/11"}}},"commit":{"parents":[{"commit":"6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50","subject":"Add CLAUDE.md to .gitignore for local development use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-17 10:30:59.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-07-08 10:06:59.000000000","tz":0},"subject":"[WIP] Volume metrics admin dashboard","message":"[WIP] Volume metrics admin dashboard\n\nDepends on https://review.opendev.org/c/openstack/grian-ui/+/952608\n(using as reference implementation)\n\nChange-Id: Ia4fb4cf1530d712c331389370cda8041ded972e0\nSigned-off-by: Victoria Martinez de la Cruz \u003cvictoria@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/f01eceb82ea6e79fd77f299e2e661afb82deab08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/f01eceb82ea6e79fd77f299e2e661afb82deab08"}]},"branch":"refs/heads/master"},"dceca39701953b9d6bfbdf0f255f68c27b386e53":{"kind":"REWORK","_number":12,"created":"2025-07-09 14:09:46.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/50/952750/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/50/952750/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/50/952750/12"}}},"commit":{"parents":[{"commit":"6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50","subject":"Add CLAUDE.md to .gitignore for local development use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/6dbe2a8b72b5b8ab0c66a4bff9eb64af4de3ec50"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-06-17 10:30:59.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2025-07-09 14:08:53.000000000","tz":0},"subject":"[WIP] Volume metrics admin dashboard","message":"[WIP] Volume metrics admin dashboard\n\nDepends on https://review.opendev.org/c/openstack/grian-ui/+/952608\n(using as reference implementation)\n\nChange-Id: Ia4fb4cf1530d712c331389370cda8041ded972e0\nSigned-off-by: Victoria Martinez de la Cruz \u003cvictoria@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/dceca39701953b9d6bfbdf0f255f68c27b386e53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/dceca39701953b9d6bfbdf0f255f68c27b386e53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
