)]}'
{"id":"openstack%2Fgrian-ui~974088","triplet_id":"openstack%2Fgrian-ui~master~I944e659f5616c43ed968c0fb8de70d2146cbf959","project":"openstack/grian-ui","branch":"master","topic":"py313","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-10 20:41:49.000000000","reason":"Change was submitted"},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2026-04-10 19:26:56.000000000","reason":"\u003cGERRIT_ACCOUNT_6413\u003e replied on the change","reason_account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}},"hashtags":[],"change_id":"I944e659f5616c43ed968c0fb8de70d2146cbf959","subject":"Declare Python 3.13 support","status":"MERGED","created":"2026-01-21 15:21:04.000000000","updated":"2026-04-10 20:42:50.000000000","submitted":"2026-04-10 20:41:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974088-py313","meta_rev_id":"92460c60884a6cde3667589f6455ee3dba423397","_number":974088,"virtual_id_number":974088,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-10 20:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-04-10 16:33:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-10 19:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-10 19:26:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}],"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-21 15:34:55.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-01-21 15:45:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-13 08:24:22.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"CC"},{"updated":"2026-03-27 10:35:54.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"CC"},{"updated":"2026-04-08 12:48:18.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"}],"messages":[{"id":"92ebc6b28bf5dff43bc71325197a770c7b936da4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-21 15:21:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9dc7375d5f0c3ca6a59c1cf8a80108fbe9b1f32","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-01-21 15:34:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9126b8009fe021b474efb868571b52cf66e50eb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 15:45:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c9229ac92c3423f9de0aaa4c083fab0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/af5dbd30d7e747399e6e941f8de218cf : FAILURE in 3m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8e7808936d78476b9f8b6f5024c79770 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dff9f16f00046e8b983d5528c4403ba : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/977219eebb03410591d1ce54f49b70e0 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b9a6d824f1e4d288af540285d233c2a : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/227cff9aa4b04f18b526ac80717facdc : RETRY_LIMIT in 2m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5716d5831fe04b798cd8e7729f5fa248 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4c06c6ddb7b70381ecef5bd34b9581988a4823a8","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-03-13 08:24:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9eac5ffd73fff7869a316f00157edd363b9b88d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 08:46:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/627fc0ffeb8c47a9a7b863bb09fe5d5b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a4764dca95f345bc94f17852c71b444f : FAILURE in 2m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8b3eab82bbe84d6b8617a04365741115 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/985d06006a93494db07349783da8127f : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/732e934bfbea4dfdb77e5b2f0ffb0bc3 : FAILURE in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64cf81c15aad48bbb0dbc19baeee02f3 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/196fba7eddd74e1db986a3c49c3a209f : RETRY_LIMIT in 1m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6f80cf5da7684fd4be7b92f8eef0a5ac : RETRY_LIMIT in 1m 39s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c51a78e487ea4be2903532a54a471fbd : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8fa522f4f6f9b343fb50c7c9fe8c8e9285486e2b","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2026-03-27 10:35:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"614cc3d45b2695ce24464ebf62d29ae6ca6d9644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 10:49:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/682ba1a5ffe04f89a15609e1778febef\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c6bae67a2b364884a287cb2b373aace7 : FAILURE in 2m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/460d0570da8c48668a22ba7e2d8c31e7 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f42d1077d544f7a0248bab3862e3f5 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8623cef7ac0a43068e6db14175ac956b : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abdd03d70ce34c5f96650bf722e39d09 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e94ca126d98a429b820b84db42e93de1 : RETRY_LIMIT in 1m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/15d07af5b64946a6a5a209138936d6a9 : RETRY_LIMIT in 1m 21s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/18f79c0d0b384c5cb98c8a964b9e1440 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":1},{"id":"bf95dd35410e01aae7bf27f8ac6b44f465a0c586","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2026-04-08 12:22:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":2},{"id":"b3e928f3c9a4c817ac47f2343d3a2ec3108d07eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 12:38:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf46c85762814098a209282fe1a57fa3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a7b4909cf63740099f79074b331c12bd : FAILURE in 1m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f9594d4069a49a785f548faf0db713c : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c85b33a527e04dc5925771430acbfa48 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a15b2d55d82040f1b453d285f69ef64a : SUCCESS in 7m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/464dd934b31e4020a44321efe017146e : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da9a3eb536274c9e8fdfc82af80ca97f : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/17b4249bdbb4447099be824e6305b9db : FAILURE in 3m 58s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/49f3681c474247a59921c271679cfc6d : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"299d985a6cc6a4d98fa0f6c195af315a66696790","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2026-04-08 12:48:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a084e5e215e8f285de7e98b2faeca0ccd9daf955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 13:02:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b7336d9d8fd4560b1d568be5f0d1615\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/77e164ab1ed94e65b22377ccfcbef596 : FAILURE in 1m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8c1e109bd6044e7888ad38bbffacf286 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9771d0eea9b44009218cd1e7dc3f9b8 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daa1ab4850fa4d36b8357ae37a555b4f : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19d344bb437847239158421e1e271fd8 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/606aaa4eac5845e98fe27a32063b2153 : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/683ddb5115514d5f8f12ecfe595b3adf : FAILURE in 7m 28s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/89601010a79349f2a755bfeee8129e78 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":2},{"id":"2c320593b06bf405d09d48f5ff9e89c916135f21","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-04-08 13:08:34.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8515bc5d749f68b316a44d94855fde4c2efa2ff6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-04-08 13:10:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5172cdbb021fee78eb4d2c80e342e2f84fac74c0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-04-08 13:21:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f9067c1d5ef91e460714250668de1947c3b660a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 13:21:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36ad4717cc234d30a812ac3586ba8dad\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/18f97a713bc34b74b213fd7a76f644bb : FAILURE in 2m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/84e41657e4c449f3a745d2a10209fb99 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e46f3eecec412a9231a5ef10664a05 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8131de159e904159aaa4dff60ac07f29 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/798be477cb684405bf90be630f635f47 : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28f6f8a7f9984420a24748d535ca6b38 : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6695c1752cd946a28d43856f52e69666 : FAILURE in 7m 30s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3e7a3ec738eb41bbb3977aae152fcca5 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"8cb2ac38dc7bd466cc6a2bcbb3c217d1897df9b0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-04-08 13:28:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6e460d1913aae0fe67b19701414e5db3645b284","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-04-08 13:36:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"352628ecddc0015fcfd3c4dd6fdbb7c0f44b7448","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-04-10 16:28:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0bda4a2cd0d6303a7c1ed2b6e43ef08a6e1df99d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2026-04-10 16:33:51.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":3},{"id":"6df0bc0f637cc4af2d7ff1185e77bc4cfdac61b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 16:40:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56eeaf42ce7949f4bc96b87db8fcc69f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0810ca5dfaa046639daa824bb96caaa1 : SUCCESS in 2m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a0a39ff8370d44f99c50f4eaa22b8619 : SUCCESS in 2m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7415dc6353364ad48bcac499964c8e6b : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2806730aaa0343069913468c87ec7acf : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97dc2c2af3504751b7cff7efa4339dd5 : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c439dff926cb49e2b2ee702a2d269502 : SUCCESS in 2m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a179eba9234841a8a9430400e23dd927 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/51f0e83e229447bf8305b2464727d58b : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":3},{"id":"92c04a664277bd038a27c9b0d8d2805bd5cbc91d","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2026-04-10 19:26:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ae5a65ae881fbb1610f141484f28a5b4ccaba7ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 19:27:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"62af4f842f426e62778b372d34f94dd065f7cd71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:41:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a05cf63470342a28dc3897a19fd8f4f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f530edaf09694801b5b240e3caa95710 : SUCCESS in 3m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3c934bc399cf454dbcff6779db6450c4 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c26157ca264579aaa40b44d418bf4a : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c9de73512e7465b84e51405bb430bfa : SUCCESS in 7m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c72c1fc431574f0093ee835010eff022 : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4e5015489e04259be22162400ce750c : SUCCESS in 2m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c15560aa59c34dec81e4b53481fdb20e : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":3},{"id":"88ab933b9b6ef21d445086c1295e41e5c9816403","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:41:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"92460c60884a6cde3667589f6455ee3dba423397","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:42:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/371e5854d3d44049b94f7180a752b653\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c696f0a21bde414aa997128bb9c4389f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c396aa19ff9171d1c7957b72389583082aaf64c8","revisions":{"57fef8e06ccd39e62882ede83d72cb5b3272dc4a":{"kind":"REWORK","_number":1,"created":"2026-01-21 15:21:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/974088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/88/974088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/88/974088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/88/974088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/88/974088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/88/974088/1"}}},"commit":{"parents":[{"commit":"94868c0e95a8741e1247bfb4d9ac784862975817","subject":"Avoid conflicting number","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/94868c0e95a8741e1247bfb4d9ac784862975817"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:20:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:20:55.000000000","tz":540},"subject":"Declare Python 3.13 support","message":"Declare Python 3.13 support\n\nPython 3.13 is part of supported runtimes for 2026.1[1] and now is\ntested.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2026.1.html\n\nChange-Id: I944e659f5616c43ed968c0fb8de70d2146cbf959\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/57fef8e06ccd39e62882ede83d72cb5b3272dc4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/57fef8e06ccd39e62882ede83d72cb5b3272dc4a"}]},"branch":"refs/heads/master"},"190971b4e9742c1c3b750eb2176083cf6afad376":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-08 12:22:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/974088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/88/974088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/88/974088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/88/974088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/88/974088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/88/974088/2"}}},"commit":{"parents":[{"commit":"80101412188bbb22e094fb88a68553e83b17d5e5","subject":"Merge \"Remove unused pcre packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/80101412188bbb22e094fb88a68553e83b17d5e5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:20:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-08 12:22:26.000000000","tz":0},"subject":"Declare Python 3.13 support","message":"Declare Python 3.13 support\n\nPython 3.13 is part of supported runtimes for 2026.1[1] and now is\ntested.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2026.1.html\n\nChange-Id: I944e659f5616c43ed968c0fb8de70d2146cbf959\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/190971b4e9742c1c3b750eb2176083cf6afad376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/190971b4e9742c1c3b750eb2176083cf6afad376"}]},"branch":"refs/heads/master","description":"Rebase"},"c396aa19ff9171d1c7957b72389583082aaf64c8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-10 16:33:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/974088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/grian-ui","ref":"refs/changes/88/974088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/88/974088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/88/974088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/grian-ui refs/changes/88/974088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/grian-ui refs/changes/88/974088/3"}}},"commit":{"parents":[{"commit":"14f09957a96fa99474c7586d245c177eee37f24f","subject":"Add pypi-check tox env and fix README rst syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/14f09957a96fa99474c7586d245c177eee37f24f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:20:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-10 16:33:51.000000000","tz":0},"subject":"Declare Python 3.13 support","message":"Declare Python 3.13 support\n\nPython 3.13 is part of supported runtimes for 2026.1[1] and now is\ntested.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2026.1.html\n\nChange-Id: I944e659f5616c43ed968c0fb8de70d2146cbf959\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/c396aa19ff9171d1c7957b72389583082aaf64c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/grian-ui/commit/c396aa19ff9171d1c7957b72389583082aaf64c8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
