)]}'
{"id":"openstack%2Fha-guide~689648","triplet_id":"openstack%2Fha-guide~master~Ic1b16e454e957aab11bf8bab320c4509e1abf29f","project":"openstack/ha-guide","branch":"master","hashtags":[],"change_id":"Ic1b16e454e957aab11bf8bab320c4509e1abf29f","subject":"setup.cfg: Cleanup and update some urls to latest","status":"MERGED","created":"2019-10-20 15:08:10.000000000","updated":"2019-10-22 09:56:28.000000000","submitted":"2019-10-22 09:55:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689648-1571738121605-5cb63f8a","meta_rev_id":"7ee04a2b7922d6510dcc4dce4f50eed28accafe2","_number":689648,"virtual_id_number":689648,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 09:55:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"recommended":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"all":[{"value":1,"date":"2019-10-22 06:06:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":2,"date":"2019-10-22 09:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"all":[{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":1,"date":"2019-10-22 09:51:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 15:26:47.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2019-10-22 06:06:08.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-10-22 09:51:02.000000000","updated_by":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"reviewer":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"state":"REVIEWER"},{"updated":"2019-10-22 09:55:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d09c8e070dd56a3369f662e246ae11a465a20c16","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-20 15:08:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87fcbb48ce867cec03ca6f8ddaa213991f6da365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-20 15:12:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2268d26d7b447b2b9d2b719941f09d3 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6007596895a6281ce958c30bd014a7cb83f51f20","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-20 15:13:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c8d4effcf54a2b02cd4bcd42bd9b4bac27406ea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-20 15:17:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e09a4d48ee4b08bb4613e07ca1d484 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":2},{"id":"1b036087a63849691e9dd5dfac83477cec636c6f","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2019-10-21 15:26:47.000000000","message":"Patch Set 2: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"8f4784a0ce7a6a8ec6c722298279c928fb2b2361","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-10-22 03:15:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"82bea8011f9199fe84c24d0984c715ed648a3107","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-22 03:31:05.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1e271ee85a66f1825d4de6b08e343918cef1741b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 03:34:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42cc8c63dc7243a6bc5aa7be987c7de6 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"6c0ec6e07f91a7d5a5e9e6f1ba56b9bd8d1b4268","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-10-22 06:06:08.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7685a4e90fd28fa2a4aa982570132e8f3266a86f","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-10-22 09:51:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"43c0e029fbcbb2bc0a31352c641d479074595eee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:51:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"829addfe53785dd0e34a2c2588c30ee32213554e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:55:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d6a0b7db2004e29b4155c67c0e1e34a : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":3},{"id":"4c890312a0e1b3d270b044f8972f4c95174eb4c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:55:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7ee04a2b7922d6510dcc4dce4f50eed28accafe2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:56:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/6bb19f142c9e418ca0646cea75ecc7cd : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cd21310d1e23cb32f5fb0f0adca4172d7cee0ec8","revisions":{"707c9ce85c5621a6d111b21453ce44cdde5eec9b":{"kind":"REWORK","_number":1,"created":"2019-10-20 15:08:10.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/48/689648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ha-guide","ref":"refs/changes/48/689648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ha-guide refs/changes/48/689648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ha-guide refs/changes/48/689648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ha-guide refs/changes/48/689648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ha-guide refs/changes/48/689648/1"}}},"commit":{"parents":[{"commit":"c52a8cd0573b69e3a617be6ee4825f23b935a07e","subject":"Use promote for publishing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ha-guide/commit/c52a8cd0573b69e3a617be6ee4825f23b935a07e"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-20 15:05:15.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-20 15:07:38.000000000","tz":480},"subject":"setup.cfg: Cleanup","message":"setup.cfg: Cleanup\n\n- pbr hasn\u0027t need the hook configuration since forever [1]\n- The \u0027wheel\u0027 group is renamed to \u0027bdist_wheel\u0027 [2]\n- Update url to latest\n\n[1] https://github.com/openstack/pbr/commit/c84876dc0f559a66fec19b2f81f5717204b253e2\n[2] https://github.com/pypa/wheel/commit/f7c9878712390414c03c64e9afa55ea4f30e965b\n\nChange-Id: Ic1b16e454e957aab11bf8bab320c4509e1abf29f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ha-guide/commit/707c9ce85c5621a6d111b21453ce44cdde5eec9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ha-guide/commit/707c9ce85c5621a6d111b21453ce44cdde5eec9b"}]},"branch":"refs/heads/master"},"6a02fa89a7dd68623127a9109b2bbda552a9fa1d":{"kind":"REWORK","_number":2,"created":"2019-10-20 15:13:12.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/48/689648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ha-guide","ref":"refs/changes/48/689648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ha-guide refs/changes/48/689648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ha-guide refs/changes/48/689648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ha-guide refs/changes/48/689648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ha-guide refs/changes/48/689648/2"}}},"commit":{"parents":[{"commit":"c52a8cd0573b69e3a617be6ee4825f23b935a07e","subject":"Use promote for publishing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ha-guide/commit/c52a8cd0573b69e3a617be6ee4825f23b935a07e"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-20 15:05:15.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-20 15:12:55.000000000","tz":480},"subject":"setup.cfg: Cleanup","message":"setup.cfg: Cleanup\n\n- pbr hasn\u0027t need the hook configuration since forever [1]\n- The \u0027wheel\u0027 group is renamed to \u0027bdist_wheel\u0027 [2]\n- Update url to latest\n\n[1] https://github.com/openstack/pbr/commit/c84876dc0f559a66fec19b2f81f5717204b253e2\n[2] https://github.com/pypa/wheel/commit/f7c9878712390414c03c64e9afa55ea4f30e965b\n\nChange-Id: Ic1b16e454e957aab11bf8bab320c4509e1abf29f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ha-guide/commit/6a02fa89a7dd68623127a9109b2bbda552a9fa1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ha-guide/commit/6a02fa89a7dd68623127a9109b2bbda552a9fa1d"}]},"branch":"refs/heads/master"},"cd21310d1e23cb32f5fb0f0adca4172d7cee0ec8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-22 03:31:05.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/48/689648/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ha-guide","ref":"refs/changes/48/689648/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ha-guide refs/changes/48/689648/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ha-guide refs/changes/48/689648/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ha-guide refs/changes/48/689648/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ha-guide refs/changes/48/689648/3"}}},"commit":{"parents":[{"commit":"c52a8cd0573b69e3a617be6ee4825f23b935a07e","subject":"Use promote for publishing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ha-guide/commit/c52a8cd0573b69e3a617be6ee4825f23b935a07e"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-20 15:05:15.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-22 03:30:04.000000000","tz":480},"subject":"setup.cfg: Cleanup and update some urls to latest","message":"setup.cfg: Cleanup and update some urls to latest\n\n- pbr hasn\u0027t need the hook configuration since forever [1]\n- The \u0027wheel\u0027 group is renamed to \u0027bdist_wheel\u0027 [2]\n- Update the urls to latest\n\n[1] https://opendev.org/openstack/pbr/commit/c84876dc0f559a66fec19b2f81f5717204b253e2\n[2] https://github.com/pypa/wheel/commit/f7c9878712390414c03c64e9afa55ea4f30e965b\n\nChange-Id: Ic1b16e454e957aab11bf8bab320c4509e1abf29f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ha-guide/commit/cd21310d1e23cb32f5fb0f0adca4172d7cee0ec8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ha-guide/commit/cd21310d1e23cb32f5fb0f0adca4172d7cee0ec8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
