)]}'
{"id":"openstack%2Fhacking~1003084","triplet_id":"openstack%2Fhacking~master~I6b9f6f515969058fbc4b47478e670b5858934d51","project":"openstack/hacking","branch":"master","hashtags":[],"change_id":"I6b9f6f515969058fbc4b47478e670b5858934d51","subject":"H216 and H904 are enabled by default","status":"NEW","created":"2026-08-31 16:46:30.000000000","updated":"2026-08-31 17:03:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3026eb55448a3a55500cebabe511980e50851d3f","_number":1003084,"virtual_id_number":1003084,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-31 17:03:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-31 17:03:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4072831039226901f06ff98c0bd8cb6bd9500baa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-31 16:46:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be5f556407963a069eddde2ee46c3ea236634b70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-31 16:47:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3026eb55448a3a55500cebabe511980e50851d3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 17:03:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27598ba6a61942dc89253638905bb79e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2690ca36f0b8458686cf840abfd0fbf1 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1838f379ff4f4a1695bbdef81d4734c8 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af8bb140ed824e02b6e8562b85600a9c : SUCCESS in 2m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10d25b3e6728412dba5a3d269b316ed5 : SUCCESS in 2m 34s\n- hacking-integration-nova https://zuul.opendev.org/t/openstack/build/4c98083114cb407e99f2cc17dfdae1ac : FAILURE in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5cf7c3ec9fdeb7e610cc1373c93a7783e2c323f3","revisions":{"9d75f10e6e5de8b4deb1959cf71197e993b30182":{"kind":"REWORK","_number":1,"created":"2026-08-31 16:46:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/1003084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/hacking","ref":"refs/changes/84/1003084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/hacking refs/changes/84/1003084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/hacking refs/changes/84/1003084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/hacking refs/changes/84/1003084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/hacking refs/changes/84/1003084/1"}}},"commit":{"parents":[{"commit":"60ad62d6d59ab9cee43356b8f5f97f116907a8e4","subject":"Merge \"Add minimal typing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/hacking/commit/60ad62d6d59ab9cee43356b8f5f97f116907a8e4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-31 16:46:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-31 16:46:01.000000000","tz":540},"subject":"H216 check is enabled by default","message":"H216 check is enabled by default\n\nFix the wrong description about its default status.\n\nChange-Id: I6b9f6f515969058fbc4b47478e670b5858934d51\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/hacking/commit/9d75f10e6e5de8b4deb1959cf71197e993b30182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/hacking/commit/9d75f10e6e5de8b4deb1959cf71197e993b30182"}]},"branch":"refs/heads/master"},"5cf7c3ec9fdeb7e610cc1373c93a7783e2c323f3":{"kind":"REWORK","_number":2,"created":"2026-08-31 16:47:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/1003084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/hacking","ref":"refs/changes/84/1003084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/hacking refs/changes/84/1003084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/hacking refs/changes/84/1003084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/hacking refs/changes/84/1003084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/hacking refs/changes/84/1003084/2"}}},"commit":{"parents":[{"commit":"60ad62d6d59ab9cee43356b8f5f97f116907a8e4","subject":"Merge \"Add minimal typing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/hacking/commit/60ad62d6d59ab9cee43356b8f5f97f116907a8e4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-31 16:46:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-31 16:46:58.000000000","tz":540},"subject":"H216 and H904 are enabled by default","message":"H216 and H904 are enabled by default\n\nFix the wrong description about their default status.\n\nChange-Id: I6b9f6f515969058fbc4b47478e670b5858934d51\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/hacking/commit/5cf7c3ec9fdeb7e610cc1373c93a7783e2c323f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/hacking/commit/5cf7c3ec9fdeb7e610cc1373c93a7783e2c323f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
