)]}'
{"id":"openstack%2Fheat-dashboard~792201","triplet_id":"openstack%2Fheat-dashboard~master~I11176ef54ca25b130bc2c1c86b8458cbf960312f","project":"openstack/heat-dashboard","branch":"master","topic":"drop-nodejs10","hashtags":[],"change_id":"I11176ef54ca25b130bc2c1c86b8458cbf960312f","subject":"Drop horizon-nodejs10-jobs template","status":"MERGED","created":"2021-05-19 16:17:38.000000000","updated":"2021-07-21 15:32:40.000000000","submitted":"2021-07-21 15:29:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792201-drop-nodejs10","meta_rev_id":"a3106ace4911f5b872f0b424e4fb3c73d258ca30","_number":792201,"virtual_id_number":792201,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-21 15:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":2,"date":"2021-07-21 15:22:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":2,"date":"2021-07-07 15:25:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":2,"date":"2021-07-21 15:18:49.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":1,"date":"2021-07-21 15:18:49.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-19 16:55:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-17 10:35:42.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2021-06-30 14:46:04.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2021-07-07 15:25:55.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2021-07-21 15:22:11.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"}],"messages":[{"id":"6a33ca44bb911039bd43874e1aff100cc9d4708c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-05-19 16:17:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a63069409cf8f964fac094e705104f967c239d5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-05-19 16:19:26.000000000","message":"Topic set to drop-nodejs10","accounts_in_message":[],"_revision_number":1},{"id":"56c7a412b68155fdd2d7b01aa413568106f16217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 16:55:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a8a07f2c9414df3a01a812f41cb449b : SUCCESS in 4m 46s\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/3724d997bac1486db8e4327494970615 : SUCCESS in 5m 52s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/8e388a7436ed4c4aa9613f49ef0b954e : SUCCESS in 9m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36178db35f74414cb4b773b6a63bf187 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41753308daaa440baffd3c2263a7fe2d : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad99492b57c24f87a01d2be3538f89f8 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae8abdcfafbf497c84d61a8cb809685c : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c285582cbd5445419b4b308fe2310bc4 : SUCCESS in 6m 43s\n- heat-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/a0f9f4b1b9ef42c78f6e832794cdc528 : SUCCESS in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c0067e69e52c57c8d5f6884bd29c0ca7f0b93a3","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2021-07-07 15:25:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"483bff8bc42f5213c9da31d8391f9e6b558b6811","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-08 11:33:33.000000000","message":"Patch Set 1:\n\nHi all, Could you review this patch because horizon team is going to drop\nthe nodejs10 job once this patch is merged as nodejs10 is already EOL.\nthanks.","accounts_in_message":[],"_revision_number":1},{"id":"26d27a8a212c1f0bd253e4b7615a8848bbe46b00","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2021-07-21 15:18:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eb234bb9d4ae04fc5f0a3ab52d94984952315e00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 15:19:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52cab2ee2a5bb421a99b6602e9991666dd3f11cf","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2021-07-21 15:22:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7d532d6720ad74c37be399374d0f320727db3107","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 15:29:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd76ca8efe074c8db7a4286820f880dc : SUCCESS in 7m 34s\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/c32246fab9644631b24e12604c620849 : SUCCESS in 5m 46s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/5f0eac5b0f8840beb33a3622f875a44e : SUCCESS in 9m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb316a722c05495b9fe86c2cec960dff : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea707817d03048a58a1adc05607649be : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14b3c4a1db1844f58c3a7f061681133a : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed74abd6e4734732883fa691e4f69730 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"91fda35737a7543b2eadbe874365cef198b363e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 15:29:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a3106ace4911f5b872f0b424e4fb3c73d258ca30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 15:32:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4409ae4958d4623a540f5689e5b993e : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b31cd8319028573439ea9301b09768c73e724d77","revisions":{"b31cd8319028573439ea9301b09768c73e724d77":{"kind":"REWORK","_number":1,"created":"2021-05-19 16:17:38.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/01/792201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat-dashboard","ref":"refs/changes/01/792201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat-dashboard refs/changes/01/792201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat-dashboard refs/changes/01/792201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat-dashboard refs/changes/01/792201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat-dashboard refs/changes/01/792201/1"}}},"commit":{"parents":[{"commit":"6e698bbe09c74b7854e088d363d3e097ce776d1c","subject":"Add horizon-nodejs14-job template.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-dashboard/commit/6e698bbe09c74b7854e088d363d3e097ce776d1c"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-05-19 16:15:36.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-05-19 16:15:36.000000000","tz":330},"subject":"Drop horizon-nodejs10-jobs template","message":"Drop horizon-nodejs10-jobs template\n\nAs we already added horizon-nodejs14-jobs template in this plugin\nand nodejs14 jobs running fine. So let\u0027s drop horizon-nodejs10-jobs\ntemplate because nodejs10 is already EOL.\n\nChange-Id: I11176ef54ca25b130bc2c1c86b8458cbf960312f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-dashboard/commit/b31cd8319028573439ea9301b09768c73e724d77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-dashboard/commit/b31cd8319028573439ea9301b09768c73e724d77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
