)]}'
{"id":"openstack%2Fheat-tempest-plugin~721060","triplet_id":"openstack%2Fheat-tempest-plugin~master~Ibf3737f1f35776705c39ccef8c744b29a2778496","project":"openstack/heat-tempest-plugin","branch":"master","topic":"fix-gates","hashtags":[],"change_id":"Ibf3737f1f35776705c39ccef8c744b29a2778496","subject":"Fix stable gates","status":"ABANDONED","created":"2020-04-19 07:30:09.000000000","updated":"2020-05-06 09:43:48.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8d32e647ebe7a669227c829221aa2bc2c3434a04","_number":721060,"virtual_id_number":721060,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-20 17:50:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 14:38:36.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-04-20 17:50:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88b8c1320245e55505c3503469f2758305d6161e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-19 07:30:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b3c90a0ff29dc6ffe6c1bdfbd780a3b1daf4978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 07:30:24.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/heat-tempest-plugin on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027required-projects\u0027][0][\u0027openstack/tempest\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: heat-functional-convg-queens\n      parent: heat-functional-convg-mysql-lbaasv2-py3\n      nodeset: legacy-ubuntu-xenial\n      override-checkout: stable/queens\n      vars:\n        branch_override: stable/queens\n      required-projects:\n        - openstack/tempest:\n            override-checkout: 18.0.0\n\n  in \"openstack/heat-tempest-plugin/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"1364c4b44f80c05605a737bf4f5ca5cdc60e5c4d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-19 07:32:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a0be8dcb3501604adc3277777dfb46ef2f37cb46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 07:32:52.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/heat-tempest-plugin on branch master.  The error was:\n\n  mapping values are not allowed here\n  in \"openstack/heat-tempest-plugin/.zuul.yaml@master\", line 9, column 32","accounts_in_message":[],"_revision_number":2},{"id":"da20d97186a9d63afad2fdca1859cba58844fa88","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-19 07:43:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c5e870783b25fb094d85a81c37d441e0fa22673a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 09:09:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85b5fdb9b52f4167bda5187bb821c1a6 : SUCCESS in 4m 30s\n- heat-functional https://zuul.opendev.org/t/openstack/build/056541fb50f844cab54d44ef88089edd : SUCCESS in 1h 23m 58s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/f2af91bcfdbd421ca0d0083cdfd3ed47 : SUCCESS in 1h 24m 38s (non-voting)\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/7ece3da1e7c24e2499d9eef7dbb720d4 : SUCCESS in 1h 01m 39s\n- heat-functional-orig-queens https://zuul.opendev.org/t/openstack/build/b53c21b9e92240e783dfe0c4b1078547 : SUCCESS in 1h 03m 47s\n- heat-functional-orig-rocky https://zuul.opendev.org/t/openstack/build/69a03beb2e10468e85af93dc73b52dc7 : SUCCESS in 1h 17m 51s\n- heat-functional-convg-stein https://zuul.opendev.org/t/openstack/build/accf055fa7e14106a40f8700b96543b0 : FAILURE in 20m 35s\n- heat-functional-orig-stein https://zuul.opendev.org/t/openstack/build/8fa0b6ffe6534b458525e4cd7f8646e5 : FAILURE in 18m 37s\n- heat-functional-convg-train https://zuul.opendev.org/t/openstack/build/de0135662dc1413ba4857303b86cac43 : FAILURE in 19m 10s\n- heat-functional-orig-train https://zuul.opendev.org/t/openstack/build/34f24df3098f4c338943c27c4df0079f : FAILURE in 18m 43s","accounts_in_message":[],"_revision_number":3},{"id":"7a855170b54ddac9206fbc646e66db7ab0d27239","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-19 09:16:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b5aca124ed000a1fe358a0bc0b2213392fca654f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 10:40:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/353b5b1868d24f7ca38c03d05ff7cd83 : SUCCESS in 4m 00s\n- heat-functional https://zuul.opendev.org/t/openstack/build/533bc4a65d054e638e10a7cebac29cb6 : SUCCESS in 1h 23m 33s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/d704c7afaf94478ca298dcda417ed84c : SUCCESS in 1h 04m 54s (non-voting)\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/527bac2ff3f74beb987baa18e2908feb : SUCCESS in 1h 05m 47s\n- heat-functional-orig-queens https://zuul.opendev.org/t/openstack/build/b1c3eea9c52744d2bdc171e6a29f57a0 : SUCCESS in 1h 02m 46s\n- heat-functional-orig-rocky https://zuul.opendev.org/t/openstack/build/ff02a4565a224e2aa67dfba49fbad200 : SUCCESS in 57m 55s\n- heat-functional-convg-stein https://zuul.opendev.org/t/openstack/build/ef71ae45e16346aea7d4f48d7c58cc96 : FAILURE in 21m 13s\n- heat-functional-orig-stein https://zuul.opendev.org/t/openstack/build/0ecc632c55564d54b6c8ebb9e66deef4 : FAILURE in 18m 02s\n- heat-functional-convg-train https://zuul.opendev.org/t/openstack/build/6ea75bb5345041db8d2f697ab8ae1008 : FAILURE in 18m 53s\n- heat-functional-orig-train https://zuul.opendev.org/t/openstack/build/e30a3930dd5e40dea66932963efb3703 : FAILURE in 18m 25s","accounts_in_message":[],"_revision_number":4},{"id":"b18ff2549c21455722b8a3756b54ff929897c5c2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-20 14:38:36.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c63654d46ce588b5355d25d882ec9d474d5a293a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-20 14:45:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"18ff4f67c4b16530929c419cb970184d677446f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 17:50:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cd78cab851f46cd8c3e7145ffbd3e8a : SUCCESS in 4m 41s\n- heat-functional https://zuul.opendev.org/t/openstack/build/e4df4e4640214e709662987e06e107cd : SUCCESS in 1h 01m 42s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/6a985b6074dd432c9d6958504ba449af : SUCCESS in 1h 03m 58s (non-voting)\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/a13972af66dd4bd98072a8f63733627f : SUCCESS in 1h 25m 58s\n- heat-functional-orig-queens https://zuul.opendev.org/t/openstack/build/fa3922cf5122481da8e2f0d6eecaf8c6 : SUCCESS in 1h 13m 47s\n- heat-functional-orig-rocky https://zuul.opendev.org/t/openstack/build/10ab2b0616724966a4e6954a5830abe9 : SUCCESS in 1h 31m 54s\n- heat-functional-convg-stein https://zuul.opendev.org/t/openstack/build/8ea48b89d3e64310a9bd1f91d8e90260 : FAILURE in 21m 37s\n- heat-functional-orig-stein https://zuul.opendev.org/t/openstack/build/b2cda7d38bfc48658865e7ad5e43dcf0 : FAILURE in 26m 25s\n- heat-functional-convg-train https://zuul.opendev.org/t/openstack/build/65d52f37e0af4d88a34942cef805e8c2 : FAILURE in 22m 09s\n- heat-functional-orig-train https://zuul.opendev.org/t/openstack/build/df4910bea3cb446bb2ec287f91bad82d : FAILURE in 24m 18s","accounts_in_message":[],"_revision_number":5},{"id":"e3337f01171eb2e28a68bf865775cef6735f369b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-06 09:43:48.000000000","message":"Abandoned\n\ndone differently","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a5d746915a9ffe546ee12820098170a7972ac6df","revisions":{"cefa747b546621828617bfe7ee484428e5d121d1":{"kind":"REWORK","_number":1,"created":"2020-04-19 07:30:09.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/60/721060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat-tempest-plugin","ref":"refs/changes/60/721060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/1"}}},"commit":{"parents":[{"commit":"37a8df97277b42693ead3f5709f23544788f5a9f","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/37a8df97277b42693ead3f5709f23544788f5a9f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 07:28:58.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 07:28:58.000000000","tz":120},"subject":"WIP: Fix stable gates","message":"WIP: Fix stable gates\n\ntempest is not supporting older stable branches anymore, so we need to\nuse specific versions for older releases.\n\nThis is WIP, updates only queens to see whether this works as expected.\n\nChange-Id: Ibf3737f1f35776705c39ccef8c744b29a2778496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/cefa747b546621828617bfe7ee484428e5d121d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/cefa747b546621828617bfe7ee484428e5d121d1"}]},"branch":"refs/heads/master"},"2de50c782679496bc94435d10123f99b244beb3b":{"kind":"REWORK","_number":2,"created":"2020-04-19 07:32:36.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/60/721060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat-tempest-plugin","ref":"refs/changes/60/721060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/2"}}},"commit":{"parents":[{"commit":"37a8df97277b42693ead3f5709f23544788f5a9f","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/37a8df97277b42693ead3f5709f23544788f5a9f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 07:28:58.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 07:32:32.000000000","tz":120},"subject":"WIP: Fix stable gates","message":"WIP: Fix stable gates\n\ntempest is not supporting older stable branches anymore, so we need to\nuse specific versions for older releases.\n\nThis is WIP, updates only queens to see whether this works as expected.\n\nChange-Id: Ibf3737f1f35776705c39ccef8c744b29a2778496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/2de50c782679496bc94435d10123f99b244beb3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/2de50c782679496bc94435d10123f99b244beb3b"}]},"branch":"refs/heads/master"},"611b4b1c6a3de2d0a3463b8c0ac28004d0781c00":{"kind":"REWORK","_number":3,"created":"2020-04-19 07:43:14.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/60/721060/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat-tempest-plugin","ref":"refs/changes/60/721060/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/3"}}},"commit":{"parents":[{"commit":"37a8df97277b42693ead3f5709f23544788f5a9f","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/37a8df97277b42693ead3f5709f23544788f5a9f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 07:28:58.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 07:43:10.000000000","tz":120},"subject":"WIP: Fix stable gates","message":"WIP: Fix stable gates\n\ntempest is not supporting older stable branches anymore, so we need to\nuse specific versions for older releases.\n\nThis is WIP, updates only queens to see whether this works as expected.\n\nChange-Id: Ibf3737f1f35776705c39ccef8c744b29a2778496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/611b4b1c6a3de2d0a3463b8c0ac28004d0781c00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/611b4b1c6a3de2d0a3463b8c0ac28004d0781c00"}]},"branch":"refs/heads/master"},"550efa83a5e56beb6e9fba4682b7b19d248a3663":{"kind":"REWORK","_number":4,"created":"2020-04-19 09:16:16.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/60/721060/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat-tempest-plugin","ref":"refs/changes/60/721060/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/4"}}},"commit":{"parents":[{"commit":"37a8df97277b42693ead3f5709f23544788f5a9f","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/37a8df97277b42693ead3f5709f23544788f5a9f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 07:28:58.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 09:15:51.000000000","tz":120},"subject":"Fix stable gates","message":"Fix stable gates\n\ntempest is not supporting older stable branches anymore, so we need to\nuse specific versions of tempest for older releases.\n\nChange-Id: Ibf3737f1f35776705c39ccef8c744b29a2778496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/550efa83a5e56beb6e9fba4682b7b19d248a3663"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/550efa83a5e56beb6e9fba4682b7b19d248a3663"}]},"branch":"refs/heads/master"},"a5d746915a9ffe546ee12820098170a7972ac6df":{"kind":"REWORK","_number":5,"created":"2020-04-20 14:45:16.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/60/721060/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat-tempest-plugin","ref":"refs/changes/60/721060/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/60/721060/5"}}},"commit":{"parents":[{"commit":"37a8df97277b42693ead3f5709f23544788f5a9f","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/37a8df97277b42693ead3f5709f23544788f5a9f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 07:28:58.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-20 14:43:35.000000000","tz":120},"subject":"Fix stable gates","message":"Fix stable gates\n\nTempest master version is not compatible with env \u003c py3.6 version. Xenial node is py3.5\nso we have to use the old version of Tempest which is 23.0.0[1].\nWe have to do the same for these jobs explicitly as these jobs run on xenial node on stable/stein gate.\n\nUpdate other branches in same way.\n\n[1]\nhttps://opendev.org/openstack/devstack/src/commit/8cd4e9a9076e05974ed1242ccab3391c55c7070c/stackrc#L316\n\nChange-Id: Ibf3737f1f35776705c39ccef8c744b29a2778496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/a5d746915a9ffe546ee12820098170a7972ac6df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/a5d746915a9ffe546ee12820098170a7972ac6df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
